)]}'
{"id":"openstack%2Frequirements~843793","triplet_id":"openstack%2Frequirements~master~Id8c3b500f2c7c247cfdef84631e35c500ca24ebd","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-05-31 14:37:39.000000000","reason":"Removed by jpodivin by clicking the attention icon"}},"hashtags":[],"change_id":"Id8c3b500f2c7c247cfdef84631e35c500ca24ebd","subject":"DNM debugging validate-projects job","status":"ABANDONED","created":"2022-05-30 07:55:30.000000000","updated":"2022-10-21 12:59:03.000000000","total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"1f31b32f73888c7f0eed7ac2ea529e7c2d63c9ca","_number":843793,"virtual_id_number":843793,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-30 09:31:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-1,"date":"2022-06-05 19:03:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-05-30 07:55:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-30 07:56:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-05 19:03:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"b4abcfc93bfc73378350e4d94649d7be2bd3ce9f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-05-30 07:55:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c2942f380ec682fffda70624f3c8d4c841b03624","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:55:58.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch master.  The error was:\n\n  expected a dictionary for dictionary value @ data[\u0027debug\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      debug: true\n      templates:\n        - openstack-python3-zed-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              voting: false\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-aodh-py38\n          - cross-barbican-py38\n          - cross-ceilometer-py38\n          - cross-cinder-py38\n          - cross-designate-py38\n          - cross-glance-py38\n          - cross-heat-py38\n          - cross-horizon-py38\n          - cross-horizon-npm\n          - cross-ironic-py38\n          - cross-keystone-py38\n          - cross-kuryr-py38\n          - cross-manila-py38\n          - cross-masakari-py38\n          - cross-neutron-py38\n          - cross-nova-py38\n          - cross-placement-py38\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py38\n          - cross-octavia-py38\n          - cross-osc-tox-docs\n          - cross-swift-py38\n          - openstack-tox-py38\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      check-arm64:\n        jobs:\n          - requirements-tox-py38-check-uc-aarch64:\n              files:\n                - bindep.txt\n                - upper-constraints.txt\n          - build-wheel-cache-centos-8-stream-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy-arm64:\n              files:\n                - bindep.txt\n      gate:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - cross-cinder-py38\n          - cross-designate-py38\n          - cross-glance-py38\n          - cross-heat-py38\n          - cross-horizon-py38\n          - cross-horizon-npm\n          - cross-ironic-py38\n          - cross-keystone-py38\n          - cross-kuryr-py38\n          - cross-manila-py38\n          - cross-masakari-py38\n          - cross-neutron-py38\n          - cross-nova-py38:\n              files: ^upper-constraints.txt$\n          - cross-placement-py38:\n              files: ^upper-constraints.txt$\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py38\n          - cross-octavia-py38\n          - cross-osc-tox-docs\n          - cross-swift-py38\n          - openstack-tox-py38\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"6d4906689b2cfa16cf5e2645f9af249d3971babe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 07:56:04.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/requirements on branch master.  The error was:\n\n  expected a dictionary for dictionary value @ data[\u0027debug\u0027]\n\nThe error appears in the following project stanza:\n\n  project:\n      debug: true\n      templates:\n        - openstack-python3-zed-jobs\n      check:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-bindep:\n              voting: false\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - requirements-tox-validate-projects:\n              voting: false\n          - cross-aodh-py38\n          - cross-barbican-py38\n          - cross-ceilometer-py38\n          - cross-cinder-py38\n          - cross-designate-py38\n          - cross-glance-py38\n          - cross-heat-py38\n          - cross-horizon-py38\n          - cross-horizon-npm\n          - cross-ironic-py38\n          - cross-keystone-py38\n          - cross-kuryr-py38\n          - cross-manila-py38\n          - cross-masakari-py38\n          - cross-neutron-py38\n          - cross-nova-py38\n          - cross-placement-py38\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py38\n          - cross-octavia-py38\n          - cross-osc-tox-docs\n          - cross-swift-py38\n          - openstack-tox-py38\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      check-arm64:\n        jobs:\n          - requirements-tox-py38-check-uc-aarch64:\n              files:\n                - bindep.txt\n                - upper-constraints.txt\n          - build-wheel-cache-centos-8-stream-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal-arm64:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy-arm64:\n              files:\n                - bindep.txt\n      gate:\n        jobs:\n          - build-wheel-cache-centos-7:\n              files:\n                - bindep.txt\n          - build-wheel-cache-centos-8-stream:\n              files:\n                - bindep.txt\n          - build-wheel-cache-debian-buster:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-xenial:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-bionic:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-focal:\n              files:\n                - bindep.txt\n          - build-wheel-cache-ubuntu-jammy:\n              files:\n                - bindep.txt\n          - openstack-tox-validate\n          - requirements-tox-babel\n          - requirements-tox-py38-check-uc\n          - requirements-tox-py39-check-uc\n          - cross-cinder-py38\n          - cross-designate-py38\n          - cross-glance-py38\n          - cross-heat-py38\n          - cross-horizon-py38\n          - cross-horizon-npm\n          - cross-ironic-py38\n          - cross-keystone-py38\n          - cross-kuryr-py38\n          - cross-manila-py38\n          - cross-masakari-py38\n          - cross-neutron-py38\n          - cross-nova-py38:\n              files: ^upper-constraints.txt$\n          - cross-placement-py38:\n              files: ^upper-constraints.txt$\n          - cross-osvif-py3\n          - cross-nova-functional\n          - cross-placement-functional-py38\n          - cross-octavia-py38\n          - cross-osc-tox-docs\n          - cross-swift-py38\n          - openstack-tox-py38\n          - openstacksdk-functional-devstack\n          - tempest-full-py3:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n      experimental:\n        jobs:\n          - tempest-pg-full:\n              irrelevant-files:\n                - ^(test-|)requirements.txt$\n                - ^setup.cfg$\n\n  in \"openstack/requirements/.zuul.d/project.yaml@master\", line 1, column 3","accounts_in_message":[],"_revision_number":1},{"id":"644e2cce337f3dca7365ceefeee4f4c3015ba69f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-05-30 08:04:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"330c94eb2c5c0a13e152b18c1e543701ce9a8596","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 09:31:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/922d05297c54466c85de605020a2e643 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9072a9b61246578d1d93ae79773d87 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/399ce719f9f846c5bd2bf7f62d75da62 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/33ce4293b7f04ebeae0c61a1a1f30bce : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2f609aa3623e4b8099ce621b368dad75 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cbde9f467d2c4bd598ff2105c4efe013 : SUCCESS in 3m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/08607f3c1c854aa0baf4f086fc39acb4 : SUCCESS in 45m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa1bb57736dd4f7bbb684c4d8f57534a : SUCCESS in 1h 25m 09s\n\nDebug information:\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/project-config/zuul.d/projects.yaml@master None\u003e matched\n  Project template \u003cProjectConfig publish-to-pypi source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master None\u003e matched\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@master {ImpliedBranchMatcher:master}\u003e matched\n  Project template \u003cProjectConfig openstack-python3-zed-jobs source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master None\u003e matched\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/newton {ImpliedBranchMatcher:stable/newton}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/pike {ImpliedBranchMatcher:stable/pike}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/queens {ImpliedBranchMatcher:stable/queens}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/rocky {ImpliedBranchMatcher:stable/rocky}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/stein {ImpliedBranchMatcher:stable/stein}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/train {ImpliedBranchMatcher:stable/train}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/ussuri {ImpliedBranchMatcher:stable/ussuri}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/victoria {ImpliedBranchMatcher:stable/victoria}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/wallaby {ImpliedBranchMatcher:stable/wallaby}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/xena {ImpliedBranchMatcher:stable/xena}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/requirements/.zuul.d/project.yaml@stable/yoga {ImpliedBranchMatcher:stable/yoga}\u003e did not match\n  Project \u003cProjectConfig opendev.org/openstack/requirements source: openstack/project-config/zuul.d/projects.yaml@master None\u003e matched\n  Project template \u003cProjectConfig system-required source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master None\u003e matched\n  Freezing job graph\n    Freezing job test-release-openstack\n      Collecting job variants for test-release-openstack\n      Variant \u003cJob test-release-openstack branches: None source: openstack/project-config/zuul.d/jobs.yaml@master#50\u003e matched\n        Collecting job variants for base\n        Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n      Pipeline variant \u003cJob test-release-openstack branches: None source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master#164\u003e matched\n      Job test-release-openstack did not match files\n    Freezing job openstack-tox-docs\n      Collecting job variants for openstack-tox-docs\n      Variant \u003cJob openstack-tox-docs branches: {MatchAny:{BranchMatcher:^(?!stable/(ocata|pike|queens|rocky|stein|train|ussuri)).*$}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#702\u003e matched\n        Collecting job variants for opendev-tox-docs\n        Variant \u003cJob opendev-tox-docs branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#235\u003e matched\n          Collecting job variants for base\n          Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n      Variant \u003cJob openstack-tox-docs branches: {MatchAny:{BranchMatcher:stable/stein},{BranchMatcher:stable/train},{BranchMatcher:stable/ussuri}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#739\u003e did not match\n      Variant \u003cJob openstack-tox-docs branches: {MatchAny:{BranchMatcher:stable/ocata},{BranchMatcher:stable/pike},{BranchMatcher:stable/queens},{BranchMatcher:stable/rocky}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#783\u003e did not match\n      Pipeline variant \u003cJob openstack-tox-docs branches: {MatchAny:{BranchMatcher:master}} source: openstack/project-config/zuul.d/projects.yaml@master#4774\u003e matched\n    Freezing job openstack-tox-pep8\n      Collecting job variants for openstack-tox-pep8\n      Variant \u003cJob openstack-tox-pep8 branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#339\u003e matched\n        Collecting job variants for openstack-tox\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:^(?!stable/(ocata|pike|queens|rocky|stein|train|ussuri)).*$}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#47\u003e matched\n          Collecting job variants for tox-fips\n          Variant \u003cJob tox-fips branches: None source: zuul/zuul-jobs/zuul.d/python-jobs.yaml@master#50\u003e matched\n            Collecting job variants for unittests-fips\n            Variant \u003cJob unittests-fips branches: None source: zuul/zuul-jobs/zuul.d/general-jobs.yaml@master#20\u003e matched\n              Collecting job variants for base\n              Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/stein},{BranchMatcher:stable/train},{BranchMatcher:stable/ussuri}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#63\u003e did not match\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/ocata},{BranchMatcher:stable/pike},{BranchMatcher:stable/queens},{BranchMatcher:stable/rocky}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#85\u003e did not match\n      Pipeline variant \u003cJob openstack-tox-pep8 branches: None source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master#541\u003e matched\n    Freezing job openstack-tox-py38\n      Collecting job variants for openstack-tox-py38\n      Variant \u003cJob openstack-tox-py38 branches: {MatchAny:{BranchMatcher:^(?!stable/(train|ussuri)).*$}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#207\u003e matched\n        Collecting job variants for openstack-tox\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:^(?!stable/(ocata|pike|queens|rocky|stein|train|ussuri)).*$}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#47\u003e matched\n          Collecting job variants for tox-fips\n          Variant \u003cJob tox-fips branches: None source: zuul/zuul-jobs/zuul.d/python-jobs.yaml@master#50\u003e matched\n            Collecting job variants for unittests-fips\n            Variant \u003cJob unittests-fips branches: None source: zuul/zuul-jobs/zuul.d/general-jobs.yaml@master#20\u003e matched\n              Collecting job variants for base\n              Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/stein},{BranchMatcher:stable/train},{BranchMatcher:stable/ussuri}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#63\u003e did not match\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/ocata},{BranchMatcher:stable/pike},{BranchMatcher:stable/queens},{BranchMatcher:stable/rocky}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#85\u003e did not match\n      Variant \u003cJob openstack-tox-py38 branches: {MatchAny:{BranchMatcher:stable/train},{BranchMatcher:stable/ussuri}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#225\u003e did not match\n      Pipeline variant \u003cJob openstack-tox-py38 branches: None source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master#541\u003e matched\n      Pipeline variant \u003cJob openstack-tox-py38 branches: None source: openstack/requirements/.zuul.d/project.yaml@master#1\u003e matched\n    Freezing job openstack-tox-py39\n      Collecting job variants for openstack-tox-py39\n      Variant \u003cJob openstack-tox-py39 branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#259\u003e matched\n        Collecting job variants for openstack-tox\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:^(?!stable/(ocata|pike|queens|rocky|stein|train|ussuri)).*$}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#47\u003e matched\n          Collecting job variants for tox-fips\n          Variant \u003cJob tox-fips branches: None source: zuul/zuul-jobs/zuul.d/python-jobs.yaml@master#50\u003e matched\n            Collecting job variants for unittests-fips\n            Variant \u003cJob unittests-fips branches: None source: zuul/zuul-jobs/zuul.d/general-jobs.yaml@master#20\u003e matched\n              Collecting job variants for base\n              Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/stein},{BranchMatcher:stable/train},{BranchMatcher:stable/ussuri}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#63\u003e did not match\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/ocata},{BranchMatcher:stable/pike},{BranchMatcher:stable/queens},{BranchMatcher:stable/rocky}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#85\u003e did not match\n      Pipeline variant \u003cJob openstack-tox-py39 branches: None source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master#541\u003e matched\n    Freezing job openstack-tox-py310\n      Collecting job variants for openstack-tox-py310\n      Variant \u003cJob openstack-tox-py310 branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#288\u003e matched\n        Collecting job variants for openstack-tox\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:^(?!stable/(ocata|pike|queens|rocky|stein|train|ussuri)).*$}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#47\u003e matched\n          Collecting job variants for tox-fips\n          Variant \u003cJob tox-fips branches: None source: zuul/zuul-jobs/zuul.d/python-jobs.yaml@master#50\u003e matched\n            Collecting job variants for unittests-fips\n            Variant \u003cJob unittests-fips branches: None source: zuul/zuul-jobs/zuul.d/general-jobs.yaml@master#20\u003e matched\n              Collecting job variants for base\n              Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/stein},{BranchMatcher:stable/train},{BranchMatcher:stable/ussuri}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#63\u003e did not match\n        Variant \u003cJob openstack-tox branches: {MatchAny:{BranchMatcher:stable/ocata},{BranchMatcher:stable/pike},{BranchMatcher:stable/queens},{BranchMatcher:stable/rocky}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#85\u003e did not match\n      Pipeline variant \u003cJob openstack-tox-py310 branches: None source: openstack/openstack-zuul-jobs/zuul.d/project-templates.yaml@master#541\u003e matched\n    Freezing job build-wheel-cache-centos-7\n      Collecting job variants for build-wheel-cache-centos-7\n      Variant \u003cJob build-wheel-cache-centos-7 branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1580\u003e matched\n        Collecting job variants for build-wheel-cache-base\n        Variant \u003cJob build-wheel-cache-base branches: {MatchAny:{BranchMatcher:master}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1557\u003e matched\n          Collecting job variants for base\n          Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n      Pipeline variant \u003cJob build-wheel-cache-centos-7 branches: None source: openstack/requirements/.zuul.d/project.yaml@master#1\u003e matched\n      Job build-wheel-cache-centos-7 did not match files\n    Freezing job build-wheel-cache-centos-8-stream\n      Collecting job variants for build-wheel-cache-centos-8-stream\n      Variant \u003cJob build-wheel-cache-centos-8-stream branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1623\u003e matched\n        Collecting job variants for build-wheel-cache-base\n        Variant \u003cJob build-wheel-cache-base branches: {MatchAny:{BranchMatcher:master}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1557\u003e matched\n          Collecting job variants for base\n          Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n      Pipeline variant \u003cJob build-wheel-cache-centos-8-stream branches: None source: openstack/requirements/.zuul.d/project.yaml@master#1\u003e matched\n      Job build-wheel-cache-centos-8-stream did not match files\n    Freezing job build-wheel-cache-debian-buster\n      Collecting job variants for build-wheel-cache-debian-buster\n      Variant \u003cJob build-wheel-cache-debian-buster branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1675\u003e matched\n        Collecting job variants for build-wheel-cache-base\n        Variant \u003cJob build-wheel-cache-base branches: {MatchAny:{BranchMatcher:master}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1557\u003e matched\n          Collecting job variants for base\n          Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n      Pipeline variant \u003cJob build-wheel-cache-debian-buster branches: None source: openstack/requirements/.zuul.d/project.yaml@master#1\u003e matched\n      Job build-wheel-cache-debian-buster did not match files\n    Freezing job build-wheel-cache-ubuntu-xenial\n      Collecting job variants for build-wheel-cache-ubuntu-xenial\n      Variant \u003cJob build-wheel-cache-ubuntu-xenial branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1795\u003e matched\n        Collecting job variants for build-wheel-cache-base\n        Variant \u003cJob build-wheel-cache-base branches: {MatchAny:{BranchMatcher:master}} source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1557\u003e matched\n          Collecting job variants for base\n          Variant \u003cJob base branches: None source: opendev/base-jobs/zuul.d/jobs.yaml@master#25\u003e matched\n      Pipeline variant \u003cJob build-wheel-cache-ubuntu-xenial branches: None source: openstack/requirements/.zuul.d/project.yaml@master#1\u003e matched\n      Job build-wheel-cache-ubuntu-xenial did not match files\n    Freezing job build-wheel-cache-ubuntu-bionic\n      Collecting job variants for build-wheel-cache-ubuntu-bionic\n      Variant \u003cJob build-wheel-cache-ubuntu-bionic branches: None source: openstack/openstack-zuul-jobs/zuul.d/jobs.yaml@master#1761\u003e matched\n        Co... (truncated)","accounts_in_message":[],"_revision_number":2},{"id":"a8bca90bfd0a47156bff621f2d43e12362b048ff","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-05 19:03:53.000000000","message":"Patch Set 2: Workflow-1\n\ncan this be abandoned now or is it still useful?","accounts_in_message":[],"_revision_number":2},{"id":"1f31b32f73888c7f0eed7ac2ea529e7c2d63c9ca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-10-21 12:59:03.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81a27b60dda7a444bbaf256109a2bd813fb3a8b2","revisions":{"f7e8eb90d69807bb77447a1e0ef6358c262a05af":{"kind":"REWORK","_number":1,"created":"2022-05-30 07:55:30.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/93/843793/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/843793/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/843793/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/843793/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/843793/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/843793/1"}}},"commit":{"parents":[{"commit":"7ac0e631833ee8c33b42ddbcc23be89c05dcc658","subject":"remove py36 specifiers from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac0e631833ee8c33b42ddbcc23be89c05dcc658"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-05-30 07:55:24.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-05-30 07:55:24.000000000","tz":120},"subject":"DNM debugging validate-projects job","message":"DNM debugging validate-projects job\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Id8c3b500f2c7c247cfdef84631e35c500ca24ebd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7e8eb90d69807bb77447a1e0ef6358c262a05af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f7e8eb90d69807bb77447a1e0ef6358c262a05af"}]},"branch":"refs/heads/master"},"81a27b60dda7a444bbaf256109a2bd813fb3a8b2":{"kind":"REWORK","_number":2,"created":"2022-05-30 08:04:26.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/93/843793/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/93/843793/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/843793/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/843793/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/93/843793/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/93/843793/2"}}},"commit":{"parents":[{"commit":"7ac0e631833ee8c33b42ddbcc23be89c05dcc658","subject":"remove py36 specifiers from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac0e631833ee8c33b42ddbcc23be89c05dcc658"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-05-30 07:55:24.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-05-30 08:04:21.000000000","tz":120},"subject":"DNM debugging validate-projects job","message":"DNM debugging validate-projects job\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: Id8c3b500f2c7c247cfdef84631e35c500ca24ebd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81a27b60dda7a444bbaf256109a2bd813fb3a8b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/81a27b60dda7a444bbaf256109a2bd813fb3a8b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
