)]}'
{"id":"openstack%2Frequirements~843805","triplet_id":"openstack%2Frequirements~master~I4f5d8afbe3d7603a2d28b13dd38c283f290b456d","project":"openstack/requirements","branch":"master","topic":"py36-glanceclient","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-05-31 14:36:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4f5d8afbe3d7603a2d28b13dd38c283f290b456d","subject":"Adding constraint for glanceclient for Python 3.6","status":"ABANDONED","created":"2022-05-30 09:19:45.000000000","updated":"2022-05-31 14:36:57.000000000","total_comment_count":2,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"27964af168aaca1bc90ef302bd0f1dcc09c5f5e2","_number":843805,"virtual_id_number":843805,"owner":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-05-30 11:16:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":-2,"date":"2022-05-30 13:31:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2022-05-30 09:51:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-30 11:16:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-30 13:31:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5d8b0d42ec57f6b2b81aa486241e25f1f48c6cc4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-05-30 09:19:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fefd67a33bc8091813ad6841dca4e571d4f87f2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 09:51:12.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8eff6d459d4d40c394109a45822ed248 : FAILURE in 18m 43s","accounts_in_message":[],"_revision_number":1},{"id":"ae59d640a55a31d37166fe88b3cc200cbbc6f68a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-30 11:16:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/794a10fc29034538af4e8eda34d9e31a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bd809794755410c87e8b3dfb84c91db : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd8da33f9b3249e3bd2a4b8a9cd2b01e : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cb2d2061df545768c3e05451c5c0d90 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bdd265b7a0e44aa8d56dcf5b189ea94 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a36816c57bfe4bd284bb2bcf21356ad4 : SUCCESS in 3m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1741f0b80c0e4e5ca88c3c24ffcc57ae : SUCCESS in 4m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a794c1b90d2247f386f98b0725ebd186 : SUCCESS in 8m 32s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/870421678f6d4e5a9ab1f508e0827ee7 : SUCCESS in 15m 20s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/dfc5970a904d4d69a1e9b5f4fb6cc59d : SUCCESS in 6m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ce63da95a3d44c27800ab7f995fdfe94 : SUCCESS in 4m 52s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3c67ce1e959f43cf878fb8ec20a9a829 : SUCCESS in 5m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3d052f08a56b430f8c2e9064fac04760 : SUCCESS in 10m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fbc6fa09229b45f29ab6482cca296334 : SUCCESS in 5m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/63a9412dc6b44c0c9489b696b1b0441d : SUCCESS in 5m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ccab86caef4f42a094715c802e89881c : SUCCESS in 8m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/37e4a51aa996493eafd0b21aa4b6c8c4 : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39c6c1a6cee24b479aba457f0f4ddf3d : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6f15207b07c64a4c9bdcea619c0b00cc : SUCCESS in 8m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/521fb3e6f43b48fe92741d51a1f04070 : SUCCESS in 11m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/34f10d30a79e441c984e0cda0c390121 : SUCCESS in 4m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a86e815bea2f47b38ae2525e53ca0590 : SUCCESS in 9m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/914b090e9deb4202932344b06d80963d : SUCCESS in 4m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/daf09f6e89c8469aacf981fa582a6180 : SUCCESS in 22m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4ed5de59ffaa49ea99cca0e2ed456be4 : SUCCESS in 12m 54s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/02cece121aa14c6997313d3cfa1aa8f0 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e9e2babb3f6049df987949cb0576aa1f : SUCCESS in 4m 22s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/496ab77f8c7b422ea86991ef46766ff7 : SUCCESS in 22m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/718d1d16e8084390892172e158107c02 : SUCCESS in 5m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b914064882404122ba5312b4edac0eb6 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0b28aee03774468583a4901502344889 : SUCCESS in 8m 15s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fd7ab7361d8d4277beee3e6678ced5fa : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e9d18b4e7dd540128c7f6f98632f13f5 : SUCCESS in 56m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c724ab1e681f4a85b47038a0d9961a86 : SUCCESS in 1h 50m 41s","accounts_in_message":[],"_revision_number":1},{"id":"15c4c393f546d21c28cd6599a5279a479954954f","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-05-30 13:31:48.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"164e5f84533cdc3605caeb3197e89f3f048c4b09","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-05-31 14:36:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27964af168aaca1bc90ef302bd0f1dcc09c5f5e2","tag":"autogenerated:gerrit:abandon","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-05-31 14:36:57.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75bac865a0b20f47cdec08180c47ab1228dfe32b","revisions":{"75bac865a0b20f47cdec08180c47ab1228dfe32b":{"kind":"REWORK","_number":1,"created":"2022-05-30 09:19:45.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/05/843805/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/05/843805/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/843805/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/843805/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/05/843805/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/05/843805/1"}}},"commit":{"parents":[{"commit":"7ac0e631833ee8c33b42ddbcc23be89c05dcc658","subject":"remove py36 specifiers from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac0e631833ee8c33b42ddbcc23be89c05dcc658"}]}],"author":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-05-30 09:19:39.000000000","tz":120},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-05-30 09:19:39.000000000","tz":120},"subject":"Adding constraint for glanceclient for Python 3.6","message":"Adding constraint for glanceclient for Python 3.6\n\nSigned-off-by: Jiri Podivin \u003cjpodivin@redhat.com\u003e\nChange-Id: I4f5d8afbe3d7603a2d28b13dd38c283f290b456d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75bac865a0b20f47cdec08180c47ab1228dfe32b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/75bac865a0b20f47cdec08180c47ab1228dfe32b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
