)]}'
{"id":"openstack%2Frequirements~844099","triplet_id":"openstack%2Frequirements~master~Ia4ff0fb3bb946dba5112510684bc36ea9c30c1e4","project":"openstack/requirements","branch":"master","topic":"new-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-13 12:17:54.000000000","reason":"Dr. Jens Harbott (frickler) replied on the change"},"33493":{"account":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"last_update":"2022-06-13 17:07:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia4ff0fb3bb946dba5112510684bc36ea9c30c1e4","subject":"update constraint for XStatic-Angular to new release 1.8.2.2","status":"MERGED","created":"2022-05-31 15:39:21.000000000","updated":"2022-06-13 17:08:37.000000000","submitted":"2022-06-13 17:07:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"844099-new-release","meta_rev_id":"1f8841448216aa245eef4f6d572c0d0e36f1d629","_number":844099,"virtual_id_number":844099,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-13 17:07:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-13 15:23:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-06-13 15:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":-1,"date":"2022-06-07 09:25:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-13 15:23:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-31 15:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-05-31 17:11:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-01 07:01:41.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2022-06-07 09:25:44.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"},{"updated":"2022-06-09 08:32:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-06-10 15:25:55.000000000","updated_by":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"reviewer":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"state":"CC"},{"updated":"2022-06-13 15:06:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-13 15:23:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"5423260ed956a28680a0c9f7b5bda084c7aaf1a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-05-31 15:39:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b005c5dfcf1540146711268eb7444b15506711c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 15:51:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/17e207eec24048259cb9fdcce17779ac : FAILURE in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"1bfa3e4658b9296984f5261e1e4a6d38bb2ab305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 17:11:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0d5660b65eb467ba1db6ef1010a2312 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8aa8124b6bbe495590315d7c290605f6 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68bb7d582501494b94bdc5c6a2deb242 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f086c50ad9e749de9246fb8fc9b4123b : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/970f97f41d8b4c02a831ea8fcc7dcbf1 : SUCCESS in 3m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86fe049354e3471c8da4a1efe246263d : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7c2073de1c59441ea4bd6a3ea44fa5b5 : SUCCESS in 3m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/56ea6d22b67f452bb8f8d53d7bde8279 : SUCCESS in 8m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5619ca47c8af4c53b0ed14665e585289 : SUCCESS in 12m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/1639a3a0a04940a08b4c1022d5534523 : SUCCESS in 6m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/fbeff43534554048a0ccebc1d242901b : SUCCESS in 4m 56s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/bec9b81f217b4bceb31bf8fddc140f99 : SUCCESS in 5m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/18a07bd82f54484d84e9e9c073dd34d6 : SUCCESS in 9m 46s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1ea438d7b0c045b0bf15672722f313eb : SUCCESS in 4m 40s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/488b0ddf98354581af917e99719a4f55 : SUCCESS in 5m 53s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b6d66f7465ec440d883a66e9f9f839ea : SUCCESS in 7m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/50308d83bf224697a1a165fd60d8bdc7 : SUCCESS in 9m 16s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/30ca24d0d7434d4aadfc437806b58541 : FAILURE in 23m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/356cb868343f4a3ea4def04f689422ce : SUCCESS in 7m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/87a051db728e4460b1ba7db6ef53c949 : SUCCESS in 16m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0fe07f0bcf654049ada70cf6d149bd56 : SUCCESS in 5m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a27b43900e3c456da8f0ff48e56bd779 : SUCCESS in 16m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/018441fd3ece4f2caaf4c19579bf239c : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/50f44546bb934a25aa709793d1a037b2 : SUCCESS in 31m 01s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9bc50a3674b3415491fe5aa2769b3230 : SUCCESS in 13m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/05071b514efb42b1a22529e0c87f79ba : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab0765edffe046a286151bc737f19140 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8a93554be9414c5cb17897409a58d8e4 : SUCCESS in 19m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d75dd0375efd4360b18576872382443b : SUCCESS in 6m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f03c2928adac4579ab9c41ad4cbcf636 : SUCCESS in 5m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/98baba49130644f7bd6d059cd34762f1 : SUCCESS in 7m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3774e40820b848a2a371e648fccafad6 : SUCCESS in 11m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bfa72ba927847cf98a752fdf905cec4 : SUCCESS in 51m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/249f555419624b60b834f9c02b537167 : SUCCESS in 1h 28m 44s","accounts_in_message":[],"_revision_number":1},{"id":"f12e82d9cd883be9f59efd3f5f185ca065e96429","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-01 07:01:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1628c10da2cf8c99afa5ca77f8ee06fae17f0e55","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 07:17:30.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c52c53e1399d495abfbae391b42cc29b : FAILURE in 13m 14s","accounts_in_message":[],"_revision_number":1},{"id":"ca60652fd7ed46670c052656ea133ab482d4fa32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 08:28:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f15057e9ed04241bd1322b3e814fcc5 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72c268b1e8f3408ca4ec16ae03562989 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7747f39ba084c3085fc45ade797402c : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03645a32eacf4a2e8f548e2fedae17ba : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b8402bb49e949ed8e9089c43d99008d : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ebf57c98d0d04fe8a88cab743cad35cf : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a5ba1129b7274e5ea96a87304adc92b5 : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/02d0b333f44f415c92241b6328ff407f : SUCCESS in 8m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/676a7dfad2c54830b01b831921145911 : SUCCESS in 14m 11s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/44f8fcd28b484b129ad7d61a79b2e2b7 : SUCCESS in 8m 28s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2542de8776834816944d1bc7a4d3998d : SUCCESS in 5m 19s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/156b7c7aa839433591a0c2ab894d29e9 : SUCCESS in 6m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/78807d6146424923a70f9b13bb74b3be : SUCCESS in 10m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6ff20ced58c44723940ae5ceeabaec7b : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/150710ce1000438cb7a4b0f321037edd : SUCCESS in 5m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7aa4a0354313456ba73425e91048c1e7 : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1c1f603833404d6ab6045e0f4c752b37 : SUCCESS in 10m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8562c084e1594dbb8480e266ffe36757 : FAILURE in 11m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/af94897480f247fb8b1fd69796f66be8 : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b5135ba152594bdf968f653f3253cefc : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/af68d7faa81f4b458b9cf8074c2781ba : SUCCESS in 4m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e8aae961f39642efab6cc531cd9c4cfb : SUCCESS in 9m 32s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e6d93e9ac8d248fcabc069e3a174882f : SUCCESS in 5m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4d2bcc0173c34369a64df4b89ec6bb36 : SUCCESS in 26m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/548155551ce541159f81fdb257d8897f : SUCCESS in 14m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/49a72464e4d04a96bdcac8e07e4534e3 : SUCCESS in 6m 19s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/882f25104b6f4074b557f8e29515cf2c : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8449da95f74e4134b8855e918ab8202d : SUCCESS in 22m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9f7ea741b51b4997897b7fa676f77b93 : SUCCESS in 6m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/be6be724404b490f9a28d6e19754f69e : SUCCESS in 4m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4ce54a2fe5ee4a0e8e60e7f2fca95440 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/35dcead5c95a4c9aa8a5c798d1fdb19a : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec00879f74b04067a2f128e8b2277400 : SUCCESS in 52m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/277be0d53182489e9a7575277c69f948 : SUCCESS in 1h 24m 40s","accounts_in_message":[],"_revision_number":1},{"id":"da8c4a7f5301cac206d0faea4b587f7afbd1f7a0","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-06-07 09:25:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6e2a4387fe4006f377a0f30d12f9b480af7ee92","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-09 08:32:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61cc750bf63f3609ecff2a5b09a4263a67034f6c","author":{"_account_id":6914,"name":"Tatiana Ovchinnikova","email":"t.v.ovtchinnikova@gmail.com","username":"tmazur"},"date":"2022-06-10 15:25:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"823ac226dbf241c18207a0babf9d02d6f6db08b2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-13 12:17:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0a65cd98855dd632a02887f831722252a1feccf7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 12:33:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a9cfa8b3e94c4243a34ad157235cab0d : FAILURE in 13m 44s","accounts_in_message":[],"_revision_number":1},{"id":"404783fb99740dd413aeb3c944f685de1973e9de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 14:18:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8556db78a6d74bf4bc5cb4914af96a45 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc0dbac68f8e434d831d3b53b1a2fc08 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d55b4bda4c2c4f7ca03fe0633fdf82cc : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb358e3b73e0410594e20e7837d68799 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87dfa9b3450340198f47dac564e0c47d : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/212511ee2f844c88a3b4fb083e5f1d56 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7d22a69ca4434f01b1fa4557ef68399f : SUCCESS in 4m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/21f7d1e0fce646f4bd931616bffae94e : SUCCESS in 9m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8449a182b9ee4812a5f59012505861e9 : SUCCESS in 14m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/603f9c2e451540bd9857f2c62a6c561b : SUCCESS in 6m 33s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/54ce928b26f840fb8485e77720e343b9 : SUCCESS in 5m 07s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d7f22baac2ae4d88b66e1567c02ed794 : SUCCESS in 5m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2d12244598444610baf0e47e14bd1b78 : SUCCESS in 9m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3e42a41e10554c65af10d36fedf1f49f : SUCCESS in 4m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/07e77781b85b41849686163397f763fe : SUCCESS in 5m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c79fce05cd2043939cadd92c06184cec : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/516f5529d75444dbb8e7af57b33f31ed : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff7466f8a7a7419284e8a9fc87bb1538 : SUCCESS in 8m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/03df1b3a716c4fc49017890e8c1bdefb : SUCCESS in 7m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0aedf35a479f495cb412f29889653870 : SUCCESS in 12m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c563cdf8057e4cf2bd4c56f7bb150d82 : SUCCESS in 4m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6e364148003944719acb8f5b43cfdd2d : SUCCESS in 8m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ca12a35deffd4f7da1cb6aaa23cf3a46 : SUCCESS in 4m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7712e8f4cbb94f66ad7cb008fffc63b9 : SUCCESS in 23m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/761f7cf24ed14e6fab87c1480f5fccda : SUCCESS in 11m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f82a8f28e2e24ce6abd67ae11972abb8 : SUCCESS in 4m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e6c0a3ac019c4745bcf58458d648e8f8 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/82dd6400303f4fc58475bc0a92f60d57 : SUCCESS in 19m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f783d0bb5b6e40f690d921b0b1fac7a5 : SUCCESS in 5m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/760ca7b5c4a74dde8bc1bc8b41249dba : SUCCESS in 4m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8ba5f95cc911469e84e9da6b3daa1898 : SUCCESS in 6m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/10d28b8a0ce84e0c959f5e0ee0616bae : SUCCESS in 10m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cc96bacc7264298baa4415bfe0508ab : SUCCESS in 47m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a417de70e663454ca2e63ad4421dfcde : SUCCESS in 1h 24m 48s","accounts_in_message":[],"_revision_number":1},{"id":"18c0d96087281bb7099c6d2609b81251b7b2461e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-13 15:06:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3bb4859fbe10f39bb6bec34ffbf3668a24f2a780","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-13 15:23:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"24cacc30e0250307990f255f5173aa8a07b8246b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 15:24:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1f0e853b47cb459ed6a1710633f989a3ad9cd663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:07:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fce3d169c15e4f9ea1eb4abb8592270a : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f92cd0191cb4840bb95f82df4bfebe1 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7db2897efc1a4e19b24f4ad94da6f43b : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd74c500a0df4beeb7c44f528b1788f7 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8b400eeac4c43d6986dc7ea9cb3ca15 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dfd82e96ba13467db40c2dbc41a5808e : SUCCESS in 4m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d6fca492d8b347cf8af00dbcfe4669cc : SUCCESS in 9m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/949a4382d79e48029d25f9d408513b46 : SUCCESS in 13m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9cc0f189912044989e05c65f967767a1 : SUCCESS in 12m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6981b0d0e9974b088bbb20a9d35f6928 : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c24bb742e1a84dc3840d922d8011052f : SUCCESS in 5m 38s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/61f1cd47e5db432c86a70c9928fbe4e4 : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5172fd7000ac4229be562994f6252c4b : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ae58e76c42fd47e4a4946ed72551045e : SUCCESS in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4dd27e74e07d4f0b963532c3be564dac : SUCCESS in 7m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a23908ca761c41188df0aac7d11e8e60 : SUCCESS in 11m 57s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fc0c2f5c660e4ac8a537cbc1ae80fc5c : SUCCESS in 3m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ba08a85912dd40fd85b3cdd15ee4329f : SUCCESS in 11m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a834fe44165546f98f21d4bcb1d1818a : SUCCESS in 4m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b078909d07384be38d36a8450aaf8119 : SUCCESS in 21m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/463cc0fa107c4564897d5c13f308620d : SUCCESS in 12m 09s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d7c347b9113d430a892015cffa7cf652 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/95145d4a0cfb4e7088471854db9f6cad : SUCCESS in 3m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b31be054c0094f8ca422e2f970d3c89a : SUCCESS in 18m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/284ee0f0cc39442ea810da9a4dbf3b80 : SUCCESS in 6m 05s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5f5160ad7e73445ba86f83b26999210a : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6a22f87c7350449b829881d61985a1ee : SUCCESS in 7m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e39acb459d8b4f338895510f54bd4eb2 : SUCCESS in 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ee12d61d03349c887d966c208330de7 : SUCCESS in 56m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f98a94c49082482d988012d6195de9dd : SUCCESS in 1h 36m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fb0fd4757a2eb23b0b1b023fc7d1159008d39390","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:07:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1f8841448216aa245eef4f6d572c0d0e36f1d629","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 17:08:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18ba521f7f57460bb05b64dc73513975 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6aed4ecbf72f274e46020611f71336511ca6f641","revisions":{"6aed4ecbf72f274e46020611f71336511ca6f641":{"kind":"REWORK","_number":1,"created":"2022-05-31 15:39:21.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/844099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/844099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/844099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/844099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/844099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/844099/1"}}},"commit":{"parents":[{"commit":"7ac0e631833ee8c33b42ddbcc23be89c05dcc658","subject":"remove py36 specifiers from global-requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7ac0e631833ee8c33b42ddbcc23be89c05dcc658"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-31 15:39:20.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-05-31 15:39:20.000000000","tz":0},"subject":"update constraint for XStatic-Angular to new release 1.8.2.2","message":"update constraint for XStatic-Angular to new release 1.8.2.2\n\nmeta: version: 1.8.2.2\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Commit: Tatiana Ovchinnikova \u003ct.v.ovtchinnikova@gmail.com\u003e\nmeta: release:Change-Id: I0ec7003111adf477bfde6001d9f416938e123b4d\nmeta: release:Code-Review+1: Akihiro Motoki \u003camotoki@gmail.com\u003e\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: Radomir Dopieralski \u003copenstack@dopieralski.pl\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia4ff0fb3bb946dba5112510684bc36ea9c30c1e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aed4ecbf72f274e46020611f71336511ca6f641"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6aed4ecbf72f274e46020611f71336511ca6f641"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
