)]}'
{"id":"openstack%2Frequirements~845023","triplet_id":"openstack%2Frequirements~master~If2bfc21f2a06956a0142a6b056a151883348850c","project":"openstack/requirements","branch":"master","topic":"req-check-var","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-06-08 10:51:10.000000000","reason":"Brian Rosmaita replied on the change"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-08 13:01:22.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If2bfc21f2a06956a0142a6b056a151883348850c","subject":"Add variable to requirements-check job","status":"ABANDONED","created":"2022-06-07 19:53:19.000000000","updated":"2022-06-08 13:01:22.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c3f6a2235c322b7bfb90b05d489fa61b27d2f8fb","_number":845023,"virtual_id_number":845023,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-08 07:46:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2022-06-08 08:12:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":-1,"date":"2022-06-08 12:23:40.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-07 21:11:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-08 08:12:52.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"ead622d775ec193f981fbda9d9ed975e09e95be4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-07 19:53:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"15051050902bcbc60835ca3fd574e054c3512d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-07 21:11:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4db39479237c499183f7e73079e4398d : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d9f104fd6274001a737fe653f6d35ec : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21165c7a95fc432d8491f44ea8f1d19f : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d56a3e3eba224b34bfdb65099bf035e6 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1317db632cf4a31a05e8ec4785c54a2 : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/290ef219638f46a7bb982da0fdc45c7a : SUCCESS in 3m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2154954c38b2400c82105aacd4d7bec0 : SUCCESS in 48m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4169a76f43d94f7185b6d60345a44806 : SUCCESS in 1h 16m 59s","accounts_in_message":[],"_revision_number":1},{"id":"df4ad220f749e3a62314437e13eef3ac15c9c26e","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-07 21:23:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93e2248fc15a8b8bc71dd4ca89b2aad220c31060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-08 06:33:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fdc488d6eb96f52ff37ff3dce43d1265b58d41c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 07:46:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ab9903a84e64ac0850181632cfcca3b : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/525845ed592e4431bec94ea618c38698 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66a5d4d65ffd46b0b8713964e54b7c7b : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b413828e8d1c4d2796ccaaec6e864a1c : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f107678434a4695b4675b5691cc957b : SUCCESS in 3m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e9bddba1974b46978abb423e074c5a58 : SUCCESS in 3m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a51ab92f9ba348b988d6244187ec369c : SUCCESS in 39m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/857ab7b7cd524013a2e0f47759d443d7 : SUCCESS in 1h 11m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a119948adf2eab83e646b6ee29eff7d44f1ece99","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-08 08:12:52.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2b47fc25f2956e02c705e669d1983a49c36de488","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-08 10:51:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7d0860c4f60dc90442445f96b5ea653ba8f49004","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-08 12:23:40.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3f6a2235c322b7bfb90b05d489fa61b27d2f8fb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-06-08 13:01:22.000000000","message":"Abandoned\n\nNo longer needed .... https://review.opendev.org/c/openstack/cinderlib/+/845026 solves this by using override checkout to get the requirements branch we want.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6ccff2bf0e25673d38777c03639632008fb50f42","revisions":{"c68d0602a530595ee94364b44263e6693defa78a":{"kind":"REWORK","_number":1,"created":"2022-06-07 19:53:19.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/23/845023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/845023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/845023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/845023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/845023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/845023/1"}}},"commit":{"parents":[{"commit":"64af4766b2cf60fffe773fd4156f9e415b98dce4","subject":"Merge \"Don\u0027t limit use of importlib-metadata to Python \u003c 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af4766b2cf60fffe773fd4156f9e415b98dce4"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-06-07 19:48:02.000000000","tz":-240},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-06-07 19:48:02.000000000","tz":-240},"subject":"Add variable to requirements-check job","message":"Add variable to requirements-check job\n\nThe use case is to run the requirements-check job on a trailing\ndeliverable, whose master branch is still the development branch\nfor the previous cycle, and thus who needs to use the stable branch\nrequirements, not master.\n\nAdd the \u0027requirements_branch\u0027 var which will default to zuul.branch\n\nChange-Id: If2bfc21f2a06956a0142a6b056a151883348850c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c68d0602a530595ee94364b44263e6693defa78a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c68d0602a530595ee94364b44263e6693defa78a"}]},"branch":"refs/heads/master"},"6ccff2bf0e25673d38777c03639632008fb50f42":{"kind":"REWORK","_number":2,"created":"2022-06-08 06:33:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/23/845023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/23/845023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/845023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/845023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/23/845023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/23/845023/2"}}},"commit":{"parents":[{"commit":"64af4766b2cf60fffe773fd4156f9e415b98dce4","subject":"Merge \"Don\u0027t limit use of importlib-metadata to Python \u003c 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af4766b2cf60fffe773fd4156f9e415b98dce4"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2022-06-07 19:48:02.000000000","tz":-240},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-08 06:33:58.000000000","tz":0},"subject":"Add variable to requirements-check job","message":"Add variable to requirements-check job\n\nThe use case is to run the requirements-check job on a trailing\ndeliverable, whose master branch is still the development branch\nfor the previous cycle, and thus who needs to use the stable branch\nrequirements, not master.\n\nAdd the \u0027requirements_branch\u0027 var which will default to zuul.branch\n\nChange-Id: If2bfc21f2a06956a0142a6b056a151883348850c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ccff2bf0e25673d38777c03639632008fb50f42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/6ccff2bf0e25673d38777c03639632008fb50f42"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
