)]}'
{"id":"openstack%2Frequirements~845082","triplet_id":"openstack%2Frequirements~master~Ie3d81f180cc41a638ed17a0013dadfbd46d9819b","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-06-08 16:49:51.000000000","reason":"removed on reply"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-08 20:21:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3d81f180cc41a638ed17a0013dadfbd46d9819b","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-06-08 05:43:11.000000000","updated":"2022-06-08 20:22:26.000000000","submitted":"2022-06-08 20:21:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845082-openstack/requirements/constraints/noclob","meta_rev_id":"7bfc5c4452a294fb7f486a58188d7373a4d58775","_number":845082,"virtual_id_number":845082,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-08 20:21:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-08 18:17:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-08 18:17:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-08 05:52:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-08 07:46:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-08 15:12:02.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"3db74f1dc891d939b8721fff182fe629b7674089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-08 05:43:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"35ea8be9ed622b2f543357d744d0a4979be494a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 05:52:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/81f0096b3030402dbcb568b6f734dc4e : FAILURE in 5m 40s","accounts_in_message":[],"_revision_number":1},{"id":"df29bba181a503ae0fa62d310aa10d9aff08cfc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-08 06:28:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0227164f516ed8cc4fefbc4255fbf8b894679231","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-08 06:29:11.000000000","message":"Topic openstack/requirements/constraints removed","accounts_in_message":[],"_revision_number":2},{"id":"3e30de1659b2b928ab39d9d59b28db3c730408bd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-08 06:29:26.000000000","message":"Topic set to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"b31d5011ad79d613f425e15fd8fb86c484de9c11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 06:44:57.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/57c382662d86441e8e15feb7b224142a : FAILURE in 14m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f447f7696074063f9a94bc01de590a7dc97a705c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 07:46:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70a134c525bc4ed68cc001d95889c595 : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9434cab7a58451ebfe0b4aa4c096278 : SUCCESS in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7c98563aa5744a190328f755567d2ae : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09fe0d3e143d42b89356ab0e9427fa61 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d131f14e897541e0bbcb10c2da1d090e : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0383d63aece44f61b61a0b9bb546f52b : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a0b46434959c4ac9a650604f6d4f0791 : SUCCESS in 3m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ad846e5138db4e798b67c5db81a28aa3 : SUCCESS in 10m 18s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4ff2080825db44c0b43af61fe95f1171 : SUCCESS in 12m 27s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7b7018f186b44684a58962d1dfc2a240 : SUCCESS in 7m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4caaf786cbe24df1936694af646cdc66 : FAILURE in 4m 51s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/b2009fbacd2e4d888269f0b64106b6be : SUCCESS in 5m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3cecd3e24d6648fba28f3be79223a9f4 : SUCCESS in 9m 18s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6a1ec30e9abf444faaf4600cb8420bdf : SUCCESS in 4m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2d67defdda6f48f4ae5f94093ce458f4 : SUCCESS in 5m 25s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c296136244e44afa901db7f29a4912a2 : FAILURE in 6m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/369f444939f44cf8a1e90c67051195c1 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/33aed867fccd43df842605e2e545f7ef : FAILURE in 10m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3160f94cce644afdb32a46fe1bf7ebe1 : SUCCESS in 8m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/74efeb991ffb4a898ac48bf7cfbb6ee4 : SUCCESS in 10m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9d5930362a49427cb25d67e8211816b5 : SUCCESS in 3m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d8b50dcdf5bf48c7b677111c812e9fea : SUCCESS in 10m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/508501e4b9164ae4ba0c9df4344d4241 : SUCCESS in 4m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4a8f77c484a74429becc1e2361099035 : FAILURE in 5m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1a6647fe83ce469a88d409b129b0742d : SUCCESS in 11m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e55436dfc799448f897ba004adbe2663 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3e55faa6001f4cb79c578d34a2a3425d : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4421aaa889484e509cd47a417f9364b9 : SUCCESS in 17m 57s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7c4dd15130a547d29aed87c3ed489f20 : SUCCESS in 6m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f3237331ecdc4de6a0900175165ddb98 : FAILURE in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/677063b16e9d44b9868f8ef00adeca60 : FAILURE in 4m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0f5ea0e7fed54a45aa1036412ca562af : SUCCESS in 11m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da66730a1c5a42a2a133b6556bd3d23d : SUCCESS in 38m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ed4bf1caea494502957bdaf56e34a8e6 : SUCCESS in 1h 14m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cabf332175877bd99229e52561d2524e7998d316","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-08 15:12:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ab4f5102bbbfa733bb964edfaf24e7115a3da9c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 15:26:56.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/43dcae7c97d24acb93a9deb25a1457df : FAILURE in 10m 04s","accounts_in_message":[],"_revision_number":3},{"id":"3d9ad5cf890efb0ad228f96888641c8860070898","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 16:41:45.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8bc608f8b7b4394a702969baa01ec6a : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201d344dc1834f87a92cb90f6f0260f9 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bac821aad664fe189ab90daeb2e2dfb : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/255dc398072f47618c9d4520c0ec9e8e : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e55cf3c5753b4ad69ac759a862f686d7 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0821c82ab5bd4c4aa785f678abdb780a : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/89f96bdff88d4516adaa99acca0625a6 : SUCCESS in 3m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/97ef911aa67148d89ae1aeb00ed9088e : SUCCESS in 9m 05s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0a40aec721c8437bb8769688661dcb1d : SUCCESS in 16m 15s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/901279d158314fc49115c25f91f53b77 : SUCCESS in 7m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d420d48ca91e4768a67924e65c8092e5 : SUCCESS in 5m 09s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/71648c258ac546ac9e9c06a5654946b6 : SUCCESS in 6m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/64d37a96d6a84bbf9e8d540210bce657 : SUCCESS in 11m 03s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/aa0c15cde6884c799aecc6acde1e70ec : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/83df173170f84bb19101fd19dd60e300 : SUCCESS in 5m 17s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6242456362a240f69c1d815301e8451f : SUCCESS in 8m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/47c89833bd55406a8f658c432ee0cb46 : SUCCESS in 9m 49s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/acfc58cfcf58459fad80be6e560ca566 : SUCCESS in 8m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a3e54c23a2094cc09f137574f9db61f1 : SUCCESS in 9m 32s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4b882afe5cf7410db60a54a4c623384b : SUCCESS in 11m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/01b96e77338f4cff893bdebcdc987094 : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/81d5391546304af38c6a0a09dcdb289e : SUCCESS in 14m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2e06bd309f5c4fce8800dd4185d6d7f7 : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1c887eeb54504b8c9145e7bec574a349 : FAILURE in 5m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/012a3aa3a4d44ae1bd6c88f931d1487a : SUCCESS in 18m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f0861f559bd647459e49e4f84c3d71f4 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3f4a8431cbaf42d1a4bb9d926267d7d7 : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5e01bddb25af4bc99d4405425a8e8547 : SUCCESS in 20m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d44268b0568846618a06d5e94d32ef7f : SUCCESS in 5m 25s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/21dc947431ef493f964ea80025a9bc23 : FAILURE in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8c527a3f5a924575b844db3b5e8b4ee8 : SUCCESS in 6m 35s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1bd581692d99421697c2656068a9dc73 : SUCCESS in 12m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b1c89f4b9c49489c830722e4d3873cc9 : SUCCESS in 51m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/721e17e875524ac2b0d2828d7f3224ea : SUCCESS in 1h 23m 52s","accounts_in_message":[],"_revision_number":3},{"id":"9c24a5301ef770d440f38ea47d09a0b22a389571","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-08 16:49:51.000000000","message":"Patch Set 3:\n\n@mock.patch.object(pyroute2.netns, \u0027listnetns\u0027)\nAttributeError: module \u0027pyroute2\u0027 has no attribute \u0027netns\u0027","accounts_in_message":[],"_revision_number":3},{"id":"18c73f7deb019f84c757058924beb7e3e586770f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-08 16:52:06.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3f487501f4e2c82df5b1917545536b21401b547b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 17:04:50.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3e5f8eddd60049ba9014632c65149d96 : FAILURE in 8m 46s","accounts_in_message":[],"_revision_number":4},{"id":"7a5e4b67af480c46097f97e4368d3124059a983e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 18:11:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a2872ad22634bf39449a1eac8564678 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62d9db88156847268115ca15bce366ac : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ce9ceadb6dd4665961138f8a7d82028 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5709ae368e6a46e7ad4c843c232887eb : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a7b264a95864e9b9b70de5d79715e3d : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c122abf91b434c829dcddb3300b38259 : SUCCESS in 3m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce01568cfaf245848d30ce981b68429c : SUCCESS in 4m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/91655fd0c0124227b4d3f3cbde220a80 : SUCCESS in 9m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0642701b78db4667b36167c8dc64147b : SUCCESS in 12m 47s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ab05ab208c464dde85a9c4aae4870453 : SUCCESS in 11m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b7ee80bd15f14d6790328feecc0809a5 : SUCCESS in 5m 29s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/91df110a2bec49c697b3f7846df27328 : SUCCESS in 6m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/be5cb35779334b46bddff0666cf1d04a : SUCCESS in 13m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ebc1b76c5d094aed8c8d398dbe290f39 : SUCCESS in 4m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3d3ad70efd224449883be8a735028873 : SUCCESS in 7m 36s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/5604786a8f2c4e9eb9f98a7825504fe3 : SUCCESS in 7m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a11e5a37841c4a2ab6794eea5737413d : SUCCESS in 11m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b645eb475fbc431e9d2d8efd1e86bb60 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/51887b4df7604896914d3e5fe8584949 : SUCCESS in 11m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d607805f47a146b389889f24ad932804 : SUCCESS in 13m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7979447039564313a8e49b52fddfd2d3 : SUCCESS in 3m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0192d81e64c54ca28c037654a80723ed : SUCCESS in 15m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7d407daa92414429ad0c03f66026e9aa : SUCCESS in 6m 36s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/644a9be6a3c344eb91c7f2ee57774ece : SUCCESS in 25m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7cc10174ca1b4321b6b6cdce52053117 : SUCCESS in 11m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/345804bbdf6e41dcb8dda04c5a59915b : SUCCESS in 4m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bf8ffdf7b03d487d8646de62fc914f34 : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/06b55dce5cc64d9db040c90497d3a1e6 : SUCCESS in 16m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/adc11c7b59c14059bd909590b4c8efb2 : SUCCESS in 6m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5445d8fc4980427e8b03f64e13f4df60 : SUCCESS in 5m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9e87b97b579040c4bd776de1b697ba6a : SUCCESS in 8m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ecdf2c1cff4f449ab8927d637884727b : SUCCESS in 17m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dba357a680248cdab9a3bc4c9e49d8d : SUCCESS in 43m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c103a7cd34574f97a5c37f977c76ce0e : SUCCESS in 1h 17m 10s","accounts_in_message":[],"_revision_number":4},{"id":"f733fe3fbca6d0cb4fabe51fd66fadc39a9ce489","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-08 18:17:53.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"da972e62be07820c2243c4080be49dcef9ff2dc4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 18:18:40.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7cde405b27552e50cffdefd2b0692aaf1127cba7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 20:21:01.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ba377647b3d494db5ae102b5dc109fd : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb08aaa3c1fc4ccabd3a6464d912b108 : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e328d0df6d141229d5e17d0ff035731 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/102b0b867e0349008ead05fa390e4077 : SUCCESS in 3m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b710a9b14cf64a1e9f9f04d3f09f0136 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/00ebee20b13c4cc1997ac607a986bf7a : SUCCESS in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b8f6eb80fcb7433fac80e5701bc50e2e : SUCCESS in 10m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8c70a96e54b249bb9f7fb0f9803f6013 : SUCCESS in 12m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/54715ce790c8434c86aec12015952c77 : SUCCESS in 10m 53s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9a186771091d478d93754049e3c0439d : SUCCESS in 4m 58s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/417712f9abc0499db3a19510f7508416 : SUCCESS in 5m 56s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/492b935c668449afaeb8409117e246fd : SUCCESS in 8m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4a2ca19d2bc24635b41c9a4968172086 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/99d92ec70fb8498f8f9f82045b2160f1 : SUCCESS in 9m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9745a06d9b2947d49d34fee79acc95d5 : SUCCESS in 8m 30s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/38d850109e5049c48e75ac55573406e4 : SUCCESS in 12m 39s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b655d71e9d1f47d488fa7082225411e4 : SUCCESS in 4m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2a903315777a48c889b0dd4dbca7daf5 : SUCCESS in 9m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8e19e32c01dc4dbc9853266e4cebde46 : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c634fd154c3945f8a56754f7d091732a : SUCCESS in 27m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6778402419664a9e9e2ef0cea60b213e : SUCCESS in 10m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3de4a4c4f9ce45359326b05f9f76a087 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/390400f2602a40c1b747ccf38d305c9c : SUCCESS in 3m 51s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e246d3caec5a4330aae7482b093fd473 : SUCCESS in 20m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/9752e39ab0684a0fbae022dd0eb8a48a : SUCCESS in 6m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/08792825f22e4488b6e5052deb3fdf8b : SUCCESS in 5m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/10d2b0147f424d54a54f66927392d773 : SUCCESS in 7m 09s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6b9b2b56c220400db7ad18ea17054f5e : SUCCESS in 11m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f432c85dfed4f629cfb0537cb812235 : SUCCESS in 53m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa056c9f904445ae95680709ed68e4b7 : SUCCESS in 1h 28m 08s","accounts_in_message":[],"_revision_number":4},{"id":"917392b543535a60a4bcfef77fd6c4bb68fec0b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 20:21:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"7bfc5c4452a294fb7f486a58188d7373a4d58775","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-08 20:22:26.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f0ae0ee2dd54b19826d4a967f00df0c : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2cc6b9f8b5929b0f85c22cad8da2e64b87b538ff","revisions":{"b998099392f52aa612355a71f5943c3910994ccb":{"kind":"REWORK","_number":1,"created":"2022-06-08 05:43:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/845082/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/845082/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/845082/1"}}},"commit":{"parents":[{"commit":"64af4766b2cf60fffe773fd4156f9e415b98dce4","subject":"Merge \"Don\u0027t limit use of importlib-metadata to Python \u003c 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af4766b2cf60fffe773fd4156f9e415b98dce4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-08 05:43:09.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-08 05:43:09.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie3d81f180cc41a638ed17a0013dadfbd46d9819b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b998099392f52aa612355a71f5943c3910994ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b998099392f52aa612355a71f5943c3910994ccb"}]},"branch":"refs/heads/master"},"cb50833d586ddf0ec4880aaf127e86b08ed1b1e4":{"kind":"REWORK","_number":2,"created":"2022-06-08 06:28:59.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/82/845082/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/845082/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/845082/2"}}},"commit":{"parents":[{"commit":"64af4766b2cf60fffe773fd4156f9e415b98dce4","subject":"Merge \"Don\u0027t limit use of importlib-metadata to Python \u003c 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af4766b2cf60fffe773fd4156f9e415b98dce4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-08 05:43:09.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-08 06:28:46.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie3d81f180cc41a638ed17a0013dadfbd46d9819b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb50833d586ddf0ec4880aaf127e86b08ed1b1e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/cb50833d586ddf0ec4880aaf127e86b08ed1b1e4"}]},"branch":"refs/heads/master"},"533c06ea9fd04026852d845fe41542d54620d057":{"kind":"REWORK","_number":3,"created":"2022-06-08 15:12:02.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/82/845082/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/845082/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/845082/3"}}},"commit":{"parents":[{"commit":"64af4766b2cf60fffe773fd4156f9e415b98dce4","subject":"Merge \"Don\u0027t limit use of importlib-metadata to Python \u003c 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af4766b2cf60fffe773fd4156f9e415b98dce4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-08 05:43:09.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-06-08 15:10:57.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ie3d81f180cc41a638ed17a0013dadfbd46d9819b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/533c06ea9fd04026852d845fe41542d54620d057"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/533c06ea9fd04026852d845fe41542d54620d057"}]},"branch":"refs/heads/master"},"2cc6b9f8b5929b0f85c22cad8da2e64b87b538ff":{"kind":"REWORK","_number":4,"created":"2022-06-08 16:52:06.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/82/845082/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/845082/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/845082/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/845082/4"}}},"commit":{"parents":[{"commit":"64af4766b2cf60fffe773fd4156f9e415b98dce4","subject":"Merge \"Don\u0027t limit use of importlib-metadata to Python \u003c 3.8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/64af4766b2cf60fffe773fd4156f9e415b98dce4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-08 05:43:09.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-06-08 16:51:29.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nheld back pyroute2 for neutron and octavia\n\nChange-Id: Ie3d81f180cc41a638ed17a0013dadfbd46d9819b\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cc6b9f8b5929b0f85c22cad8da2e64b87b538ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2cc6b9f8b5929b0f85c22cad8da2e64b87b538ff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
