)]}'
{"id":"openstack%2Frequirements~845661","triplet_id":"openstack%2Frequirements~master~I4bd5a74aab1307fcee16fefb0f53926ab1293b38","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-06-16 09:56:28.000000000","reason":"Change was abandoned"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-14 15:49:39.000000000","reason":"Removed by Dr. Jens Harbott (frickler) by clicking the attention icon"}},"hashtags":[],"change_id":"I4bd5a74aab1307fcee16fefb0f53926ab1293b38","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-06-14 02:45:02.000000000","updated":"2022-06-16 09:56:28.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"848230bf522e0a0d392c36d5fecd4bc24f7696c8","_number":845661,"virtual_id_number":845661,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-16 08:53:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2022-06-16 07:29:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"date":"2022-06-16 07:29:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-14 02:56:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-14 03:27:57.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-06-14 05:00:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-14 07:19:24.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-06-16 07:24:08.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"a68f0e5ea03d457987475e4e52df189b6de57a6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-14 02:45:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8102dd2c28ac9f175a0b6701eedca955a064d95f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 02:56:36.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a8f5ba879f034f9a8ee3cc683d3dec80 : FAILURE in 8m 49s","accounts_in_message":[],"_revision_number":1},{"id":"dade322746e4b2a3d14112ed841836c4db8dd7a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-14 03:27:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e63f334971aca991b2988e3641c3e6352b3fa9f8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-14 03:27:59.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"88ea848927b4b1ee3e4aaa25e8a278126c3fd518","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 03:47:02.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a82f7fb1578448858b4b9c1fcc215527 : FAILURE in 17m 10s","accounts_in_message":[],"_revision_number":2},{"id":"b246fd00d8a74e1e7ef9a21177f58f99fbeba8aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 05:00:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4525dc727ea64b0389c89f48f7dc22e3 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04e0a27c2c994fd6baea093c59a8da2d : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9cd5edf32e44c10955c244152491685 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fca53b6d7ea948589a423480b1de7f82 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/41b008baafd34b65b19f5a4786aa79d3 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/86d457bcf47b4a31a7acaeb0e6ecbe89 : SUCCESS in 4m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b53cc442e94456ca5a5a148ecbb49e6 : SUCCESS in 3m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/09cd508dfc2447a98e2e868cc859a9dd : SUCCESS in 10m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0eac10a0eab643c2bcf6fb9622c53625 : SUCCESS in 14m 39s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7578fb4f62dd4d75b18476adbfdaf2a7 : SUCCESS in 8m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/46e6a14020344d35aa7a0d4b376e26a4 : SUCCESS in 6m 28s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/6001f02ec1924e51ad7a1e3da68243d4 : SUCCESS in 7m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ff1393be1c7e41c99ae86b477e8fdeb2 : SUCCESS in 11m 09s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/916de8225e54496b9397906d97d5c1c7 : SUCCESS in 6m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8db37ef7455a48689296aa3741c8224b : SUCCESS in 6m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/fd1bab6c9719430f8e70310da586a7a9 : SUCCESS in 11m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d5b8d68c4a984881a5b583cefba26d9a : SUCCESS in 10m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/39ae26251f2f4e67a9951ff1e515430c : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/311521efdbcc46e0a08ea2972dde0fd0 : SUCCESS in 8m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5b5e63d7e3964f1d9b0718f84779e2d1 : SUCCESS in 11m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/635e196f094a4feb8539bd87794490b6 : SUCCESS in 5m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6c80f57a7bc04ed2b52f01c0ec524dce : SUCCESS in 9m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d6320f6be86847d784d285c693dc88ea : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c217d83a69664925aad47be4398f5b9a : FAILURE in 24m 13s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/995f0ee614234bec9f05dfc12aecc11f : SUCCESS in 12m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bff8defcacfc4d4a94063b77d7058540 : SUCCESS in 5m 47s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ea954ac52f648419d178b2ab50e4ea3 : SUCCESS in 5m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f9d75c186dad481c9974b9d606ceebea : SUCCESS in 24m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/62f814aa6f9d4e768c3cbe6860adae0d : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b3997f3ae88047a7a9720f09c9748df0 : FAILURE in 6m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3c8cef11abc44dd4bf588ff9b69940e9 : SUCCESS in 7m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f5539e9fc38148d695e17afd9271b2f5 : SUCCESS in 21m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/525ba8ac746947dbaad61d3e534d6dfd : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3daa31b0f3f54812ab27124f100ed8a9 : SUCCESS in 1h 29m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6b241b700aeec49b6bd177be7b7a98c592987338","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-14 05:11:21.000000000","message":"Patch Set 2:\n\nhmmm, still broken","accounts_in_message":[],"_revision_number":2},{"id":"53940466155b1ba28ebb5e042e4feb0a77fa4567","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-14 07:19:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e5d13fe7b08ed9d1cef9b35c5f7cf605d7b62e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-14 09:29:47.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a08d84d85f28bda6664ab04e37528f474637745c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-14 09:30:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"518343edb3d1731d6ce66cd2f3d599fe74cd3457","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 09:45:16.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/074fb2bf0b0146c587681fb8995abb4b : FAILURE in 13m 45s","accounts_in_message":[],"_revision_number":3},{"id":"2be8f3debd1d8bb2edec8347dc2f4d25688d7094","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 11:17:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e9bcad946a347fe93c9f651e4ae4ff6 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f631b096d624f42b52bf0286fb0a3bf : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48b7d447f72c44f18e98d04fc9cb0f55 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/892044e54df1489db1f23aef5f742350 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06f69eeaca4a4c06ab7588ad5cc29c63 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5dfa54bedc1f4a24a16b5f6ac787b184 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f45fc0d82a5442a4bd28765209469496 : SUCCESS in 5m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8ce6770ff29847cf9a25dc094c551921 : SUCCESS in 10m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b6367d15e9b45fc9f59dad8c5e8c252 : SUCCESS in 14m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/1eee71d4cf1040a9b45a136572055627 : SUCCESS in 8m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a91b53bda49e4f1f9bdf0a0d285f32ef : SUCCESS in 7m 03s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/920c48dfdfa8447c9992a5cad91fc64b : SUCCESS in 8m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/30067736b6bc49be9697df483bb4c39d : SUCCESS in 11m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3b5b4601825743d8a6360554b635e9ca : SUCCESS in 6m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3e5929e53e224277a251cf75b9cacff1 : SUCCESS in 6m 06s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/79c9e3e8c44246de96410f179ab3e437 : SUCCESS in 9m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e01b90a54d7b491696c1dc26d2f05318 : SUCCESS in 9m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/32529771aea146f981af52e143e533e8 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4bf65ae35d03451e848782579719a6fb : SUCCESS in 9m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/db4caa611cc44ec3bfeb262a5c071c6e : SUCCESS in 12m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c8a8d4e24f0a4c3e98151c4ee17f4ada : SUCCESS in 5m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f209694801cc4f46b9cc028956f865a9 : SUCCESS in 12m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/78203d230617420d850f32cfe6266721 : SUCCESS in 7m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/338142176a894903a32bdedd65270e0d : SUCCESS in 29m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5d058ef80cc94a4ba602c61b5c51573d : SUCCESS in 14m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/34978d22a6c04a2eaefd3f9d4f9c7352 : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2604f5d0b89b46e297c9a96051d385a7 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63cd7a09fc8a45a79d14833d865d8371 : SUCCESS in 24m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/46ab913e788047f9bfa4a5cacd4efa36 : SUCCESS in 6m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2350bc08757b4ba9bbfd81ad6ecdbc05 : FAILURE in 6m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7ad8f14aafeb4994a4ed0c7496657632 : SUCCESS in 7m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b7c866451a284901a80add3e27b9d7bc : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1a3e444b5e984d96aa91082daf82fb5a : SUCCESS in 55m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/633fbcf6a1314b248308c1aedc33d8a0 : SUCCESS in 1h 41m 15s","accounts_in_message":[],"_revision_number":3},{"id":"fff60395df9f3f31d7200413c8342923e717cb0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-14 11:49:17.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"8ca82e939a7812c6dd3f10626ecb19fe72b5e4c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 12:04:40.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6fff65d7ff464326920614231dc530ed : FAILURE in 10m 12s","accounts_in_message":[],"_revision_number":4},{"id":"f7da4893188aa21c9d37dc75ac34d7bb104b3fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 13:51:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3b6071d57c4183ac6b95632ca547e9 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1cbbf4d3714b37974f9f0b9efedd39 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eadc833407dc4e79ab759e12926e0107 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32e0ecd638fa4dc8b56379b65a9350d6 : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c654174c6d014c44b8da7de6c78e2877 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/31a37e0044254cd8b7ab56c1e9873d00 : SUCCESS in 4m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6b3559650e634bb4b193f86dba5a43bc : SUCCESS in 5m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ef9edded28174ac5b0d17e7e813dcdce : SUCCESS in 11m 17s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0f511a44c5de4c71babb6628cdb16811 : SUCCESS in 15m 13s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f2b040d5c6aa45ce97aea36973caa0b6 : SUCCESS in 9m 11s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a107d199ab7a4b238ca2efc9be46778f : SUCCESS in 7m 07s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d20b10a4b33f4f23b1d078a080eb1df7 : SUCCESS in 7m 14s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c7b5237688e74bbd890848f2e6d45283 : SUCCESS in 12m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7b236a5b56b24099b35239434e1a61b9 : SUCCESS in 6m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/aa2d239cc7d54fd39837921ac5f26b75 : SUCCESS in 6m 25s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2eefd49e136041069283de6f35dfe34a : SUCCESS in 8m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dede7f8cdc6b4998adcc45a936f4378b : SUCCESS in 10m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/079286fcc32f4d15ae2191b8b792b8b6 : SUCCESS in 9m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4c1c9dcdeeee477ba329660a628215d6 : SUCCESS in 9m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/47bdc20150cf45cc881146b5b0e31b2b : SUCCESS in 11m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c31b2c0739434b068e0e8433b4b868df : SUCCESS in 6m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/925568208b924f9cb65bda107c824737 : SUCCESS in 10m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d574dd666913447e94aef33f32869329 : SUCCESS in 7m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a1b604bc7eca4a7d8a12f75f87c8d4f5 : SUCCESS in 23m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/77d870cad8e9474bb20310be04997bf8 : SUCCESS in 13m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/460e6cf8493a44179d36b153c9691184 : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/940ea0d8c9954fb99a0fcab3fd464612 : SUCCESS in 6m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c55fc07ebed54b128da3dfd763516a98 : SUCCESS in 20m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1bc3b44707914ffe8c369e1e643c5851 : SUCCESS in 7m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b60c52cdfee143409b8120255459d3fd : SUCCESS in 6m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1b6099574bdb458e8da80f90d2e80cee : SUCCESS in 6m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/68c684c347634461a96aec99ef2fc399 : SUCCESS in 13m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/529f38cccce849ac9b8ad6d60dbd2e57 : SUCCESS in 44m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f3f0d24cfd19496b858754db55871d92 : SUCCESS in 1h 25m 56s","accounts_in_message":[],"_revision_number":4},{"id":"8fd2fb45b05a4251546b4d91a4a5455150aded5a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:24:08.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ca4e060d42235dea7d5de02e3d23773b4c8ce82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:26:18.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a632a6242a9677964feedf74c305e694945c1757","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:26:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"74917f6220e44471abf9aa59f82473d719c9c029","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:29:31.000000000","message":"Patch Set 5: -Code-Review -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"61c3b859d2ac695f406299e2aee47c2fb4943a97","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 07:40:57.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3a266ef6176d496fb03315d721d440a4 : FAILURE in 9m 23s","accounts_in_message":[],"_revision_number":5},{"id":"aa1884f2dd3617ecc6ebdf8b480060d5dd79ee56","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:53:33.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aadc0aea8b84b8283f71e8a81afedb5 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61c86bc244974ad0a6ba28df15348af0 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/33f6b029959f4340b121529facbafdd3 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cb5dc17bba644b5a2427a18528b00a6 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4fe7c0b023ae47e2aa9ddcd9e16780f1 : SUCCESS in 3m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dc14c2848e024bc9a0ccf1247f018cb0 : SUCCESS in 3m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a68214ffe14545ac9043a69f88e22d43 : SUCCESS in 3m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/969ab710eb9646339d5c72a22eaca69f : SUCCESS in 11m 06s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/575da679f0e14fb599f12fd953f3815e : SUCCESS in 15m 06s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d7449ff2d9d0425a975ce13d853d2602 : SUCCESS in 7m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/83136c746742420f8b56733c14093496 : SUCCESS in 6m 53s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/21f2c84813ff455d9882da6057ce4588 : SUCCESS in 5m 58s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/28fe4de422184deabd0f2ea0ed0a5121 : SUCCESS in 10m 50s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/be44a8f5689b470aa2473fe362a73869 : SUCCESS in 6m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7928ed66d2024ec998528a30ea32d63c : SUCCESS in 5m 51s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/48a09564e82b4b79b92f140d45d82444 : SUCCESS in 7m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3750c15f07264c8196133e90dc4ef1e8 : SUCCESS in 12m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/adef6038f68e493e9570b99ec0627df3 : SUCCESS in 8m 06s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9b907d6a680544a9b393e7be45517209 : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/12700fd4f463479ba5ec8356ea3d0f6c : SUCCESS in 11m 36s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ed867661e868495c8c850f62b27222b0 : SUCCESS in 4m 25s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0105bc40b82245aeb6f2c5ac1fe083ca : SUCCESS in 9m 06s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b2a818d4c4504caf8f95aa3e9bb4ad52 : SUCCESS in 5m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fd4b25e7c8f247f79b7401fa2c6600c8 : SUCCESS in 27m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/34b1628ff95d460fb479dbd06cfba3ef : SUCCESS in 14m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/112ce47971c2495498e2cf53b2a4672f : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ead6f19a1bf642e3b2f3ae0117a689ba : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9a5b088a648744c3b3cec05b542daf59 : SUCCESS in 19m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/86ed3de6a7ba4cf982cd0b0d6400a60b : SUCCESS in 6m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/470756b2ff6841f5b121c8cc0941ceda : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2ab13003eaac4917ad39d64d141b9f35 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c5d62c8a66c94e179eb0dd255893afe0 : SUCCESS in 10m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cef31f292a9b44179f2272962060befa : SUCCESS in 45m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/753a1804d75c4d15bacfd4218b68a0a0 : SUCCESS in 1h 22m 39s","accounts_in_message":[],"_revision_number":5},{"id":"848230bf522e0a0d392c36d5fecd4bc24f7696c8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 09:56:28.000000000","message":"Abandoned\n\nreplaced by https://review.opendev.org/c/openstack/requirements/+/845871","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e0b7418d74dc75f57563192ebd5588196e506830","revisions":{"5eda7755f64c2605ed2442b087ca4891a4be758a":{"kind":"REWORK","_number":1,"created":"2022-06-14 02:45:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/61/845661/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/845661/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/845661/1"}}},"commit":{"parents":[{"commit":"44f2fa7205c39ac4307e02affdab53df401cfe67","subject":"Merge \"update constraint for XStatic-Angular to new release 1.8.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f2fa7205c39ac4307e02affdab53df401cfe67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-14 02:45:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-14 02:45:01.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4bd5a74aab1307fcee16fefb0f53926ab1293b38\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5eda7755f64c2605ed2442b087ca4891a4be758a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5eda7755f64c2605ed2442b087ca4891a4be758a"}]},"branch":"refs/heads/master"},"b7d476c00523aa3a6c8373c306f4e7ceebb85ea9":{"kind":"REWORK","_number":2,"created":"2022-06-14 03:27:57.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/61/845661/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/845661/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/845661/2"}}},"commit":{"parents":[{"commit":"44f2fa7205c39ac4307e02affdab53df401cfe67","subject":"Merge \"update constraint for XStatic-Angular to new release 1.8.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f2fa7205c39ac4307e02affdab53df401cfe67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-14 02:45:01.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-06-14 03:27:41.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I4bd5a74aab1307fcee16fefb0f53926ab1293b38\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7d476c00523aa3a6c8373c306f4e7ceebb85ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b7d476c00523aa3a6c8373c306f4e7ceebb85ea9"}]},"branch":"refs/heads/master"},"be8edbc0d344180c9bdd4a61374bd310ee83b7db":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-06-14 09:29:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/61/845661/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/845661/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/845661/3"}}},"commit":{"parents":[{"commit":"44f2fa7205c39ac4307e02affdab53df401cfe67","subject":"Merge \"update constraint for XStatic-Angular to new release 1.8.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f2fa7205c39ac4307e02affdab53df401cfe67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-14 02:45:01.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-14 09:29:47.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/845715\nChange-Id: I4bd5a74aab1307fcee16fefb0f53926ab1293b38\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be8edbc0d344180c9bdd4a61374bd310ee83b7db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/be8edbc0d344180c9bdd4a61374bd310ee83b7db"}]},"branch":"refs/heads/master","description":"Edit commit message"},"45f95a3445741916b8b8cb6ddf031ceff299f31d":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2022-06-14 11:49:17.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/61/845661/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/845661/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/845661/4"}}},"commit":{"parents":[{"commit":"44f2fa7205c39ac4307e02affdab53df401cfe67","subject":"Merge \"update constraint for XStatic-Angular to new release 1.8.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f2fa7205c39ac4307e02affdab53df401cfe67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-14 02:45:01.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-14 11:49:17.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/845715\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/845171\nChange-Id: I4bd5a74aab1307fcee16fefb0f53926ab1293b38\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45f95a3445741916b8b8cb6ddf031ceff299f31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/45f95a3445741916b8b8cb6ddf031ceff299f31d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"e0b7418d74dc75f57563192ebd5588196e506830":{"kind":"REWORK","_number":5,"created":"2022-06-16 07:26:18.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/61/845661/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/61/845661/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/61/845661/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/61/845661/5"}}},"commit":{"parents":[{"commit":"9c6e1b39d26efb3fabfbcfc1f37c41f207688afa","subject":"update constraint for taskflow to new release 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c6e1b39d26efb3fabfbcfc1f37c41f207688afa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-14 02:45:01.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-16 07:26:06.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nDepends-On: https://review.opendev.org/c/openstack/neutron/+/845715\nDepends-On: https://review.opendev.org/c/openstack/octavia/+/845171\nChange-Id: I4bd5a74aab1307fcee16fefb0f53926ab1293b38\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0b7418d74dc75f57563192ebd5588196e506830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e0b7418d74dc75f57563192ebd5588196e506830"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
