)]}'
{"id":"openstack%2Frequirements~845859","triplet_id":"openstack%2Frequirements~master~I98ffd87fc321ac44b296a19e2f2bdbb73c3ef228","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2022-07-18 23:00:57.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I98ffd87fc321ac44b296a19e2f2bdbb73c3ef228","subject":"DNM manually bumping jsonschma for testing","status":"ABANDONED","created":"2022-06-14 23:37:50.000000000","updated":"2022-07-18 23:00:57.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"c175c383ae335d1037db757516c1f063b86eaae9","_number":845859,"virtual_id_number":845859,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-15 01:37:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-14 23:48:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 01:37:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-15 01:48:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-18 21:39:22.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"}],"messages":[{"id":"f8edbc0360ce71ed97d5fb3e8ff07b94d7a84308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-06-14 23:37:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe854dd21723e69b8c634408a24f15f0b562f069","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-14 23:48:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/06ed295bb8244c16b48e2698c1049be9 : FAILURE in 7m 05s","accounts_in_message":[],"_revision_number":1},{"id":"e7cdea2598ee5e88a1c01600ea8e5ed41ed0b6a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 01:37:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f98d5f7b1e42489893a9b7feb493aad1 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffa46f1798594d6c83760e4aa44fdeaf : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e2aa224749c7492d8e1ef444217b4fad : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e1708b6f89f4147bf50888dcda51874 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/480fb97dfd0b42cfb97ce3af0d7b696b : SUCCESS in 3m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/09bb431e6a15457093264c475de905e6 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8014cc53be19429b9461977b5895fccb : SUCCESS in 4m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/27aacf18ee824cd983eec6e4fb20c3f2 : FAILURE in 5m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4107b30a73714fd594ca18f7cf56a5d2 : FAILURE in 6m 56s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/9c0cdb7200134616a0ae0977a786ed80 : SUCCESS in 7m 58s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e598c2d00d35400eaf5e5bc0bce7c3c5 : SUCCESS in 5m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c893276446eb4fc1becb74b21419a22c : SUCCESS in 6m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1e4cd993f709409e8963912f9cf5e13b : FAILURE in 5m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/75b6d1dff31e4af8b239f7b2e3086bbc : FAILURE in 4m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c4f010e2f67c48a8837cc9396c13faa5 : SUCCESS in 5m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/28bf8bc1e5e7419596744def7094f60c : FAILURE in 5m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2e273988d090499e8f094ebfa0b6acb0 : FAILURE in 3m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/70e50be9561c409a8ff6f3bb973e05bd : FAILURE in 6m 30s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3fea58ff9fd646489bdd3242560b7d20 : FAILURE in 3m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e99befd7c7c7471baf086d0e088aa35e : SUCCESS in 13m 05s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/25424fb694b84e9d8b2f410d6d756fbf : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/25d6e5342ba04b4b902a57cf9b758f84 : FAILURE in 4m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/97a98a0a71874239b981bbc4d65bc8f5 : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f04daf57512d4c6d8dc421361e85c0e0 : SUCCESS in 22m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fb098cd4a5d942128258af21cd1786d6 : FAILURE in 5m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/92837fac394d4e94885cf5ba77ef21d8 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/45cf1049fc5a40a0a013ae22ec4fef34 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/49a23a47b6cd460daf4a334917fec7c0 : FAILURE in 5m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/00b1b94914b24bbfaaed4134b4368b96 : SUCCESS in 6m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dfd376723c6447bdb0aa4dc552f169f0 : FAILURE in 3m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d61e00f98f214f39987629108cf8f7c9 : FAILURE in 3m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/af9bd52efd404541a043dc0b489cbfbd : SUCCESS in 13m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b57646bd45a4384a30fa7ce6e73e08a : SUCCESS in 45m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f1ef70419c754cc0a7a82174b5486e63 : SUCCESS in 1h 27m 21s","accounts_in_message":[],"_revision_number":1},{"id":"8e7e21559c958c76d1fbb355ae4a7ff92ac11f2c","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-15 01:48:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"56ce67971cdd600b1579ad2923c8cce42c0cee8d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-07-18 21:39:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c175c383ae335d1037db757516c1f063b86eaae9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2022-07-18 23:00:57.000000000","message":"Abandoned\n\nabandoned this in favor of: https://review.opendev.org/c/openstack/requirements/+/849102","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"042b56df33ea8bad01611d4188d4dd00d337982d","revisions":{"042b56df33ea8bad01611d4188d4dd00d337982d":{"kind":"REWORK","_number":1,"created":"2022-06-14 23:37:50.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/59/845859/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/59/845859/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/845859/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/845859/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/59/845859/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/59/845859/1"}}},"commit":{"parents":[{"commit":"44f2fa7205c39ac4307e02affdab53df401cfe67","subject":"Merge \"update constraint for XStatic-Angular to new release 1.8.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f2fa7205c39ac4307e02affdab53df401cfe67"}]}],"author":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-06-14 23:37:44.000000000","tz":60},"committer":{"name":"Sean Mooney","email":"work@seanmooney.info","date":"2022-06-14 23:37:44.000000000","tz":60},"subject":"DNM manually bumping jsonschma for testing","message":"DNM manually bumping jsonschma for testing\n\nChange-Id: I98ffd87fc321ac44b296a19e2f2bdbb73c3ef228\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/042b56df33ea8bad01611d4188d4dd00d337982d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/042b56df33ea8bad01611d4188d4dd00d337982d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
