)]}'
{"id":"openstack%2Frequirements~845871","triplet_id":"openstack%2Frequirements~master~Ieb57695815b5dd1478b8be5e878fb162de33d8fd","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob2","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-06-15 15:16:13.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"Ieb57695815b5dd1478b8be5e878fb162de33d8fd","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-06-15 02:46:15.000000000","updated":"2022-06-16 13:32:29.000000000","submitted":"2022-06-16 13:31:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"845871-openstack/requirements/constraints/noclob2","meta_rev_id":"dd1bc5dd222a321c8c715fb00823b02a16816f08","_number":845871,"virtual_id_number":845871,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-16 13:31:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2022-06-16 09:55:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-06-16 09:55:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-15 02:59:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-15 03:03:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-06-15 04:38:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-16 09:55:55.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"d6fedded99fcf942a8283fab43b5cf0533126e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-15 02:46:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8bb81595e10508ca9f27d36b879ee2db439cc87b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 02:59:44.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a80693a2bc0046929b51c6c013cdfe06 : FAILURE in 8m 46s","accounts_in_message":[],"_revision_number":1},{"id":"ea97e742cb8c7efc8b5308593f3cadb2feae22e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-15 03:03:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"978f0a547e3d7f43a6b9227133351edcf7579636","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-15 03:03:45.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob2","accounts_in_message":[],"_revision_number":2},{"id":"eda87be958bc14d7b72e73edcf5bb2dec8385d74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 03:20:28.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7fb870e34e234fe594d81ed47ae3389e : FAILURE in 14m 07s","accounts_in_message":[],"_revision_number":2},{"id":"597ab9367de8cac5389c4daa0f072464d087b1f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-15 04:38:13.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf3ffab5fef416aab5427678038cc80 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20118ec4c2a2475d8d29097e79efc8d7 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/842bc18326b44c4b91f811e8b9af10dd : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/591d6ae9e94241cca6eb79c500a13cbc : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9823f835f21441e78f60ba0300b04c31 : SUCCESS in 3m 27s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85c452f796664ce18e1f6c7d2de0f80b : SUCCESS in 3m 55s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/3e6ec9d303a44426a256f2cdf2bf88cb : SUCCESS in 3m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/068c09796c2e423d9bcc1a7506ce501d : SUCCESS in 9m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5ee17199ff214d15a910203f0800e001 : SUCCESS in 15m 06s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/c9007f1e87f1473280b4c686ce1e7659 : SUCCESS in 6m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8c857117140846ffb6db320383f67775 : SUCCESS in 5m 31s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c684f0eeef774b3aaf9b495af4411368 : SUCCESS in 5m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4e905b31d6644c1ea9f69fd9223680b2 : SUCCESS in 11m 27s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c1950effd25b4b218f81836c781e1e24 : SUCCESS in 4m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ecfdf73bb3b94865adc859a55badf4e2 : SUCCESS in 5m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/9c42df5a872a46e58458b96bdb7ddfbc : SUCCESS in 7m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/fcde87ab3f0d415d860837f90b3979c3 : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/012904d170bd47a8a9370c957e73ff04 : SUCCESS in 7m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/efab7848978f4a268f361217c315aef6 : SUCCESS in 9m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6c936cfee2734306a391621c6c8a6614 : SUCCESS in 10m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cebf46d56c7f4d3f99cdc7de0c9927af : SUCCESS in 5m 10s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c84c7c3c3dbf4699850076d4860f8def : SUCCESS in 8m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d6fe436e68c2481aa183576f26a2dc6a : SUCCESS in 5m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e8a3292988da46d1aa645bb8ed2bd534 : SUCCESS in 25m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f9a1d2aea13746cdbdb21a0f5a819ee7 : SUCCESS in 13m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dcbbd95c95984edb8d427c67f2d8d2c1 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6dd48321fcf348b5bd11f13825c21a8e : SUCCESS in 5m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/35404037ecb44b9282ef134d588abd7b : SUCCESS in 20m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f2485e39e99b48ec9264ae2d7631066f : SUCCESS in 5m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c3eea4a7b5ec4e889de159eb2740e244 : FAILURE in 5m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dbcada2b77e440379e3703bd74aea912 : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f8db2f3930a04cc7815b12468128b7b5 : SUCCESS in 10m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1b7ebad241224fcea01de8ec0a270685 : SUCCESS in 56m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7acbcafbfb264bcd822dd59d852021ab : SUCCESS in 1h 28m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d7b2b2d2967ee8731d583e82c7dc35117f2bcc86","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-15 15:16:13.000000000","message":"Patch Set 2:\n\nstill waiting on octavia","accounts_in_message":[],"_revision_number":2},{"id":"81fb3b07f3b64dd089a44bcb268b4898207811b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 07:28:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5389df196737dd4598d02e542ef494ad22d892bf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 07:45:28.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/347f37a9e1414562b0d6ece41f26d91b : FAILURE in 13m 37s","accounts_in_message":[],"_revision_number":3},{"id":"8104b03786016e74d522f52fdb50033b27bb8ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 08:59:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a076fe313f594b5883aa63415571bc30 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a9577b60878466195826f3c56e15d01 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9eeb5cc5b1d4e5fadfcba3d5017da48 : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb1dcf6f3a0c4b69af83bc6305ff2f69 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71c18ad6a9de441f94efd59350d7a242 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b780a68c1093438f8ecdd99f799fb2d5 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33ccfc0e6e84467d867d0a26173a3c2e : SUCCESS in 4m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/77b504606f124c9590ec806e1c4c086a : SUCCESS in 12m 39s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4f7d1bdfdb814567a78e4dc274e3e646 : SUCCESS in 15m 35s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f99bf5798fcc4de4b3de20f505a20a0c : SUCCESS in 10m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/efb4a908a67942719c84f9a2701d50e8 : SUCCESS in 7m 22s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/dd616beb262247d2a712e3957d6da52d : SUCCESS in 6m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/87d75f996d274c55bd80d25ae643ef21 : SUCCESS in 9m 15s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/39e168c442184b1fb12712eaac82efa5 : SUCCESS in 5m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6b929f59519947c2888306fbed5b9514 : SUCCESS in 5m 25s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e5b48e81cddb4c4ea0e2dcd5b6f99c6b : SUCCESS in 9m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/323de4aac2954146b3cdd884e248f295 : SUCCESS in 11m 27s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/675e23582971448b94b1f67fd58c7faa : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b127f87d617b4a57ad99dc212857eabd : SUCCESS in 8m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/19cda1f73c794182a96172886fcf1759 : SUCCESS in 12m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ff9404140ced4c86bc2c2687700cc9ec : SUCCESS in 4m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3c2eb15eebc64b148fc74a511944e8cb : SUCCESS in 11m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1cb5c84da3de4071b6c6b6ccbc5bab64 : SUCCESS in 5m 25s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f51928f335d649e4a1f38725466305f1 : SUCCESS in 26m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8e11b9c2e90545e1b1a2dd859a7612aa : SUCCESS in 16m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3c7eaefe0ae045e9838f82b946f83260 : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e95b275938b3471096f94fa98ceaad00 : SUCCESS in 5m 00s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f79297d4cc74161b897ee2488bb1a18 : SUCCESS in 21m 51s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2e5aa02016a74286b0ab8ead013d1225 : SUCCESS in 5m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/27fbc52f7b1e4fcf807b626d861529b8 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/af06512d88eb46d99cc1578e56fbf425 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/083a848bda0640f5bba6807a01003254 : SUCCESS in 12m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a11b9c49179b480fb0801c5338a1b805 : SUCCESS in 50m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dcd23d42686f41cabcbd92c03501c425 : SUCCESS in 1h 26m 41s","accounts_in_message":[],"_revision_number":3},{"id":"0414d055285dc66bf8da033faf0766704cf1ed23","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-16 09:55:55.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"981e0eb2e340cce893b937e1ea2bffab7c661d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 09:57:01.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"07c4a89509e7a198a86ef1093dc5c1f0e2a91844","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 13:31:09.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ffab155f8f4a608b737e3b41671aec : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3dc031f741b402383fd7bcf6fcdb9bd : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f91df5363640dbbc878beb46d66bc4 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/325b48261a9743cb85935d12005f78b3 : SUCCESS in 5m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/baeeaf84cdef4aca8aae489ce392532f : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/96e710574ba345bb8a175f070dd95877 : SUCCESS in 4m 51s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/4610725f685a43dab55b78b4f4128573 : SUCCESS in 11m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4b86a97c1d0c4813ac9b0ac3e702597b : SUCCESS in 15m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3dde14fba8054c4ba2313ede9bcf7bbb : SUCCESS in 10m 44s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1e7afb307a034194b381e2cf018de7f8 : SUCCESS in 5m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d612bb56d4784f218f2f4ab74d704b13 : SUCCESS in 5m 30s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8bc1ae30b6af42a39e6ec8378cdca4aa : SUCCESS in 7m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/13be694840e2449f83c8bbc577b803d0 : SUCCESS in 10m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/236fc4c040ca4d6d8a03d6321523a2a2 : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49b98dc8cfaa42309e6b09955aefce2e : SUCCESS in 8m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2dcb2b01d1af4116ab8f1341ba2daa25 : SUCCESS in 11m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5e2aacf7656349b09f950a41f91f9f8a : SUCCESS in 6m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/964d682dcc5c47b0aaf84562c807ae9d : SUCCESS in 9m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e261fed44a224e25833030905f5b22f5 : SUCCESS in 5m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/590d79e199604bf0afd0bbe624715033 : SUCCESS in 25m 36s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/96086cafa69c4a54a9904fffca425ec4 : SUCCESS in 11m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/561b79b1f7714368bdd0047680464952 : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ab92cc4288624ecc951b9cc378ff3d62 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/b90ad82ced324200a368e259a8572c9c : SUCCESS in 17m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/577aa392ef5142359bfc3349422b2046 : SUCCESS in 6m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f1f81d819ce14501ad5889efa36a2f18 : SUCCESS in 5m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b886cb7e4eac4600874c068700a3f2c5 : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/67d7679c35884469ba091b8bd158f465 : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e35ea1b6ee544fa1be469680665ecddc : SUCCESS in 40m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f8da2ba13b2749d881f74b44ace5d6be : SUCCESS in 1h 35m 31s","accounts_in_message":[],"_revision_number":3},{"id":"d71df77fd3445cffb0afbf6134a8423321486e99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 13:31:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"dd1bc5dd222a321c8c715fb00823b02a16816f08","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-16 13:32:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02dd2c31bec7466fbe5fa9e09d43e69e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"33a0385bbac3bd506593dd5479b5efb2565673ca","revisions":{"671874f5a779e57ab7260edf826b9aa30831dd53":{"kind":"REWORK","_number":1,"created":"2022-06-15 02:46:15.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/71/845871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/845871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/845871/1"}}},"commit":{"parents":[{"commit":"44f2fa7205c39ac4307e02affdab53df401cfe67","subject":"Merge \"update constraint for XStatic-Angular to new release 1.8.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f2fa7205c39ac4307e02affdab53df401cfe67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-15 02:46:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-15 02:46:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieb57695815b5dd1478b8be5e878fb162de33d8fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/671874f5a779e57ab7260edf826b9aa30831dd53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/671874f5a779e57ab7260edf826b9aa30831dd53"}]},"branch":"refs/heads/master"},"b1c4d9a6206cabd25456abe20c80f1c82c53e2e4":{"kind":"REWORK","_number":2,"created":"2022-06-15 03:03:44.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/71/845871/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/845871/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/845871/2"}}},"commit":{"parents":[{"commit":"44f2fa7205c39ac4307e02affdab53df401cfe67","subject":"Merge \"update constraint for XStatic-Angular to new release 1.8.2.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/44f2fa7205c39ac4307e02affdab53df401cfe67"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-15 02:46:13.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-06-15 03:03:39.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieb57695815b5dd1478b8be5e878fb162de33d8fd\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1c4d9a6206cabd25456abe20c80f1c82c53e2e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1c4d9a6206cabd25456abe20c80f1c82c53e2e4"}]},"branch":"refs/heads/master"},"33a0385bbac3bd506593dd5479b5efb2565673ca":{"kind":"REWORK","_number":3,"created":"2022-06-16 07:28:47.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/71/845871/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/71/845871/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/71/845871/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/71/845871/3"}}},"commit":{"parents":[{"commit":"9c6e1b39d26efb3fabfbcfc1f37c41f207688afa","subject":"update constraint for taskflow to new release 5.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9c6e1b39d26efb3fabfbcfc1f37c41f207688afa"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-15 02:46:13.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-16 07:28:37.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Ieb57695815b5dd1478b8be5e878fb162de33d8fd\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33a0385bbac3bd506593dd5479b5efb2565673ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/33a0385bbac3bd506593dd5479b5efb2565673ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
