)]}'
{"id":"openstack%2Frequirements~846277","triplet_id":"openstack%2Frequirements~master~I31f3323345a15c699327c3feff8b979504e052d4","project":"openstack/requirements","branch":"master","topic":"bug/1979031","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-06-17 08:38:11.000000000","reason":"Rodolfo Alonso replied on the change"}},"hashtags":[],"change_id":"I31f3323345a15c699327c3feff8b979504e052d4","subject":"Cap pyroute2 to 0.6.9","status":"ABANDONED","created":"2022-06-17 07:34:56.000000000","updated":"2022-06-17 10:36:19.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0c171ae9263906d54f719104d0006924e317bc39","_number":846277,"virtual_id_number":846277,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-17 09:04:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2022-06-17 08:37:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-06-17 07:39:12.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":-1,"date":"2022-06-17 08:38:11.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"date":"2022-06-17 08:37:31.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 07:39:12.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2022-06-17 07:42:26.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-06-17 07:45:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 08:37:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8e7f8f3a8d4afc7978d2c52d61cb6de65b0c73ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-17 07:34:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"976a36291ecac7546a2daae50649542e05ec3efa","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-06-17 07:39:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b38711e830c55595d1a2ff46a1ce0ab8c663ddd3","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-17 07:42:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"907c388d26f3c5f7de18362fdb0a5dec7f15455b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 07:45:09.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43e0e96a372b56f2ab2c0a53708dbf34879a9573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 07:51:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/de63741dab374cc4b02bd63ae208eee5 : FAILURE in 14m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f2eb7897deccd3b022a86e61c9894e8789dc9612","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-17 08:37:31.000000000","message":"Removed Workflow+1 by Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"1c79aeb7fd46a7050e18ffae793d296e8fb09b57","tag":"autogenerated:gerrit:deleteVote","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-17 08:37:33.000000000","message":"Removed Code-Review+2 by Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"282dec52c6ba88bb4497f2bdaa8ec41c12fafcbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 08:37:52.000000000","message":"Patch Set 1: Verified-2\n\nThis change is unable to merge due to a missing requirement.","accounts_in_message":[],"_revision_number":1},{"id":"b02a3a7d7a46a0f77476b440fbc424e568631bc6","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-17 08:38:11.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"785f39547416e7166b202dcaef6fa7ffe925ec6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 09:04:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef01ff976fc84cd3806f947fc4f3ef01 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29a1db236db54673ad6162bdc9392f3b : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/765ffc356f104dcf861e018c7cbaffa8 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/742475fc59b445539304530dbf2847a3 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26e8d843f26b40b09123b63cfcf4411d : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/877191a6b13849cab5f394f35bcdf760 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6511a754b74f44a297e8bdaf4f72f064 : SUCCESS in 3m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e44ab9fd9da64cbe8c165f1236444e20 : SUCCESS in 11m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e72a909fde2a4bc1bacc1cd0ebdec906 : SUCCESS in 14m 45s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a3f3ec442f174457afdbb2e853b6c6c2 : SUCCESS in 9m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5a1cbce9bffb43cea2dcb8402f7e87d5 : SUCCESS in 5m 39s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/75f80bb65256411f9d8c7ef3b1687218 : SUCCESS in 6m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/577ab3444dc24f55af0e0bbd8f157b23 : SUCCESS in 10m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/99fd7e295ab342b7bb948e69c658d7e5 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a79eeebb78b74200a443af336612ab05 : SUCCESS in 8m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f6945fa2148b49e08ec54b6d9da8ec12 : SUCCESS in 8m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0e43abc9a7724455bb6adcd09e49fdd0 : SUCCESS in 13m 07s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b47b33a866e44f449dbabdd8a380cfc0 : SUCCESS in 8m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/43edf69d73d344e3944c53ad5bac9a67 : SUCCESS in 8m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/06fb4677d66c4903b33ecd09dab111ae : SUCCESS in 11m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b5281ee4e23d4cc5a2a3c5d7573cc837 : SUCCESS in 4m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cef932a536344733aececf8308110412 : SUCCESS in 9m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b073b87ac2344ce891b1d2702f9801af : SUCCESS in 4m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/806be102a4c34e4d9cd97926d1b006d3 : SUCCESS in 23m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/da576867fc134b56965aa1376d9d81f9 : SUCCESS in 11m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c1370da5644749519c511dcf6f6133e7 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f472edfa6cad400fa4a0b163e2a1ef47 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/56d353adb5a244758cbc47fd11183a86 : SUCCESS in 20m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5afa945e085f45bb8667fd74efac9b8f : SUCCESS in 8m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3cc9c07d174443bba2344700f17b3640 : SUCCESS in 8m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e741abe3bac14c3982f5af90d31e642e : SUCCESS in 8m 06s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b65f1e673bd4435badd8205895f707db : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ece298ff237a4f44b3f5fc19129b1987 : SUCCESS in 51m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78d265f09ced42c2ab5166e7c26bb3ff : SUCCESS in 1h 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"a64ea8346215d6b552c99d6ea48345484458fc51","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-17 09:14:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68a4930292b6b7423d70419402d1c9fe1f5018a3","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-17 10:36:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c171ae9263906d54f719104d0006924e317bc39","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-06-17 10:36:19.000000000","message":"Abandoned\n\nCheck https://review.opendev.org/c/openstack/neutron/+/846286","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03fc36e2e323a029a8c90a293596ca0d7f92c11a","revisions":{"03fc36e2e323a029a8c90a293596ca0d7f92c11a":{"kind":"REWORK","_number":1,"created":"2022-06-17 07:34:56.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/77/846277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/846277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/846277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/846277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/846277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/846277/1"}}},"commit":{"parents":[{"commit":"46db00e91baa6ecd46e4dfb0b4dbdf185102b4a1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46db00e91baa6ecd46e4dfb0b4dbdf185102b4a1"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-06-17 07:32:53.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2022-06-17 07:34:53.000000000","tz":0},"subject":"Cap pyroute2 to 0.6.9","message":"Cap pyroute2 to 0.6.9\n\nUntil we fix the regressions introduced in 0.6.10, we need to\nlimit the maximum version of this library.\n\nRelated-Bug: #1979031\nChange-Id: I31f3323345a15c699327c3feff8b979504e052d4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03fc36e2e323a029a8c90a293596ca0d7f92c11a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03fc36e2e323a029a8c90a293596ca0d7f92c11a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
