)]}'
{"id":"openstack%2Frequirements~846299","triplet_id":"openstack%2Frequirements~master~I7c5c4669eb61ade89333ab0daab4b22cbf73b365","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7c5c4669eb61ade89333ab0daab4b22cbf73b365","subject":"update constraint for oslo.vmware to new release 4.0.0","status":"MERGED","created":"2022-06-17 10:17:17.000000000","updated":"2022-06-17 19:14:37.000000000","submitted":"2022-06-17 19:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846299-new-release","meta_rev_id":"cbb3641c86e844a60c70e7022a4578054018bda2","_number":846299,"virtual_id_number":846299,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-17 19:12:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-17 14:57:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-17 14:57:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 10:33:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 11:50:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 14:57:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d3407c232d5a8576ccc14c6bcbeccb436bbeeffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-17 10:17:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b38677e5a397c70848b9efb3333943cb5ae5809b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 10:33:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7c41878c5a764c36bbc0b1e45975746b : FAILURE in 13m 15s","accounts_in_message":[],"_revision_number":1},{"id":"ae8facbd7c116f79ed413dec01aedf9d3dc123b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 11:50:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d8a600ae0f64d1ba00d66d7f8405be5 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c350e8163cf148339b5a8fcbb756964b : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/82384ad35f4d4ac7aeb11996b1106aad : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c1f9641409e40cbbb6715834a3ea686 : SUCCESS in 5m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cd42bda14ea648d68d3b13fab5923db8 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d646c7f7853d496a83495050c65181eb : SUCCESS in 4m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c9e9dbb43684e7ca01dbd1c98c8feed : SUCCESS in 4m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/96a54ff7de714fa28d851fe14769947b : SUCCESS in 10m 30s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f2d9b33d4425418e9fdebea24e961c2f : SUCCESS in 16m 01s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/345d05fd47104eb5b3da2ba3f7da831b : SUCCESS in 9m 46s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c9e0a2248be54a0e9f4e9a0da1d58ac0 : SUCCESS in 6m 28s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/99d9ec13b0794dbe82475f3ba8b81d10 : SUCCESS in 7m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/cb51a0a9ad5f462584b59a674573a4e3 : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/57b7008982314fb399bc76cdea012044 : SUCCESS in 5m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/50ef1d782d614ab385c176a9a575d2b1 : SUCCESS in 7m 09s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0bde9e12ec334ca7a44565a9f990f61e : SUCCESS in 9m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/978b3a254a0c4b62b29a8c404b41dc0f : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/21779520fec54d7c9bb23bf6480d863f : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8a93916066aa458d99ab0fc1af318110 : SUCCESS in 9m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7a7b3eaf77f34f4881c48516d9efe942 : SUCCESS in 12m 28s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1121266031144cbc8e5d6c7af81a4a20 : SUCCESS in 4m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4a8abe0745ea4d80b7c77ebb4f5182fb : SUCCESS in 9m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a91b6f59f3fb4814b7f3f2404a25bf31 : SUCCESS in 4m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/612fe7c631a8487784d21731e61a6835 : SUCCESS in 23m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b94b947f624244429367de81b2be9190 : SUCCESS in 11m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/800e30a65a0d48c3818483bed183d21c : SUCCESS in 4m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c58ea46a2bd84bf5ba50548dfda58af2 : SUCCESS in 5m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb51e7d7412c4e689ac030c18d92e4fa : SUCCESS in 17m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fdb3f8da52b84afd8daf4eb963691744 : SUCCESS in 7m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a60425ffa5dd493bb89c17b45561fab1 : SUCCESS in 7m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/17a0312c01b74b6d988f4d19a86ed9cc : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b8b3ff7979594498a7d579f03ee66dfa : SUCCESS in 12m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fec10aa7f44d4c588811c7a75d2a672d : SUCCESS in 47m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b9aaac3461f4b7f9bb4c4b3acc9df56 : SUCCESS in 1h 30m 18s","accounts_in_message":[],"_revision_number":1},{"id":"0f5020b73edfda23863a2592c6abb110bbfbcab4","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-17 14:57:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77907d0fb1436bfafaa8a4f93b729b0dd22243a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 14:58:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"253b6919bbfde5dac195fe2ffbc5f5cde761f2fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:12:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf4787129914b10a0fc35f5b7fc8cf8 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f0df52de397409792729260197871f0 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5dfa1c0d7fa547c8bb2aa3590ef1326e : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cce9a25ffe84c14b31418b963fa9588 : SUCCESS in 5m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/256ff90b33e74648a347bb9eb4d4e264 : SUCCESS in 4m 11s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d5865bfddaa545cc8668437024c8f906 : SUCCESS in 5m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4a1e43881cc437c91ebc641da2cde84 : SUCCESS in 10m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bc5111e6baca4644ba71c682135690c1 : SUCCESS in 14m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9a00064d943b4686b3a914d1d7556aef : SUCCESS in 13m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0bcc258d85104b80a50bc3a1afe208f9 : SUCCESS in 7m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/762d6a787638434aaa170ac0eb7532a2 : SUCCESS in 6m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b9656ef547744ec1ab53bdb08430ff88 : SUCCESS in 11m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bef69e9ceef3498aabbe29ff227e0054 : SUCCESS in 9m 05s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66be4ab69de14290b5944567ed015c62 : SUCCESS in 12m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b934f95c6ce64e7781121d193977348d : SUCCESS in 9m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d3da336d84a8453083e134a41806f418 : SUCCESS in 13m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0e482497f83e4ad1aaa456e0ca6719a4 : SUCCESS in 7m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e3ffee0847f0453091c749e537439f2f : SUCCESS in 11m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f6ce9b78fb24b44a17819a9a3cff1b0 : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/81278ab9c1f94ca89a7636dd1e6cea79 : SUCCESS in 22m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2515b7e7cb544276aee793a30619b071 : SUCCESS in 12m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2c2d3e5a1701425d81f65eb9ae629ee5 : SUCCESS in 7m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/50b70aaf98de4a8aa6315b41499b9d46 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c37ba22da7104cc3bc59d01e441d4aa6 : SUCCESS in 18m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c58095de462a4379a63f54b25c7e8ac7 : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3e2ade97414b4300bf4f43237d5b17c9 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a264db5b15ff42a9992ab58b01820ec8 : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cd500a7a71f540cfb88b67332df42c45 : SUCCESS in 12m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bbcf50aef0484239995233f7cc4b9125 : SUCCESS in 51m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fba5485eb087470d96be06d336909f0a : SUCCESS in 1h 17m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b29c8021f5b43a36a8135fd83d24ea690d260b87","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:12:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"cbb3641c86e844a60c70e7022a4578054018bda2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:14:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00882efd4eb04c05a9256fdeaf7b61a3 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bc220272ac4924220888beaf209d158300e1497e","revisions":{"bc220272ac4924220888beaf209d158300e1497e":{"kind":"REWORK","_number":1,"created":"2022-06-17 10:17:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/99/846299/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/99/846299/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/846299/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/846299/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/99/846299/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/99/846299/1"}}},"commit":{"parents":[{"commit":"46db00e91baa6ecd46e4dfb0b4dbdf185102b4a1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46db00e91baa6ecd46e4dfb0b4dbdf185102b4a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-17 10:17:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-17 10:17:16.000000000","tz":0},"subject":"update constraint for oslo.vmware to new release 4.0.0","message":"update constraint for oslo.vmware to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I813f817dce190fa91d8a9aecdee242f7787505c9\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I7c5c4669eb61ade89333ab0daab4b22cbf73b365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc220272ac4924220888beaf209d158300e1497e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bc220272ac4924220888beaf209d158300e1497e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
