)]}'
{"id":"openstack%2Frequirements~846302","triplet_id":"openstack%2Frequirements~master~I61fae8f7de90f9fc892e841ebe7477ce33a49bc9","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I61fae8f7de90f9fc892e841ebe7477ce33a49bc9","subject":"update constraint for virtualbmc to new release 2.2.2","status":"MERGED","created":"2022-06-17 10:20:38.000000000","updated":"2022-06-17 19:17:47.000000000","submitted":"2022-06-17 19:15:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846302-new-release","meta_rev_id":"5b5e565f811103088e2c99f4340923c1fbb46143","_number":846302,"virtual_id_number":846302,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-17 19:15:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-17 14:57:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-17 14:57:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 10:35:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-17 11:48:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 14:57:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6e394f31474dd635910d4744a26e6e9bedae9ab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-17 10:20:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4acf2f37fad111a86d48e780684789e0ebd73d27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 10:35:55.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/4baf0a10ba434514b103f2e5ec5c5009 : FAILURE in 9m 49s","accounts_in_message":[],"_revision_number":1},{"id":"8b96f4ca01d4b1ce7462c6f61a1ddde9b2ef1e36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 11:48:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb889e6288644c159b282aefdd0423fa : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/706dc7ef3f2f460ba77c37d645134c7e : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/003ed772dc804877b96452b29189c70a : SUCCESS in 6m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44447804f6e94fa1803995c46050e756 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70a2f2adcf784537b095cf6e38413a27 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/620e17b438f348cf90cfaeacd7d30ad1 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c67f6a1d0d40491b96e6b3118432b99b : SUCCESS in 4m 33s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0466212472aa4a4682ee30ad04746d7c : SUCCESS in 11m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b279cd73d62840b68c14fdeb884b28fe : SUCCESS in 14m 28s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f088286b395248d5957a7b0fb420ce52 : SUCCESS in 6m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8c2885bdbe7e4f068f7d255b183425bb : SUCCESS in 7m 00s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/458774ef2b3244fd97de95f32b8ce376 : SUCCESS in 5m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9f95911074b14454a9b270e2ce4a9b59 : SUCCESS in 10m 14s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f1f14b5cef0144ab9ec1a23e99fbde3c : SUCCESS in 4m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8d001e9816dd4b9391f204d2ede93fd2 : SUCCESS in 7m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c81d98a4ef1744288c8d980758bba481 : SUCCESS in 6m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/303b7599bcda402b84682d10c5e0ea22 : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8df79a2bbb224b20b9884b3cf341bebf : SUCCESS in 11m 04s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9d4bd16f5ad246159750c42c365081f6 : SUCCESS in 9m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e228aa6065ee45058ff9b511b237b484 : SUCCESS in 10m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e2febf4e895d48a2b955bf9cf1788aea : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ba6c0c2ba7684c62b4ca1208884a5015 : SUCCESS in 10m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b530ce90f6574850943841d853b2b9fc : SUCCESS in 5m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/019c0719dc9c468e8676787f9a5e5a6c : SUCCESS in 24m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ff536c06af9840a0b10c663381021c58 : SUCCESS in 12m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/69bed4d4c34c40b7a820c374e11892ba : SUCCESS in 4m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d24ae475051f428ea43d1dc785a9fa64 : SUCCESS in 4m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a349f90fdca24e52b62ef506f337a6c7 : SUCCESS in 19m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/23f39c6fa8084c6791ba5ed21175f38b : SUCCESS in 5m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6509b14e13ec46368a4dd2693b377443 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3ff4319d527640aaa86eb472ad617c97 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8e1bb3b5dffc48399e2b1213260a4797 : SUCCESS in 13m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93246fb825a0479c935d426a7816cd3f : SUCCESS in 46m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a59215c8e58c45f19399388e02f7b3f4 : SUCCESS in 1h 25m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3c0b2dcb04f971f1a8c2fc59c511712c87b9ce05","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-17 14:57:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c3fde4dc463d84633c2ac40b200958757b7f61c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 14:59:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ab556dddefc4ccbaccf2c9602faf9244d8d6d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:15:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf3d1b0e9224195b4119cabb7857519 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69e6cdc0d33443f2bc20ef81eeff29e0 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83b7ffb24e014c7986998678f45e8773 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4bba904f21143d598bcb6e56145aed4 : SUCCESS in 6m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a12d5ecbf71f41c7944cd9c209321d8d : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/360304bcb1774731b5da0d56366c6ba8 : SUCCESS in 4m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/79194b56ccb74a80a047f1244f93a83c : SUCCESS in 10m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/530f0aaba5194fc6ae170b79d25c62c9 : SUCCESS in 14m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/be98f730db71493789bf97ad5f74c319 : SUCCESS in 10m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/de4a396860494b71b70ee0c6e62e18b4 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/03e87660273444918d1c824697945b03 : SUCCESS in 6m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/583da9b4d0ef4a11a5b69fc13f50d9af : SUCCESS in 9m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2c302e6d8fe145f0935f1eb179b41800 : SUCCESS in 12m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/4b3b5594a1a14457bddce4cd46721e6c : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4650462455df4a47b3edfdb6a5b8d546 : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/116574e758c449f2af544fa3d56a5752 : SUCCESS in 15m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b877fa2133544747956082887ad0d3f4 : SUCCESS in 5m 30s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7d9c33758f2b463e85b0f2654375f5b3 : SUCCESS in 9m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b4c934b594ee457aa3205f1de9b55c41 : SUCCESS in 6m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1816251f8976497fbe8029610f34575b : SUCCESS in 23m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c2dacea365354381a4b99699e0a69fea : SUCCESS in 13m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6ae303761765434480a3971efa3b77d4 : SUCCESS in 5m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca4fb482d06745bb99e79cc17b1ecb08 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f86933a634ba408fbe0ff66a003a70e3 : SUCCESS in 19m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c0dbda4ee41441ba88ada7bb1468722a : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/64d3d390309f48b3b743ffaa73a4017c : SUCCESS in 6m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1e73c3ce8e0b4a0083325f882b046887 : SUCCESS in 6m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ca12fe57517147b3b741946652a5aa17 : SUCCESS in 11m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58e67632b1b549ee8f0bd844eca129ec : SUCCESS in 53m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/92b631bc09e5418b805d1b04bf74414c : SUCCESS in 1h 14m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c1362f4d7cea5660f87119cb1749b05d38797223","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:15:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5b5e565f811103088e2c99f4340923c1fbb46143","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 19:17:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fded9c8d532a4b249260719299746908 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57031c3618f3a7431644e79bc7a9fd3865726e8b","revisions":{"57031c3618f3a7431644e79bc7a9fd3865726e8b":{"kind":"REWORK","_number":1,"created":"2022-06-17 10:20:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/02/846302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/02/846302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/846302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/846302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/02/846302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/02/846302/1"}}},"commit":{"parents":[{"commit":"46db00e91baa6ecd46e4dfb0b4dbdf185102b4a1","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/46db00e91baa6ecd46e4dfb0b4dbdf185102b4a1"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-17 10:20:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-17 10:20:35.000000000","tz":0},"subject":"update constraint for virtualbmc to new release 2.2.2","message":"update constraint for virtualbmc to new release 2.2.2\n\nmeta: version: 2.2.2\nmeta: diff-start: -\nmeta: series: independent\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Commit: Riccardo Pittau \u003celfosardo@gmail.com\u003e\nmeta: release:Change-Id: Ie292a2d83b9e67a1f3b5bdb2ab629665668261dd\nmeta: release:Code-Review+1: Iury Gregory Melo Ferreira \u003ciurygregory@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I61fae8f7de90f9fc892e841ebe7477ce33a49bc9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57031c3618f3a7431644e79bc7a9fd3865726e8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/57031c3618f3a7431644e79bc7a9fd3865726e8b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
