)]}'
{"id":"openstack%2Frequirements~846475","triplet_id":"openstack%2Frequirements~master~I3c8675b6f0e78e67b538f2e7e9144aee58b4668c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I3c8675b6f0e78e67b538f2e7e9144aee58b4668c","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-06-19 02:47:27.000000000","updated":"2022-06-19 05:12:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fb8d99bdab3f80025e225692ec1dc30d57550283","_number":846475,"virtual_id_number":846475,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-19 04:05:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-19 03:03:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-19 04:05:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"459634e050d2aa608ac410b2d45883bd8126e510","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-19 02:47:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"515ed0f90e6b829432f7542309b9d8dfb0f9784d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-19 03:03:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6daeabc210a3489f9a9ad7614f2a5bad : FAILURE in 12m 01s","accounts_in_message":[],"_revision_number":1},{"id":"07f1adcc57e3a4986eadb31c386cf21f8c5ef754","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-19 04:05:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f86ee10ee5c4b8da62a9c7fe2626e94 : SUCCESS in 3m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cceea84e56b41268301dac37e172331 : SUCCESS in 3m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f5729edb1a94944b6386752003ed005 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/419eda730e574f1ab72cda16f70572e2 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bbb26350bd04840a90e9866965ac633 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0325e4b936fc4444af6a6e04fc779d16 : FAILURE in 3m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9c5ff6a851494b4fa2f380ae6dcbcec4 : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b5c9253c2326404fb36f74ca8920adea : FAILURE in 4m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5b1a4a63863d479b8600c509756f4013 : FAILURE in 4m 34s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/6120ea22b4db4687bbda849a5cbf51d5 : SUCCESS in 8m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d80a80a1e58040c787d7cb34b30ad8e7 : FAILURE in 5m 09s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/4506c7c2e02241e885d192114e8ea45e : SUCCESS in 6m 48s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2652143039e5477fa89a1793f797200e : SUCCESS in 11m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1bbca9c8eedf48f48931c865f1396438 : SUCCESS in 5m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/017d228fa43b49f6803a873b7a556a99 : SUCCESS in 6m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3f4f9f82bd564f0390f5f5587b12f63c : FAILURE in 8m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0dbfa23534de41d7b354e9d21a6cf207 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fa26d8ac27844a67a074291bb0317e36 : SUCCESS in 8m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/60cb4350e4e04008a35b8ea65827beed : SUCCESS in 6m 37s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2ffcd48183fa4bb794bac1d20604be04 : SUCCESS in 11m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5471b34a28544409a4776a6983247711 : SUCCESS in 4m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c67d79eb780b4af3a7f72a6c85e1a37d : SUCCESS in 8m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/76fdc58c45f24b25b08a049cf268cf10 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/41a8908189b64689bc7e62bf385c4bf9 : SUCCESS in 26m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/216446bccca64104885799d8e4e27d79 : SUCCESS in 13m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c48dbd9a57d14eeabf6262726a742525 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fc7cf5b25cd240198d4d191e33073d63 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3c900bede5754d7e81a912dba9a029fe : SUCCESS in 20m 50s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/85d436aa69bd45ff8306d7cd2ea2d81d : SUCCESS in 5m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6708a4a2600846cfaedd647e84f49705 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6fdfa386c6fd4641be2dc7fc13951298 : FAILURE in 4m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6a700e9d7d35499eb161dbfd7395fd09 : SUCCESS in 14m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2a3bf542a4a49f29ac703f480556d44 : SUCCESS in 32m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83f013aeacc847a1ba1a85cf7618581b : SUCCESS in 1h 12m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fb8d99bdab3f80025e225692ec1dc30d57550283","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-19 05:12:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c032747a2c291b09b42fd32df5432dcfbef77974","revisions":{"c032747a2c291b09b42fd32df5432dcfbef77974":{"kind":"REWORK","_number":1,"created":"2022-06-19 02:47:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/75/846475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/75/846475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/846475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/846475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/75/846475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/75/846475/1"}}},"commit":{"parents":[{"commit":"264c424c399b0ef931a729caa3aa76888e5727c7","subject":"Merge \"Skip networkx 2.8.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/264c424c399b0ef931a729caa3aa76888e5727c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-19 02:47:26.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-19 02:47:26.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I3c8675b6f0e78e67b538f2e7e9144aee58b4668c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c032747a2c291b09b42fd32df5432dcfbef77974"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c032747a2c291b09b42fd32df5432dcfbef77974"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
