)]}'
{"id":"openstack%2Frequirements~846785","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I7b09604e8b983c440ffc2e7a9e493a149fc6990e","project":"openstack/requirements","branch":"stable/yoga","topic":"yoga-neutron-uc","attention_set":{},"removed_from_attention_set":{"15554":{"account":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"last_update":"2022-06-21 12:45:11.000000000","reason":"Change was submitted"},"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-06-21 12:45:11.000000000","reason":"Change was submitted"},"21798":{"account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"last_update":"2022-06-21 08:44:16.000000000","reason":"\u003cGERRIT_ACCOUNT_21798\u003e replied on the change","reason_account":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"}},"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2022-06-21 12:45:11.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2022-06-21 12:45:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7b09604e8b983c440ffc2e7a9e493a149fc6990e","subject":"Add neutron to stable/yoga upper-constraints","status":"MERGED","created":"2022-06-20 16:11:05.000000000","updated":"2022-06-21 12:45:11.000000000","submitted":"2022-06-21 12:45:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846785-yoga-neutron-uc","meta_rev_id":"c3bce8f24772c48f54fbb24eb61078f68cf31ad2","_number":846785,"virtual_id_number":846785,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-21 12:45:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-20 16:58:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-06-21 08:44:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-20 16:58:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"}],"reviewers":{"REVIEWER":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-20 16:14:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2022-06-20 16:14:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2022-06-20 16:14:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2022-06-20 16:14:33.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"state":"REVIEWER"},{"updated":"2022-06-20 16:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-20 16:58:48.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-06-20 17:47:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f1b36f7ebf22e65014b7612322f5b6f2e592356c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-06-20 16:11:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4d994b500219fbfcf1f217cc8532d66c3b0eb053","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 16:22:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3a39cad2f79449f0b1b6c9dc189a4109 : FAILURE in 8m 17s","accounts_in_message":[],"_revision_number":1},{"id":"9655f03aae6fc3306e1cb3b4a9c7f797076e614d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-20 16:58:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nneutron makes me sad :(","accounts_in_message":[],"_revision_number":1},{"id":"2311c8a5103710eb15b0aae9821431b43174d2e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 17:47:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8170b22d99f4bc388c355360d9a89f1 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/51ffd197826f487aa7190dda5e4797a1 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd380a40392a437786803c3887966221 : SUCCESS in 4m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b99072c5cffa40f6953545f8d3900be0 : SUCCESS in 3m 46s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f2a7cbaa3c60418dabdc910c9983e7ab : SUCCESS in 3m 32s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/88ad7cbb05fe4e0f87883dd531f31340 : SUCCESS in 9m 08s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e5d518f1d90f427f9b82d473b646ade1 : SUCCESS in 10m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/117a1b2619e6446fbd7ba472006bb4e7 : SUCCESS in 6m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2b4035987c2c4d589f9dcc94542f3c41 : SUCCESS in 10m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fb3658ad74224a879ce2211bcba0324a : SUCCESS in 5m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f53e040f002542cab6a70be092610b48 : SUCCESS in 5m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/33791b8a96bc4e64b5e0dabd4e394a76 : SUCCESS in 10m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79743223772942caa0caf81c4a5ee1f5 : SUCCESS in 8m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/85e7e8ad79c44712a0423567bf65ccd3 : SUCCESS in 7m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3969a1af88704b4490e94125767d1188 : SUCCESS in 11m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4315a9c4be4b46d7b8ebf5887cab255d : SUCCESS in 4m 18s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ea695bd66b1e486a9d1ddfd3a16557d4 : SUCCESS in 10m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5c341c62c9574974aa02a054cf2e587e : SUCCESS in 5m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7dcb3f0ccff8439aa68ed2adea322ad9 : SUCCESS in 23m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/65c87a85b9f5448d8471ba1fbb559872 : SUCCESS in 13m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8f01e593c5ad47b0bf82f64479936ce7 : SUCCESS in 5m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/705ce9cbcd684475b9d9ee120d901df5 : SUCCESS in 3m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/17f909fc6dce47658de0d2ae0397c89b : SUCCESS in 18m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/eb27be877a334a4a87698acde828e3a2 : SUCCESS in 6m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/23cb61c9abda4b5383cf196294f2df29 : SUCCESS in 4m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/079db26ce3f241809e8f87dd4c0966b2 : SUCCESS in 6m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/6b43998344cf4eaba2c2b4850ff84824 : TIMED_OUT in 40m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f15296d9cc84bfc86d34b597a19013e : SUCCESS in 49m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c39e3d6da664ebaa101e5265d48299a : SUCCESS in 1h 31m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d656fee7d88095aec3d671c69d78b8b7b8f3cd78","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-20 18:22:35.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5b93aaef8675d0cd52263fd99d9a4e67826433ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 18:34:31.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/794d75831ce14006bbaf9bd2e55d5e42 : FAILURE in 11m 28s","accounts_in_message":[],"_revision_number":1},{"id":"60d0b99a04560e7b0b3d6df0791bd12b7604363b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 19:57:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/853e498603bb463abcedeeb84bbbe7e0 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b99dfc8de75049ff85599f08b75fdd1f : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c6459a9b5484c8faa4f2d1424caf1ac : SUCCESS in 4m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a7d7273a414f401d811630073e596573 : SUCCESS in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a17b498b492d48aea3fa9ce307ca6373 : SUCCESS in 3m 44s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/5ff9b99952444a7eb4849729e5b54067 : SUCCESS in 9m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/01634e7a1b6d4d1cabdeb964e2941d15 : SUCCESS in 9m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1695d5521dcb4c9988e6315b7f3d9b86 : SUCCESS in 5m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/057e3e6ef0a24171a3e86b6725c0ba25 : SUCCESS in 10m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/372054d5c1a341f8bf5830f92e5432eb : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c60faeabd9e149829f6eb0d33909621d : SUCCESS in 5m 21s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/691f51fbf73a4716981a76f0017a56ea : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3de98ab11d2a48c288c567f5d451dcbf : SUCCESS in 9m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/04998ef30c2e44c9b19ae5ca2f8ffa95 : SUCCESS in 8m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/84de8d15859e45e28e2a6ea648287258 : SUCCESS in 12m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dcd8cdf1f72d42a1b23352168449acae : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9812998810df4a5ebc58be613849dacc : SUCCESS in 11m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dcf70050b56247ba8f483dc2f9f36c77 : SUCCESS in 5m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b12d068fb4e54aa898321eb8c00b6b42 : SUCCESS in 20m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/702c92cceae7429aa43b2b54d8c74680 : SUCCESS in 11m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ae0ee61aed2a415ebe171ea39357f740 : SUCCESS in 7m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2b8779f8a54a4196a8391dde1cd7ee1e : SUCCESS in 3m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ff5697d5ce6b4798be042e61cb0075c3 : SUCCESS in 18m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/047b2e0e2ea14749bfaaeea50472f88f : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fbbeb003425941efa3d66a725e00a08a : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4a5b9fb1e07348af8aa6b39d4aa3dfc8 : SUCCESS in 6m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ef42680c7d6e4eefa55bf9ffb917af39 : SUCCESS in 8m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e65efb529eb424796d9a6ad3462dbc4 : SUCCESS in 55m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a9d06cc6e1a9495ca207e33d661b6771 : SUCCESS in 1h 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4149030c87945cb5b3ea876dbaec5331c09a9115","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 19:58:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06024964f08fdfd5ca08df1d15e3fbbb12f584c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 21:21:40.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1a55d27067424b97eaad9d9c0e104b : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b9e79de5e5f4a9db4a6f730a4177608 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fe250237f4f47a880ca2bd5ce9a4d77 : SUCCESS in 4m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eb9c72bc66c743f2aa079db158782021 : SUCCESS in 3m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4395b4433aa4de5a42693ea9cc797cb : SUCCESS in 3m 57s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/93b288ac631d4ab7948d47108e5a36a1 : SUCCESS in 8m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/569f711552ae4edd96c103a7bdc0c6d6 : SUCCESS in 7m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a9d26ff4ba89469ea76b87e2ddf72344 : SUCCESS in 9m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6f4c8d3700f642de9973d83f8b43b452 : SUCCESS in 4m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/730f78281f294abf965fe04f9911a238 : SUCCESS in 4m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/299618a14a644690b79f0b1a9c070d7f : SUCCESS in 9m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b7e07fdda48b46069e5b6174542b44c5 : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d836f6cf949c4befa020a87b473244a2 : SUCCESS in 6m 48s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8ff98faac0364b7f8e13522147d08cf9 : SUCCESS in 14m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/80a891b083b24b178a395455be183763 : SUCCESS in 3m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b7d3cc9e3bbc4fa79db416d985f624cc : SUCCESS in 8m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b8fdc9430b6840a3830a0ed3e6360867 : SUCCESS in 4m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9a1ea63f90674937b69ee809082332f8 : SUCCESS in 20m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/04b50dd2d26845f2ba6064eb70e31bd8 : SUCCESS in 11m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/72b38ad0f9e843b3aacb8e2ed08baf02 : SUCCESS in 5m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/75af5abfc5484f4ebe92c9288c131724 : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d60ed0e4fe74ff8891feedcf5cc9c54 : SUCCESS in 19m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/61b943e93cd44447b35cf81113fa5598 : SUCCESS in 6m 01s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/03a7635ee507499199d7fb81b3a11e21 : SUCCESS in 4m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/04153d5891d24cb185df5f472bb01958 : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b48996c6279740ee8b9a04c351e9e65c : TIMED_OUT in 41m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/db1c3cd18a2b48adba6948121bfc922b : SUCCESS in 45m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5929c5361e24bbe99172f0e51e45caf : SUCCESS in 1h 22m 45s","accounts_in_message":[],"_revision_number":1},{"id":"5ac1ba7c622eddf2aa71007905a3e35c936919b6","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-06-21 04:44:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae5a463e11d7fc636f36b4fe99b38f0d81bf2eed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 05:54:24.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/776a018eef3f4d00bb516a56a26aab16 : FAILURE in 17m 03s","accounts_in_message":[],"_revision_number":1},{"id":"54e63e282916a95d085c62c288f98dd28d9e9f54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 06:36:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/774a339477a14e9fb332f2016f971fcc : SUCCESS in 3m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/404a00f2eb5f423c9c14aba7f774470e : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08db04e004f74df6a365f883f63bd467 : SUCCESS in 3m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/adce4338b72c43298aa734b044991649 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0c05b36e14a541cf85be9c556324d780 : SUCCESS in 3m 23s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/e8c3eef504754c9b9cb8717b8df53b3b : SUCCESS in 11m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3d45c2e69bba408790454cd77671bb7d : SUCCESS in 9m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f69b1e92a406489a976d2a5e23fecddc : SUCCESS in 5m 11s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2be5762fb96e4c85bd1585b60389cc8c : SUCCESS in 9m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7c21552c8ad74c79bb880c7b66f8550e : SUCCESS in 4m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3b769ffde5854e4585e8f11bd8fb3468 : SUCCESS in 5m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a5108063cd5a45c797a959f2b039d670 : SUCCESS in 10m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6f500da2e7a34330846a6a9f623daa26 : SUCCESS in 9m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ba82061157064da1822fd0609bfc8c0c : SUCCESS in 8m 22s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/58959bf21db74cf3b355c03bf94d8f80 : SUCCESS in 11m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4f28fce11a344bc18542c03c0775ff7b : SUCCESS in 3m 58s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b6d3176bb3f143df900112f13c16a58c : SUCCESS in 9m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1e3158452fd84c05819208fd5e5ab56b : SUCCESS in 4m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/868518e4d6bd42d797fd64fb98f4711d : SUCCESS in 30m 09s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a269f421fdfa4bed9320e2a72be710b3 : SUCCESS in 15m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/182a20075bd14037a9e926c774e8ea17 : SUCCESS in 5m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a2f8f840951c4a74826eaa7fc4786f48 : SUCCESS in 3m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cc72985306b94543906345af536255cf : SUCCESS in 24m 56s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/491cce6834374c5290e6d38d7f48513e : SUCCESS in 6m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/07a03660819f4109ac3b620378d37045 : SUCCESS in 4m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7f290d2cc2874befb52a38ba56a0fbf1 : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7bdfbcb871ee4595b4d1b6d0868ea2af : SUCCESS in 11m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a77f181c1da4424195d3076dce2fb37b : SUCCESS in 57m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5710210acf44eeab3398d5db2ce9fd3 : SUCCESS in 1h 45m 41s","accounts_in_message":[],"_revision_number":1},{"id":"a50de74b13a7e62b25527c51a4bb027250a22315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 06:37:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8eb1770261276c9859b959bae73a2fb92e0a651e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 08:07:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c579dd7a90d48409e00e8ba167d8eba : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75b9eb151f2b4efc8b24f56d396d20a7 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/755f5805984349b6a21e3789bf667a4d : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/41cb89e2374e4fce962f5b918d004f46 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7651461e3b614914b1ae46ae2c6d2dbf : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/79451d0114db410eb2616d2c37ac89d8 : SUCCESS in 9m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f8f518bb722b4c53a1579e8758ac75cb : SUCCESS in 8m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7519079f00f844999049354d383f915a : SUCCESS in 10m 17s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/dd6bae9fc74d426ab9ebf79d20087d5a : SUCCESS in 4m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/837a03c6e6064aea93e02cc61d671ca9 : SUCCESS in 5m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c73b82b8f0e343289413cf3d0568fcd8 : SUCCESS in 8m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c2fe4160b1414b8698882abd5022928a : FAILURE in 10m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/07b4b91dce44434691f0a294845cecae : SUCCESS in 8m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/af6c0d3bea5f4e30831afbf436fd4db7 : SUCCESS in 27m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4006a603b99247fa9755e6a542399eba : SUCCESS in 4m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/47a9668dbcc04b6c9bd2af84a31de543 : SUCCESS in 16m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a87dc84adb234c5c93476c43466d87a3 : SUCCESS in 5m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/81396d7e9e0b46d8887387d53deb696e : SUCCESS in 19m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e65f505d78ad4ef49bb2c1176c1abdca : SUCCESS in 11m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9bb376fe3a9349cbb25a37b3deaea774 : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7a36a93fa64640e9ba30f3a95599ee1f : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0e6c45ae0bc94227a0439709a72453a7 : SUCCESS in 18m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c9f10f3448e444eca11a14f773029a76 : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3458851e3d65437e96a74652a6df8b16 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/805318bb3d4044718beeb9235e955ebe : SUCCESS in 9m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5f37fa4edb894fdf9d10aa9a8fbd0304 : TIMED_OUT in 41m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab7932d947e14ddd948a1261588cb09a : SUCCESS in 45m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b25d42c022504dc5944f9fecebb14f31 : SUCCESS in 1h 28m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d3b2787e8da85d338fff777fee6c42dfdb2dc815","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2022-06-21 08:44:16.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"681848cf165fdd7fb80f601c46125720ca58c293","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 08:59:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f8f67f5382e94e0187b5b4da1b68dbf7 : FAILURE in 12m 53s","accounts_in_message":[],"_revision_number":1},{"id":"5e09f30dc8bf41a35c2f6055af8930077dfb9617","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 10:33:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6e8abb860645fbb2a92d74943d4572 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2cfd0b01c64249dd8548cb252d79ed55 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/05f000bcec364a939293afd301224d1c : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3087930fef1a4faaa3d2708ed07963a2 : SUCCESS in 3m 16s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/1340d635c33b436fb1ae1d5da56c9db5 : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a9d423e09b724c1882a4af3525d8d6ab : SUCCESS in 8m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0db50200e6a14df099fa81b4883c4a58 : SUCCESS in 9m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/11ccb5bc30244d69a8b41cf300f582d5 : SUCCESS in 5m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d877f74d7db746fc9040c37e4202afd9 : SUCCESS in 10m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6dcff18fe1cf40baa15abff28e0ebbc4 : SUCCESS in 4m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/2ff7a52354be437c8544ddb2f04351d9 : SUCCESS in 5m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5c83c93a9f3442e194ace46b5f194b33 : SUCCESS in 12m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a4782a1d874544cdaa76dae462e9bc84 : SUCCESS in 7m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5327fd80aa5c41c6a47d6c3230c32927 : SUCCESS in 8m 16s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/85ce3db4a2cf46c588d076d4a0348edb : SUCCESS in 11m 45s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9ce5b8461b7c411696d8dbf1afa2d8d8 : SUCCESS in 3m 40s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9da707c38e394255be65fa7910dff5fd : SUCCESS in 9m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d5339acf91604f5badb8205884fba22a : SUCCESS in 5m 08s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e3abff41878b4fafad0605eaef820279 : SUCCESS in 32m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c0c99300026346f18f200209c7075419 : SUCCESS in 12m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/139121cf647a4da0b23f2a7f27a42d25 : SUCCESS in 5m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e43ecb1422a48dd876cf70ca3c6f397 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31d5dd87583243569b003446e573250b : SUCCESS in 20m 35s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b161574fa68447dc9b62eb29b0176b4f : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bb69c236fdd6450f812552a4d29bcfac : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8432dba3922b464786354f89149fa8b0 : SUCCESS in 8m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8656a210e9524637a29c157e58f079a9 : SUCCESS in 9m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0927355a49af49cba38ac5713b3731f2 : SUCCESS in 38m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc64d632b35f47f299e5004cff3248f2 : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":1},{"id":"21fa13d6855699d8c4c0971911710d9e08fb6c41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 10:34:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fdc0af6476de226cc2224ff7e5ee671528b99e57","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 12:45:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce53e47966a8482fb9bd65d3a9f4e1f2 : SUCCESS in 3m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cff95dc7c4794b0d922cfb4390af6f16 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8d5b73613c1407f8235c1105ef4b2af : SUCCESS in 4m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bfad1a6dad44fbd9e817e1e4149e7ea : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/7a0c17eb50cb446b814d80ab12716985 : SUCCESS in 4m 14s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0f77b916212442678aabe5203137e9c0 : SUCCESS in 9m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e08a3663b74440fa137c0e6df67aabf : SUCCESS in 9m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dd8dd8a96a154684b7137700e343f808 : SUCCESS in 10m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a7d49c6cb551456ebe42df3bd422c83a : SUCCESS in 5m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/33addd5f69764f1ebc6c013bb0bded5f : SUCCESS in 5m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/19c9c87ae4be4135833b0bc999f53793 : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6ffe799b2d4240d9833d362adf0d942e : SUCCESS in 8m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/84afdacf23904bedab3920a665d306a9 : SUCCESS in 9m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2939b8c3e5164bef8bf7eea50e6b9305 : SUCCESS in 11m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/98520ba1b1114d4388389c2f2fd2e96d : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f65da7730c2b4bbba8a50955c3f0e877 : SUCCESS in 9m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/76e24fbca94b4787b9118f025f5bba18 : SUCCESS in 4m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/aa982ed24613449eafbcdc2df6f77bd9 : SUCCESS in 23m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/cb522f97a5b64283a1a8aa1f381ea76d : SUCCESS in 11m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bedbb70084954dccbd531b780db1f74d : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/37f28c9bfd3f4d20bbcc0c67d8b3c406 : SUCCESS in 4m 08s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/90e04a7d37f04762bfc37e2f7ffe0315 : SUCCESS in 18m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/54a74a7e83d7495baa223caa11fdc7f5 : SUCCESS in 5m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9c35294e342f43c297bd30b9f8447db2 : SUCCESS in 4m 50s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0a7d4b28c96449618a075a16a1ba0eee : SUCCESS in 8m 17s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5c41da02ed7c45c39d90934943a3f08d : SUCCESS in 12m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f61955cd61a43e384c80bc7f81e5bea : SUCCESS in 46m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02a4bb5bb5a34ef2b52282b12be34543 : SUCCESS in 1h 23m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c3bce8f24772c48f54fbb24eb61078f68cf31ad2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 12:45:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d513e12939d148e3653d8ea4a9dbe955e9315a48","revisions":{"d513e12939d148e3653d8ea4a9dbe955e9315a48":{"kind":"REWORK","_number":1,"created":"2022-06-20 16:11:05.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/85/846785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/846785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/846785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/846785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/846785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/846785/1"}}},"commit":{"parents":[{"commit":"5228a6045ccb57150016c9303359d89207a28334","subject":"update constraint for python-designateclient to new release 4.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5228a6045ccb57150016c9303359d89207a28334"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-20 16:09:32.000000000","tz":120},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2022-06-20 16:09:32.000000000","tz":120},"subject":"Add neutron to stable/yoga upper-constraints","message":"Add neutron to stable/yoga upper-constraints\n\nSome projects, notably neutron plugins, are using the neutron service as\na library. There is the neutron-lib package, but progress is slow and\neverything need was not moved there yet.\n\nThis is being added directly in stable branches only. Master needs to\nget the latest version available, so these will need to be added at each\nrelease so the stable branches use the correct version of neutron for\ntheir respective stable release.\n\nTo allower upper capping, this commit also drops neutron from blacklist\n\nChange-Id: I7b09604e8b983c440ffc2e7a9e493a149fc6990e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d513e12939d148e3653d8ea4a9dbe955e9315a48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d513e12939d148e3653d8ea4a9dbe955e9315a48"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
