)]}'
{"id":"openstack%2Frequirements~846848","triplet_id":"openstack%2Frequirements~master~I1cc0fbb3c4377adf02334dba627875839e2ceea2","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-29 05:03:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1cc0fbb3c4377adf02334dba627875839e2ceea2","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-06-21 02:42:30.000000000","updated":"2022-06-29 05:07:09.000000000","submitted":"2022-06-29 05:03:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"846848-openstack/requirements/constraints/noclob","meta_rev_id":"eaea1026d3c1d52b875459c4af49ab8a1766b7b0","_number":846848,"virtual_id_number":846848,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-29 05:03:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-29 03:48:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-29 03:48:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-21 03:01:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-21 03:53:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-28 09:58:01.000000000","updated_by":{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},"reviewer":{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},"state":"REVIEWER"},{"updated":"2022-06-28 11:10:52.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2022-06-29 03:48:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a9704668ef41a81093b499b6adfb8adb42fa7cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-21 02:42:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2069cbb09ed2d0c9ce8e1f195e34c75d41b1064","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 03:01:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/29a6fb80b6d74b488e6df6c0fb22fdbf : FAILURE in 12m 23s","accounts_in_message":[],"_revision_number":1},{"id":"71be72bf919e2a0ecb0c5e8bb5cf1e47a739291c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-21 03:53:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4aba93d33aab43deaa5407dc50beb04b : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4efd1bf609f040ff82789e6e2449a240 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d322a05885f4f219eb538919dad6307 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0fa09ba50e14d1e83286d4a1276e9b1 : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd402abd7c4c45a1a4fd36b0bfd16e9c : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f16669e44194d6fab09f97caaf08a59 : FAILURE in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/22d5236b18fe4d8897556e155af70f73 : SUCCESS in 4m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3b97ab785132480787f6508e6d6f0937 : FAILURE in 5m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7acefce7ff904828873dbda77d607696 : FAILURE in 4m 50s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ecce31d2322240a8b356a171869d43a7 : SUCCESS in 12m 03s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/da715debfb5342e79ae9561e4592740b : FAILURE in 4m 47s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3d94af9266004a538f490f578004c099 : SUCCESS in 8m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ea2c3a89ae4d429cba4efbc7afe5396d : SUCCESS in 11m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7a02abba7b184f2ca5a9241a68b72943 : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7a7e08aecbbf4317a13a23191ee1a209 : SUCCESS in 6m 50s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b49c063fc5b24aae8b76fe0f247c3566 : FAILURE in 8m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f080b2e8bff44d868d9c9ba2cc36bd5e : SUCCESS in 8m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cb3d2e2f5195483297a25c6651aae556 : SUCCESS in 12m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f14758b115584e7fbfab20749bc6e26b : SUCCESS in 8m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c361f97a91034a7eb82e012cc7412652 : SUCCESS in 16m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/447311fe22b74c70a4c00a4a5cad9604 : SUCCESS in 9m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b91a758ce25847bea4ede461b08e48fc : SUCCESS in 14m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d6807b1630ee41099189f0cb34e837c6 : SUCCESS in 8m 05s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b5abc3057ce84c298978597967311a26 : SUCCESS in 24m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/92216e6c16b14d329f5f15406ffab425 : SUCCESS in 12m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e3e71ec213524fb3855e0609f28cc51f : SUCCESS in 5m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a8649e0a61da4908bced7a928f37f011 : SUCCESS in 4m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a710b886bb1d483cb11346e8bf609b4b : SUCCESS in 19m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/05ff21557150448b8744b355f2d794a9 : SUCCESS in 7m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7a1ec4dd37704ff7be0c9cfb34241f43 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a57650ac3e114b12b71b0e14d42dcc27 : FAILURE in 4m 43s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/86e572f5034a43a9aea95f6fe09f5078 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f5614a6239ca4d2184f1b7420fdc045a : SUCCESS in 43m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fade3eee4f1a436594f926fcef6ff0e5 : FAILURE in 1h 01m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f97effefebf785ed308f6f1f834d53eee0d7b6b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-22 02:48:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a2a5d42c677f7a1f9183d74cf44a60d1cd08f3d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 03:19:59.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/be1653a69f9640759197530e6bf69c42 : FAILURE in 18m 05s","accounts_in_message":[],"_revision_number":2},{"id":"c44a34b5a01ff9bdbe096ea0f422dc010377e22f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-22 04:21:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f17644312b4017b53849950cf27c70 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29aa408415ac475fb20fd37394fc228f : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bdf658e8aa04b869649a59ad8de8a72 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/824b4520fd5544169a42e8e7dccbfda7 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/26156350027a4745bd00a8b7a9352aef : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8fe3235aa77c4f2fa75f0422325a77bd : FAILURE in 3m 45s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/102dd8b0a5094c2a9987fe878f0413cb : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/96fcbdb7a4d941878649227d5822fc9c : FAILURE in 3m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a6a77a7559b44f038c7b6151d290964a : FAILURE in 4m 13s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/15576047714d4881a4cb1dbf70f0d7f0 : SUCCESS in 7m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/22679a0944d54c9084b926d35243e417 : FAILURE in 5m 33s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/b709587a9e7349bda70e3678d510f5f2 : SUCCESS in 6m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8106147d8a5243e289209f36114d65be : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/69f4f93c15464a61a2f8a4daa3a537c0 : SUCCESS in 5m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f3ac83a366c24d0bbea79de7729d9032 : SUCCESS in 5m 46s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a54a8dbb4a0343c7947b1cd7a25245d5 : FAILURE in 8m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d107d3b34df64b1a97bc0fb3fdead7ed : SUCCESS in 11m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/76459d1b77924c61b89520209f2ff79e : SUCCESS in 10m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1b60d587488e429d9e8dcd4482f75b7d : SUCCESS in 8m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2be0052e6e5f40e09fbdf0aace6d3c52 : SUCCESS in 12m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c3d36ee1f5bb47ccb6eedf273b86eac1 : FAILURE in 4m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/298034c15ce244e88071cf41e27f0e87 : SUCCESS in 10m 02s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/364c2b872fab4d0ea4404c96fb3c04b8 : SUCCESS in 7m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/487bb2fa5b214270b809ddf6a0697532 : SUCCESS in 23m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ec491f974bcd4a9fbae45aa36852cf97 : SUCCESS in 15m 12s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3c24ef56826b40fbb8991c306389fde7 : SUCCESS in 7m 40s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/86f5fea7a6e54ac0b27edc16ad4e9383 : SUCCESS in 4m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0fa9222202234293877ae09fbb324840 : SUCCESS in 20m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/98ea0afe20f84abab7df8c004b5a485d : SUCCESS in 5m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/916eca2efa984d8182c6d80a9ae6f424 : SUCCESS in 6m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b343a065196f441f97979d7efa834546 : FAILURE in 4m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/972449e31cf84ddfb8ed1417ee0d85bc : SUCCESS in 11m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49ff1d52ef8d436cb497e64046d8829d : SUCCESS in 43m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e96ddcf168524480a67f48a2df4d3140 : SUCCESS in 1h 20m 25s","accounts_in_message":[],"_revision_number":2},{"id":"eeb632a566b6a90ea8ee7cb28772d954fb15334b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-23 02:39:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6c3f84c7e96a00956c813f6307f5f360a010c330","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 02:58:54.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/61a393441ceb4109ad61559a0ba5537b : FAILURE in 12m 41s","accounts_in_message":[],"_revision_number":3},{"id":"745cc9839214fb592918ea8c9051ae2a0f76c3fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 04:23:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26da27ebd52e43a995c04256b6acce66 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b243689d8646fe83b34a5d513f4746 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9bab1c68fac4cf8a6d1302e8341a1df : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac6284e89ed84798ab66128f16faca56 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf38ca10428a4ab0b5da09803b30dcbd : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7a31eb6aa7c846d6abcf6808388720aa : FAILURE in 4m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/21063a481c6842ec8272f40c76440aab : SUCCESS in 4m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/27bb2f85d8734c0d92b7543a6baaa8dd : FAILURE in 4m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/893228017f4348adab6a03ced273b16b : FAILURE in 4m 09s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d2d7136aff8e4dcf9ae9842d155cb3e9 : SUCCESS in 7m 30s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/660f41418a3a43f3ba56bc812c17bb93 : FAILURE in 4m 59s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d5d971cdda89493bbd4d11caf1811434 : SUCCESS in 6m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e8a5cf8ab2c84cc199502b3ef21b042e : SUCCESS in 9m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/772f223be1c44610b320e152215c7bc5 : SUCCESS in 6m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/28f772517a3c4863b2cfcb2307783425 : SUCCESS in 5m 03s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/385b9ac52328485686ad4e3ee8c727ea : FAILURE in 8m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/57d3493854c2415586f0bec5e2b7d0f3 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ff5ad17287e04378b5e58ab1abf305a1 : SUCCESS in 8m 07s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2013dcac043842c8b28ce4443e64a0e0 : SUCCESS in 9m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c7b7d8213a4c4bd79ef3d248c1752a5e : SUCCESS in 12m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ad1f4ee88cee43c0afb008f63dacd9a4 : FAILURE in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c3ff1866a2b543858462a9a91eb6a006 : SUCCESS in 10m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6b7e1edbfacf43e28a396218b9889400 : SUCCESS in 6m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d8ff40a45a1141fe9c23bfe3af93d6ab : SUCCESS in 26m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7dec95d7ed6c4d82971430a821b9bc85 : SUCCESS in 13m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3a9835e666094932b5aa95f5e965be31 : SUCCESS in 5m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4b15f74b15cf448095fdbc411995a924 : SUCCESS in 7m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0072a9e269cc43f0bd3b54bd42cb094d : SUCCESS in 21m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/50b6d3066e2840c6b496621d0b5b080a : SUCCESS in 6m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/84ebf75fff9d4bf0a58a7450e4e4cb12 : SUCCESS in 6m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d9f30262cce449adb2d6dbf9a9d991bf : FAILURE in 5m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ea2ae2aa2d94477e831701dd675e3298 : SUCCESS in 12m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0cf34f99beb8488dbd44e4924f0e0390 : SUCCESS in 48m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/469c34b9662341e0b2decfdf7d50fe06 : SUCCESS in 1h 31m 01s","accounts_in_message":[],"_revision_number":3},{"id":"8f61a0c50cb5263b328be26ca0fb93bc413bb5fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-24 02:40:13.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"73ea5f8f28e72e2ce03da39bd078f8f104ef72fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 02:58:56.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/db3e241181e24dca9d2b4268f970ee82 : FAILURE in 9m 14s","accounts_in_message":[],"_revision_number":4},{"id":"43dd81493d2c9eec42806465f25fd3f9d6d8649c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 04:24:29.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d24b448f90144001ad105b9c6b17124f : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26c88d3afc6445c3b40029e3edb3546e : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad5676a60d4843319ba8435f2a91819c : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9767cad3d37d4e339ad039d41611f699 : SUCCESS in 4m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b65b34346051489d9f1d250902160830 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40f5e788dc2d4c18a204b4bd161f0436 : FAILURE in 4m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/142ca2eaefdd4e33855014adeffdf52d : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a24b5a2eb398474d89b7804846e9c548 : FAILURE in 5m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2569f8fa6c7c4430be19f0ce75ff8549 : FAILURE in 5m 06s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7ca0bc202dcc451c9f0a7eea3d89b93a : SUCCESS in 10m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dc93ae65043d40d8a098b43543f245a0 : FAILURE in 5m 43s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/65085b46a2504c868733d736aa18dcde : FAILURE in 7m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3c2ef1414f7b470e9edd467bbd4fb0a5 : SUCCESS in 11m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/399a657f6d9f46b0b443d44847fb9a7e : SUCCESS in 6m 19s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ecd6dc3021a34ee982ef06edd71c8fd1 : SUCCESS in 6m 06s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/123c134dabf040deb0d13703651461f0 : FAILURE in 9m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c7d7b8ed13884021bae491819fb617dd : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/eb4ff40641c247d0813294e799749f58 : SUCCESS in 12m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/79abc088ead94337ac64d9e01209f3d7 : SUCCESS in 10m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3637eb24d9704a43be25cf2e6cc61241 : SUCCESS in 18m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/15d03e2b2ec043bd8022ca8fdede50f2 : FAILURE in 6m 20s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/40b7549baacf491da92214d423225fb1 : SUCCESS in 13m 11s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/553648fd98774d6e836bebc3e3292b50 : SUCCESS in 7m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/52bda629a7384cb6b4d8a415b9d1212a : SUCCESS in 27m 25s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/403fbbbec9be41698221f041abd44632 : SUCCESS in 12m 50s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9dec61eef0c442f2ba8995e309b53f0f : SUCCESS in 5m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4f317fb4ba8247b2a2b04d59ecdd0550 : SUCCESS in 4m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4899639db1794bada5ddfab8fd0130ec : SUCCESS in 20m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/174d3df1f21b4c0782941365e9d645b2 : SUCCESS in 5m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3749d36af07f4ca7ba5099c920de7b7e : SUCCESS in 6m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a308567b831d45398cb6e782aecbee45 : FAILURE in 4m 55s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/40929566fb6e495bb3e9b23f5d7fe0eb : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/88822f8702264833bdc74d98fb1b130e : SUCCESS in 50m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93a466527a8146bea58186f98e620633 : SUCCESS in 1h 27m 56s","accounts_in_message":[],"_revision_number":4},{"id":"d4aa739da81827dcdeb1de78f8752d0e23141976","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-25 02:49:16.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0a3dc875168737d155a171016f78e66259a29bbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 03:07:32.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e5f94260ac1143c19fd6c552070cea1c : FAILURE in 6m 17s","accounts_in_message":[],"_revision_number":5},{"id":"899ad88fbe49df339340f60252320ab794cd9402","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 04:35:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbeee2335602480d91d5a96570ba1dd7 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df65c8de8a2434b9866f20ee768f26b : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f5a9a9ce7c4266bfa7c172d1fe3c88 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70e5cd02e03e40539960f27ab8b34cbe : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/189dfb80c4ef49d4b56317807626f9e8 : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a2c92da1349343958b7ebca32e7b1cef : FAILURE in 4m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9586bbf124484360b3c0f32be58bfd6c : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/839882f2aa0d485bbb862a5355784d51 : FAILURE in 5m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8dfbf8225cdd4d539b7ba13b2966183d : FAILURE in 4m 12s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a03d84a60950432f8b3e3f200415ffc5 : SUCCESS in 8m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/959c8336099c4fcea4b25d780b99747b : FAILURE in 5m 39s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d899b41cb1a84232a13517b96b184e54 : FAILURE in 6m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d830e75d541a4fad9f015aba6857457c : SUCCESS in 10m 50s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8bf52c306c8e4b71b486ebde72d237f3 : SUCCESS in 4m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9b37430ca0f3408a8943eb71be24d653 : SUCCESS in 5m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/666089e25b9a4f6aa81fb848fcb11e6a : FAILURE in 7m 56s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b94ea5b8c4014085b867eb35554afcdc : SUCCESS in 11m 06s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c9abe96ea93e45c388be0be1065dbe49 : SUCCESS in 13m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5769b357fdcd46efa497009dc7f36eda : SUCCESS in 9m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e88bde33e7d348caba24d1fad60a6d09 : SUCCESS in 12m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/da71319c290842f8a204962ad0b071ee : FAILURE in 4m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1ff10f9dde204e25bf39c1704b4b259f : SUCCESS in 11m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/00472aca31b64065b4a3e6f6fa768e78 : SUCCESS in 5m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/13ae88d87b8345d8951156b64da54bd0 : SUCCESS in 26m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/06e576aed8464325a467f83e6aed2270 : SUCCESS in 11m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/12c9f5b9d6c842b4b10e4223330c3549 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5f839e91f32a4b8083ceafcc372aff29 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4afc591447854a48ba32d6164e197bf0 : SUCCESS in 19m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5fd08698d83a4dcc98ccb04254ff3e63 : SUCCESS in 6m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7351ec0135bd47678f16965619841d58 : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/cc8505c6a0cf42e08f1590201fb7ec51 : FAILURE in 6m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/15118304f0ae45578c4bef53290b24a8 : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/07d6be29b1c44ed6b94e6e044140d575 : SUCCESS in 36m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3f35789cd2224439b74d8d52662313e1 : SUCCESS in 1h 33m 18s","accounts_in_message":[],"_revision_number":5},{"id":"eb04253abb01505840b0ac8d7f6f73bf5f998c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-26 02:46:55.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"60c22e0b80f1f1c9e2df21cd6583bd6ea76e7e5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 02:57:38.000000000","message":"Patch Set 6:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/14794a123fdc4a099ef456a441bd087d : FAILURE in 7m 04s","accounts_in_message":[],"_revision_number":6},{"id":"9e0ac8bd1ab8807e3f60dc76454db05de95d3d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-26 04:33:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebaf2ce111f140128411ac530863ae90 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233764b75e904bf780ae34040b763c32 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3297cc44a56b4b4aa1133f41c369c661 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8661ad55f624205b12d01c97bc2430c : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/642fab8050f548e698200b82cf6b0268 : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df18a52d30464e33a2716e0d3d087cb3 : FAILURE in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a30f889c2b0b40f686c7097588e79fee : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6d33c389fcc6408f85e32bca0d196a64 : FAILURE in 5m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b0bdeb1498584a899646343b7a432583 : FAILURE in 4m 45s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/3d42c5e63a76437cad88e52e5f01293a : SUCCESS in 7m 45s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1380a66d44ef4ddfa4a4935f68bcdfe8 : FAILURE in 5m 00s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/1e8536b0d4c74f30acbfc98cadbb8822 : FAILURE in 5m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d458e2cf9d7d4a489937fa366a925e3e : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/cad2e9e5af734e558c4dc820a4b22376 : SUCCESS in 4m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/528941a454e34541aac4b7f58ca93a8c : SUCCESS in 6m 29s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f9d5e33a025d47e0b52e26e937206780 : FAILURE in 7m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b1c23c36b4454f58afd9531ecda72bd5 : SUCCESS in 8m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5de1bfa5ee6a4344aa16393ab1b8d1a0 : SUCCESS in 9m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1581621f5cc14045ac515384ec993d69 : SUCCESS in 7m 21s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3b04fa4055f94c1ca78f974c68eaf46c : SUCCESS in 10m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/5dbfd2e50cc243db895a0ff5915654dc : FAILURE in 4m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c7197a8de24c46f18cf734dbdc914825 : SUCCESS in 7m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cb6309e6473a429fbe0445c1c0762b52 : SUCCESS in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1e56dc65edc84ad2ae475640a222ae3b : SUCCESS in 23m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9f806688a8bf4088957ef68a7617add3 : SUCCESS in 14m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4cfe7584aca44b269f6cc8db1343538c : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/144c95b0867c4ed79402578b7a5bd060 : SUCCESS in 4m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d8b04140766d4645913544c8f0639101 : SUCCESS in 21m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f4af841473e84db99b0d002ed2237cf8 : SUCCESS in 5m 50s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9114bc50b92c4bfb8d65d1d92413853e : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7122c95da35443785ea26857caec301 : FAILURE in 4m 54s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/067fc65eb6d648f9b9ac752a72e86383 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fed5acf232142baaa071a169e77daa6 : SUCCESS in 49m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7d838930cc994038ab7a2623ae2b040a : SUCCESS in 1h 39m 14s","accounts_in_message":[],"_revision_number":6},{"id":"b19409cd451bf590f699c079db1ed68fa91060c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-27 02:47:58.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"fbcdc56f0980c001d4cb943e3046e45971a0931c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 03:01:34.000000000","message":"Patch Set 7:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/abaf14f581a64e9c9a5cda433ebc0e18 : FAILURE in 5m 02s","accounts_in_message":[],"_revision_number":7},{"id":"d2326ac66b97bb34ea570cbb50ad3d8d9ca48397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-27 04:26:08.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/310ebaae80b241adacddb1a1b808e941 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13dc30e2bfde495cbcb206347b3d3a63 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8db6ad0e5b0c4463aee8b85f7198cdc1 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e78863e51eec439f9358176947a2a9e0 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b3345e240e34eb69a51fbfafd50b83f : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dcd87a0e30ce4b86a33f88ce93939534 : FAILURE in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c887f2bfb6a14b9f97d6628bdf861a1d : SUCCESS in 4m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/695829022feb4611933e5d169694be69 : FAILURE in 4m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0ea2c0da44a94be192920ad9c35a41ad : FAILURE in 5m 13s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/3b535d6f52f7455b9365d25a2236d01e : SUCCESS in 7m 46s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a8864d6fdcf94670845d5d871bb859c4 : FAILURE in 5m 55s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2a871727ff924861b6cf06b946123721 : FAILURE in 7m 24s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/565370c2f77442b1b6486a4e5cdd2143 : SUCCESS in 10m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/72994307a08d4df78551f1fe4249a800 : SUCCESS in 5m 03s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/93c2101694be4c5895ab0490d62ba8b2 : SUCCESS in 5m 49s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/be5b2e9ccf544b888507778085e41aae : FAILURE in 8m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b9628255387c4e7bbc2c406020be503e : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6293c6562d64ee69cb40994fe53d652 : SUCCESS in 8m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/619f1474192c4ce8a51bf0f082257b9b : SUCCESS in 12m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/44f95f34dbed4f8bb0bf778ab1ffa762 : SUCCESS in 18m 04s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a487e05e7e984ffcb2051be5520ede64 : FAILURE in 4m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a03a7e0ad2e04c3e98d1887ca949c624 : SUCCESS in 11m 07s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8a411cf4e075487492f89b7b26e19800 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ed4d83ac6f254e24ac45514d9d2d68f7 : SUCCESS in 21m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/de1bf43a48314db7acd7267a7890ef03 : SUCCESS in 12m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/03ebbaa0ffd347639e3f57fe513c5f01 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/056e55e1284b49a884d5298f7ab82113 : SUCCESS in 4m 25s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/905bc4cf57a14754b9b42c036a6c3bef : SUCCESS in 20m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3b021ebd8270454fb5c1f57336f949f4 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/70637c665f1841bcb86ea090dc0309ce : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7d7c28d67d7243e99294da1014c82b4f : FAILURE in 5m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3dc88230fad2488aaa73b5dc815140b4 : SUCCESS in 10m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/790900e089fd43c489e9bee16f75c41f : SUCCESS in 45m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b6a4f36e9784310b79d5e454f6e0a70 : SUCCESS in 1h 29m 40s","accounts_in_message":[],"_revision_number":7},{"id":"ec6fadadcc1485fdc32c5fc27cd22b2d81e6e768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-28 02:44:30.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f3a533d504bd207f566f02d56628bc2917177eca","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 03:02:07.000000000","message":"Patch Set 8:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0a455f297fdd4ff182724af8503be263 : FAILURE in 8m 45s","accounts_in_message":[],"_revision_number":8},{"id":"c50fb89e5883f91f44c6eec6fa622a55a60627a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 04:45:51.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cbe728faba84c54a4d40b7911b47e4b : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e52fcfbb33e3454a95cdd7285f6d18c2 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cadff957e7754d0bb2ac00a47b71c70a : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dee1f5a1c7784a48b9dedd615ff13d93 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51aa19d73c9e48c3b7c73466ea92777b : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6410f13d280147aeb84bbdf85a495828 : FAILURE in 4m 17s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/946ef15646544591a46bc029c23b77e5 : SUCCESS in 4m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/601b7886b4d34b28af1d67280a9c5112 : FAILURE in 4m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6e6bdcb50f1c4ba5b80d359997b1f8d5 : FAILURE in 4m 07s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ea9fcd38ba6643f9a8a404a4815c6abd : SUCCESS in 8m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e9d0aef800514419b1e3123af0b2eee3 : FAILURE in 6m 06s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/0ef6d39f25324eaa991a6b7b7813cdcc : SUCCESS in 6m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5988692c06414122b15b7721ed11fc14 : SUCCESS in 10m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4add7ea17e7444ee8730d095da83997e : SUCCESS in 8m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/90251effca244d77919e5a2208a9178e : SUCCESS in 5m 58s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7b0dd23e03a243f7a901667a7655897f : FAILURE in 8m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/14aeb0f0c532488b81e056a0a19fcaf8 : SUCCESS in 12m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/02b8526865d24ce7948f329d147840ef : SUCCESS in 8m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f07af5402ed34f92abd907bd2bfbe6c1 : SUCCESS in 11m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/20393c4ebc894b64b03556cdb0df5b6e : SUCCESS in 12m 13s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ffff913573f64afbb2cb6f7b7b3e8213 : FAILURE in 4m 19s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/f00f5bd377824858911a0efb8b8fec19 : SUCCESS in 8m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/15d6d1c6ec554dd1b09ff4a8fc204697 : SUCCESS in 8m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f929687c49e74bdc9e1b5bb2b7e5b955 : SUCCESS in 26m 58s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ddc8e4bffd6640efa7377c0536a69426 : SUCCESS in 11m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/00239672358547a7bca5b7935ddc21a8 : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/324b92d15893447090b63b6eef4fc56c : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/021decb5b1924e06a30d33ab99b9c261 : SUCCESS in 21m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5cff9421cb424f23bcf274b5896fd014 : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2d8eb1885c5144f6aacad735da5233f8 : SUCCESS in 7m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ee21e26f88564d8c87006cb8a56d1e2f : FAILURE in 4m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9c4b8e4c453b4873a7231b80914de9c1 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fc6671dbba65402988fbe14be6e141e0 : SUCCESS in 51m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b7a906192ed40a08c35f95afae7da2d : SUCCESS in 1h 34m 08s","accounts_in_message":[],"_revision_number":8},{"id":"089977230045cd200f71470c614250895ae1fd77","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-28 09:25:10.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b8ffa26883cc67bd69a7448a152117d19c9a2e8c","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-28 09:25:40.000000000","message":"Topic openstack/requirements/constraints removed","accounts_in_message":[],"_revision_number":9},{"id":"403716d2d965d9415f404cf545a137ad0636238f","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-28 09:26:19.000000000","message":"Topic set to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":9},{"id":"6c12939ce77c43b83e5632a5097edf85a06f71c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 09:36:53.000000000","message":"Patch Set 9:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e05e7af144744e3b89bb2b06c618db66 : FAILURE in 7m 37s","accounts_in_message":[],"_revision_number":9},{"id":"d9006eecf0b9089fae3340be11f6fa77366d9185","author":{"_account_id":33543,"name":"liuhuajie","email":"liu.huajie@99cloud.net","username":"liuhuajie"},"date":"2022-06-28 09:58:01.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3787f9c7da0640c57d300db21e6b9d8e9c8c7d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 10:54:38.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a279907749e42ac9c4ae547c4b3140c : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/925f9913a4a143bd833589d183fa38f2 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f670371257346b8abc4f5278764d459 : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/397acccf43464725b09e0c6b91f3beca : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5661cfcc8fc14ae9920ae943ff5a8fff : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ee07ff4638446339c5e6e733fe8fb8a : SUCCESS in 4m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0185b40d0fa14142ba96c4cf5738280f : SUCCESS in 3m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/a37811af627e4084b2f299a7eac3e993 : SUCCESS in 10m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3db0c490e4c14b22bb4fe9281f5bcf8c : SUCCESS in 12m 50s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/488c24b3da924b2a9587de51746c75dc : SUCCESS in 7m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/2a8974e2002744ff92b0c370ce4c7188 : SUCCESS in 6m 42s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a31517fe826140059a990d5540861c48 : SUCCESS in 7m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4b70ff0d6c614f868bb97e196fff6fb6 : SUCCESS in 12m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9c0ed872b28943fe9a7484b310e0fd8f : SUCCESS in 5m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a14ed4ef567d4e078b43e88d884e0d26 : SUCCESS in 6m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/189369177aff4c9ab66783130b7771dd : SUCCESS in 7m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/891ed0ac275540979f80d4de87dc1be0 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bd62e472fd5f49b79c3b2aa32641a306 : SUCCESS in 8m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bc09e0e99df241a7a7189bb1695d3225 : SUCCESS in 8m 03s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8af23729caea441ba03cf2d9856e5aa9 : SUCCESS in 11m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/90a8ff08debd43198b48368135367ecc : FAILURE in 4m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cc0d2fd8b33e47538533ddc65bde6426 : SUCCESS in 10m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2ab132387efc4fa8ba356c741c8b8a81 : SUCCESS in 5m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3076c40c55004db4982686616ef302b6 : SUCCESS in 21m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1e6c509ff74c4e8b8d37829cc7f2671e : SUCCESS in 12m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b9b638c727b341b08c55d45acbb11aac : SUCCESS in 6m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/352be76adc954101809d0ebf766c39c2 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f5263cc1918347ad83cd1aeba81c271c : SUCCESS in 17m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6c39da458e5142beba976cd94f92874b : SUCCESS in 6m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dfd02f30485a4d0d8c33324ca9795dc6 : SUCCESS in 5m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/33b42cefe2284b7893dba302b5a5f49c : SUCCESS in 6m 48s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2e986957335746a287ade34555c57c8e : SUCCESS in 9m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/50c736abb73e467280c3a376eace52ce : SUCCESS in 44m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/811f744d9da8491b988d38a57d750e73 : SUCCESS in 1h 26m 22s","accounts_in_message":[],"_revision_number":9},{"id":"afedf9882d5a3f87be6522e662c2e3c034727d34","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-06-28 11:10:52.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f357ad13789185ea2c8b429fe95177a04e4c0c6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-28 11:45:23.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"13118104b82f006d8ff9dcfc9b2c191f2ba1e633","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 12:02:07.000000000","message":"Patch Set 10:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e142fe8382074aca97eb0c35faf6fa44 : FAILURE in 8m 20s","accounts_in_message":[],"_revision_number":10},{"id":"5a43f9f32d0f629c4e645457b42a0cc2468eaf8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-28 13:13:13.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150c47b5de7e45feae99554e6646da9d : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ce4726bce446febdd0c6e35833f364 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e22996e6764b4d28955ef116a8d469c3 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7edb4173e3c4219a8d45886b64169d6 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9f5536dd04a4727b91e046e89c9c7db : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aecd8a6f25524fd6b1f83f65de41f70d : SUCCESS in 3m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e57591ac26be4bcf8c2a0c4de5c25873 : SUCCESS in 3m 30s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3908360d51424bcb955a1b6bc2934111 : SUCCESS in 12m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/378bf92098bd40cc89913f29106ce8b1 : SUCCESS in 13m 50s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/0d3bbb5b022d47fcb76622d9c46ba86b : SUCCESS in 7m 46s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e1fa0a3225b2465db91157e55be23334 : SUCCESS in 5m 31s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/1acf4f9034f645e5bb18f8794fb622e6 : SUCCESS in 6m 29s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fd7ee87114fe4df6aa07b9eca3a914b9 : SUCCESS in 10m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ae4fcb911ae346a8be4455dd8c63a009 : SUCCESS in 5m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/edf0771bceb74ca695c8c13b1650d11b : SUCCESS in 5m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/75b8a2d82b96420fb7a5b401da0620e0 : SUCCESS in 8m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b6f848e4967f470eb8b0363ec57d451f : SUCCESS in 9m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/96f77722481b4ceda2b40617df5fc42b : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cef187618b404b7e8c0047959cdbd9e7 : SUCCESS in 7m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b5cb470d98d74148a8fba0b7fb8e7e58 : SUCCESS in 12m 17s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a2d5db52aeba49a895dbc8058062b956 : SUCCESS in 4m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/deeb9b5468e54a9682969e516521d697 : SUCCESS in 9m 41s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ae2a890d57614ed3803213c9bfc49b8f : SUCCESS in 4m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c7d850599697447ca2b6186f8eada257 : SUCCESS in 27m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/aa24d0e6338645ce93c3086a92531cc5 : SUCCESS in 11m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c8e5b4e8f8f1427c9d3055a336fd6aa6 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/783cc1a6786e4913a6aec4d42ea6764f : SUCCESS in 4m 04s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/525b8c0cca414bb18f667ac7db46809a : SUCCESS in 22m 49s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c95706cac9484a018e0119e531b6638f : SUCCESS in 5m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d064fe72698a4d07835110f6ff128d05 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6173c516e06a46a0ae1498e87203639b : SUCCESS in 6m 04s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0a09432e6b8a4a96b2bef52c1f6eec72 : SUCCESS in 12m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26c8fc23deda466788d00e6067341156 : SUCCESS in 47m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65e4d9046ed6499fb39a2bf3f72a84fb : SUCCESS in 1h 22m 22s","accounts_in_message":[],"_revision_number":10},{"id":"2b7db1a0c7f201f7f34711aa02393fa5ac6dee00","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-29 03:48:26.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"c33a0752a5b4ff9af3ba63e6827000015c048011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 03:49:22.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"60446ac4111aa27650c9f13c7c156f8bb38c3524","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:03:49.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/170ea10902b2459aab0b0b89255f00df : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9289b7a7c46c4c89908020123ec1c005 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/602863fcb9b14450921d5de16adfe7a6 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e2b4ce5b42344ca94f0511ad6b638b2 : SUCCESS in 4m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/024ff1a0117e4cae8207fd944fcce841 : SUCCESS in 3m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/234d336cf1a1435ea174d4d60bf25106 : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f38c4f1373b1466dbdd96ae732b0b90a : SUCCESS in 9m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4a1ce8672ca74cc4a69e2832526f1387 : SUCCESS in 11m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/f8c2943cfdfb4a0fa1a60bb89af92f5e : SUCCESS in 9m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/16199b0c04e6453f815f26f83df66dd4 : SUCCESS in 4m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d2eff62be2a74a8b9e7f2afe92464564 : SUCCESS in 5m 46s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2a942888f328431a8c1519b1fe02bb50 : SUCCESS in 7m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9b62e07f78fa46d78c76a50f5949519a : SUCCESS in 8m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db47b03a07864cc099b75b6f8d989753 : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6b12e23a60e2498c9d7598aedcd31b51 : SUCCESS in 6m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/8e9f2e138a104cd1a1e3db60ba90ec4a : SUCCESS in 10m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2f8fc7d0069d401da853297a5b0452b4 : SUCCESS in 4m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c55d248c054a49c9964c49bc8e700cdd : SUCCESS in 8m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/df9144aba245466889bae571dcfe99ac : SUCCESS in 4m 22s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/00150056a94d4a30a73c9d7c4619dccb : SUCCESS in 21m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b37b0b549fd744efabdfe54868bbb8e7 : SUCCESS in 11m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b771e5029b3546f4a536d5af5f7f726f : SUCCESS in 7m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f95dfc9e6ec425aae5222a0bdb56577 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/914f420661ac44a1914fa7ab75f2e3bc : SUCCESS in 17m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/312571858c7848f8a7151f834f66d64b : SUCCESS in 5m 06s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f4bc231e62d54a069cb52027f9a15f9f : SUCCESS in 5m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/484c318f0c164481b451deececcd1003 : SUCCESS in 6m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b48b0abe63564ddb85a647771489c212 : SUCCESS in 8m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fbea3643c644eda8463bda7f62710c5 : SUCCESS in 44m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8eac7f0ac49c42bfa5b57a5eb2f4ebc2 : SUCCESS in 1h 11m 48s","accounts_in_message":[],"_revision_number":10},{"id":"74355f17f62998ce756bde638f16f81d13754b2b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:03:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"eaea1026d3c1d52b875459c4af49ab8a1766b7b0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-29 05:07:09.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97fab79391be4f2eb1a4867ccc6064c9 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"e592841dcac7f621b7d3dc492fc7292cb33f892f","revisions":{"9e6e5dc5f68d520262f39fb08980fcabba8d4565":{"kind":"REWORK","_number":1,"created":"2022-06-21 02:42:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/1"}}},"commit":{"parents":[{"commit":"264c424c399b0ef931a729caa3aa76888e5727c7","subject":"Merge \"Skip networkx 2.8.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/264c424c399b0ef931a729caa3aa76888e5727c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-21 02:42:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-21 02:42:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e6e5dc5f68d520262f39fb08980fcabba8d4565"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9e6e5dc5f68d520262f39fb08980fcabba8d4565"}]},"branch":"refs/heads/master"},"067ac185aefb3da06ec60df587b0d4b61e2742fd":{"kind":"REWORK","_number":2,"created":"2022-06-22 02:48:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/2"}}},"commit":{"parents":[{"commit":"264c424c399b0ef931a729caa3aa76888e5727c7","subject":"Merge \"Skip networkx 2.8.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/264c424c399b0ef931a729caa3aa76888e5727c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-22 02:48:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-22 02:48:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/067ac185aefb3da06ec60df587b0d4b61e2742fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/067ac185aefb3da06ec60df587b0d4b61e2742fd"}]},"branch":"refs/heads/master"},"1452ed21eb19ad855d1c5571f3a5cde57e2e7ee0":{"kind":"REWORK","_number":3,"created":"2022-06-23 02:39:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/3"}}},"commit":{"parents":[{"commit":"264c424c399b0ef931a729caa3aa76888e5727c7","subject":"Merge \"Skip networkx 2.8.4\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/264c424c399b0ef931a729caa3aa76888e5727c7"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-23 02:39:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-23 02:39:01.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1452ed21eb19ad855d1c5571f3a5cde57e2e7ee0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1452ed21eb19ad855d1c5571f3a5cde57e2e7ee0"}]},"branch":"refs/heads/master"},"e6e6f5b87e8c02d57a457071883bf87b199319ec":{"kind":"REWORK","_number":4,"created":"2022-06-24 02:40:13.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/4"}}},"commit":{"parents":[{"commit":"699157ca9f7356985959ba5bd3c493b4450e5b5e","subject":"update constraint for horizon to new release 22.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/699157ca9f7356985959ba5bd3c493b4450e5b5e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-24 02:40:12.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-24 02:40:12.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6e6f5b87e8c02d57a457071883bf87b199319ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6e6f5b87e8c02d57a457071883bf87b199319ec"}]},"branch":"refs/heads/master"},"e92aaf17f09420e1d8d0be354255d3d13ab4df7e":{"kind":"REWORK","_number":5,"created":"2022-06-25 02:49:16.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/5"}}},"commit":{"parents":[{"commit":"bac67dfafa83cf3483007115b1e3099fed3f3589","subject":"Merge \"Revert \"Revert \"updating gnocchiclient\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bac67dfafa83cf3483007115b1e3099fed3f3589"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-25 02:49:13.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-25 02:49:13.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e92aaf17f09420e1d8d0be354255d3d13ab4df7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e92aaf17f09420e1d8d0be354255d3d13ab4df7e"}]},"branch":"refs/heads/master"},"90f084ad0344d554f01e9af30b5df7d9a1fc2de9":{"kind":"REWORK","_number":6,"created":"2022-06-26 02:46:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/6"}}},"commit":{"parents":[{"commit":"bac67dfafa83cf3483007115b1e3099fed3f3589","subject":"Merge \"Revert \"Revert \"updating gnocchiclient\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bac67dfafa83cf3483007115b1e3099fed3f3589"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-26 02:46:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-26 02:46:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90f084ad0344d554f01e9af30b5df7d9a1fc2de9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/90f084ad0344d554f01e9af30b5df7d9a1fc2de9"}]},"branch":"refs/heads/master"},"4aa5786aafee818657c6edb10f2ac09f97c615e0":{"kind":"REWORK","_number":7,"created":"2022-06-27 02:47:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/7"}}},"commit":{"parents":[{"commit":"bac67dfafa83cf3483007115b1e3099fed3f3589","subject":"Merge \"Revert \"Revert \"updating gnocchiclient\"\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bac67dfafa83cf3483007115b1e3099fed3f3589"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-27 02:47:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-27 02:47:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4aa5786aafee818657c6edb10f2ac09f97c615e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/4aa5786aafee818657c6edb10f2ac09f97c615e0"}]},"branch":"refs/heads/master"},"c27f45e3a6efe3ec488b9be24ac1bbf1f8caf3b5":{"kind":"REWORK","_number":8,"created":"2022-06-28 02:44:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/48/846848/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/8"}}},"commit":{"parents":[{"commit":"acce288f18915f09bff2843727d4da800b663948","subject":"Merge \"update constraint for oslo.utils to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acce288f18915f09bff2843727d4da800b663948"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-28 02:44:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-28 02:44:27.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c27f45e3a6efe3ec488b9be24ac1bbf1f8caf3b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c27f45e3a6efe3ec488b9be24ac1bbf1f8caf3b5"}]},"branch":"refs/heads/master"},"b6187749c2cd78db7670b691dcb5e4ec5135b309":{"kind":"REWORK","_number":9,"created":"2022-06-28 09:25:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/48/846848/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/9"}}},"commit":{"parents":[{"commit":"acce288f18915f09bff2843727d4da800b663948","subject":"Merge \"update constraint for oslo.utils to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acce288f18915f09bff2843727d4da800b663948"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-28 02:44:27.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-06-28 09:24:55.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6187749c2cd78db7670b691dcb5e4ec5135b309"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b6187749c2cd78db7670b691dcb5e4ec5135b309"}]},"branch":"refs/heads/master"},"e592841dcac7f621b7d3dc492fc7292cb33f892f":{"kind":"REWORK","_number":10,"created":"2022-06-28 11:45:23.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/48/846848/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/48/846848/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/48/846848/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/48/846848/10"}}},"commit":{"parents":[{"commit":"acce288f18915f09bff2843727d4da800b663948","subject":"Merge \"update constraint for oslo.utils to new release 6.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/acce288f18915f09bff2843727d4da800b663948"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-28 02:44:27.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-28 11:45:23.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I1cc0fbb3c4377adf02334dba627875839e2ceea2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
