)]}'
{"id":"openstack%2Frequirements~847309","triplet_id":"openstack%2Frequirements~stable%2Fxena~I079a8be093869e457ad021b0de1c46b7b7a98653","project":"openstack/requirements","branch":"stable/xena","topic":"new-release","hashtags":[],"change_id":"I079a8be093869e457ad021b0de1c46b7b7a98653","subject":"update constraint for oslo.messaging to new release 12.9.4","status":"MERGED","created":"2022-06-23 10:55:52.000000000","updated":"2022-06-25 01:54:10.000000000","submitted":"2022-06-25 01:54:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"847309-new-release","meta_rev_id":"2068c2b577e7143e5e7b2d5967da774f28556d93","_number":847309,"virtual_id_number":847309,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-25 01:54:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-23 13:55:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-06-23 11:19:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-23 13:55:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-23 11:10:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-23 11:19:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-06-23 12:34:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-23 13:55:52.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"1037aa074553dfbebfe8d92f3c092a5614e01eea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-23 10:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18d26bb2cbfe16241ec88c11130afd2573896abc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 11:10:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/faa2d72559f347bfaa3030277ced063e : FAILURE in 8m 12s","accounts_in_message":[],"_revision_number":1},{"id":"30bde1784a52194e706ab12253b45c4e1529f101","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-23 11:19:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f0515d5d3521e3e57ae5349ecb4f525f40640dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 12:34:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b158208bf9d54bdeb5c12cc63163cfe2 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/977f7250b565425286da3a94c4eb806b : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64e339eb8be947aeacd1c294c7180daa : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21fa20524b224cbe8b979d82b7fefd7b : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d8b88bbbca0046f89e44564dfc89b553 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/95b3d990ec2a46439538bfe07e7cca1a : SUCCESS in 3m 41s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/58bbe87c7a0c458bbaca5fdec3d857d8 : SUCCESS in 8m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6a21362c143c40d8ba467875d60bc7f7 : SUCCESS in 8m 08s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/0576c742a290445d958989ca25193444 : SUCCESS in 4m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ad9f6bebe99545e683b279ca58ee048a : SUCCESS in 9m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/203acda91a944ef3a448962c3990b42a : SUCCESS in 5m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/437e60fb038042c982646f38e5a44035 : SUCCESS in 11m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f7bd83609a1425baeabfec74aa6f7a3 : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a7ceeb4db3ea46d0973f8c9cb63df69f : SUCCESS in 8m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ee36d955e53b4f78a6e6e77a4c1d0e7d : SUCCESS in 13m 16s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/015e590d280d4c2e9d2c39e4e51a62b2 : SUCCESS in 8m 57s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8497439e64c740dfb5f3655ef1d77b62 : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/21c3a7daa1a84d24a8d65f4f6da7f399 : SUCCESS in 24m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6cd02773b3784ecebde500a4d0b865f7 : SUCCESS in 13m 24s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/77c35ccc519c4706b42ffdaa4a4fb9f6 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e414ad7bac93444e9c9a8f64cce53a40 : SUCCESS in 3m 12s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1133922c0e2847d2b5ffd67586a59909 : SUCCESS in 17m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d1958c77fec7429988a7d3fb154255ca : SUCCESS in 5m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/60695c3e7cbd4d4789195bb1f58454bb : SUCCESS in 4m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/216b898d6e2c40608e911940269c3b9c : SUCCESS in 8m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2facc31df625405d93c9b911f5c5ca89 : SUCCESS in 12m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43c63d61322b4fe392fbc1e4074dbb8f : SUCCESS in 50m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c73f67cdb6d4daa85c588920f0f8797 : SUCCESS in 1h 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"b06dbbab395947a08412e7cb7c01c858dd84af3e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-23 13:55:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d0d07da7181d5e5872fde9c48d02e8426b5392a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 13:56:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c34c123259324586524ba1f1022dd73fbb296a6d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-23 15:32:14.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac18e9b0d49e49f89487f7e7a9b693f1 : SUCCESS in 3m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b24a406d7baa4041a3c0ff163aa0c194 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8261787ce8745b098b545ab278bb7d3 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/276b69cfea024231993ae9a3e8044f99 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a4603749dedd49fc9f90e04ec5c890f9 : SUCCESS in 4m 54s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3e067d90d8664de6871aea823785de3e : SUCCESS in 9m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b10d57e887394a8bb28e89a545116e67 : SUCCESS in 9m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2b97ad5e693e4b43b1ccdda93a31fd8c : SUCCESS in 9m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3ddc0557f9884d06aa557da6ed622ebe : SUCCESS in 5m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b2666156755e4e87af13f8f6520ba0fd : SUCCESS in 9m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a325f31a98c7467fb39686be40ba56ac : SUCCESS in 8m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1f618a8286df44dcab52f4ae86a9e22a : SUCCESS in 11m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b6cbed265ba34561b6a62254aa1a5222 : SUCCESS in 44m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/53c44e01217b45d5bd7febd757cc5d41 : FAILURE in 17m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/693815ea9593413a96dae4692430dc53 : SUCCESS in 4m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7ff41e9bf17e4633ada152e5f4fe05f4 : SUCCESS in 23m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9230e35ccb6b4b6e86613fa9c4412b00 : SUCCESS in 11m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b2db07aff900409a9043d4f22ad93980 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8e8f46f4f26a4aeca3232762a912df8f : SUCCESS in 4m 29s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/63aabef86afe48368d160077a2cab96e : SUCCESS in 20m 18s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5792956231a34b58a8252c3997814f41 : SUCCESS in 5m 56s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f3e36b9a4c08475798b910cb26518938 : SUCCESS in 5m 10s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0778816672994ebea85e8ed5aea4d33b : SUCCESS in 5m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2b40a50b5c74489e8a3a6f1837010421 : SUCCESS in 12m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/11f11d97b75843f58a7843df12cb7720 : SUCCESS in 38m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4fc691141d2340b5a817a0112acc9abc : SUCCESS in 1h 29m 41s","accounts_in_message":[],"_revision_number":1},{"id":"080cdbad1dfe4cbf1c71fd9d513276a2e8fab5bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-24 23:25:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"020be8c36103b7c97429317cc87dd904c292c461","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-24 23:47:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b38863c052de4fb385b64da505d29aeb : FAILURE in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3797a30c8f62e7756b49cb1c96a28ccc6f0972db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 00:47:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0df3e3b1c2d4cd1a4d70eff384e9060 : SUCCESS in 4m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e2ec7ae4c214013a98d7834f73762d3 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d167276504e04330a37ccfabacf7eff4 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eeef80772c944bc7b9d4e5ea8329afca : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ab65bf23179e428da04a9329cb9c6126 : SUCCESS in 4m 15s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/acb1403fc5ae48e785eb53eebb327939 : SUCCESS in 5m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/6054f81deeb5428e8caaf78edf5d6b7b : SUCCESS in 10m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/919fc698af8745d9a7d90da2a052dd9e : SUCCESS in 8m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/087b4c7c6c2c41bd8401edf4ea895402 : SUCCESS in 4m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a164c9ce3271425a8934f181f6b6b1b3 : SUCCESS in 10m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4f7a67a4d9484c109db9fb5752853c8f : SUCCESS in 5m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5f4072916c8b49f599fd8f95c8025ba8 : SUCCESS in 8m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fd02b29cc43f4474b086482292ca8323 : SUCCESS in 11m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/726f669599884d88a6c7ded70677c1c1 : SUCCESS in 8m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0e3c4c4b58444a44b75fa318298d6321 : SUCCESS in 31m 12s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/79d68c49abee4efb95e7ebbbfed76dd1 : SUCCESS in 12m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/18348497e819446d8949c7b09b13eec7 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1b172dcdbb054381b4ed6a528286ed54 : SUCCESS in 28m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3d10707e469a447781ca7ba541dc9e8f : SUCCESS in 13m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b0bb6821eef245e1ba131ef1586cca5f : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b153a558c1a740d7b661aea07a94366e : SUCCESS in 3m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89f576a1356b47688fa4767ab65b9581 : SUCCESS in 16m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/30bdd8f47f6045bebcb9e89531419e4a : SUCCESS in 6m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b013fb1ecd194aafa5e4671a0cf0ed15 : SUCCESS in 4m 52s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bee96eb61ab74dcc89c1eb0a89e386f4 : SUCCESS in 6m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/860fae876c8846f8a2db9d39cc335317 : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19c5832c6f0f440aa9b80803d107572a : SUCCESS in 43m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fee473fdf3674ea4bf65024e76374981 : SUCCESS in 1h 19m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3bfbd8205d585f312b981ada91f358e225657d18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 00:48:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c66c7d40096373e2e07e8ff77a4da020d0f6ee29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 01:54:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6ed73ea2ab142fdab4f6913630a95c1 : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e22d8600aff6411d8b64a17f65aab410 : SUCCESS in 3m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3aed7fb7ec64285aab731a77c3c8ece : SUCCESS in 4m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c5fd920614a241689e6a373348e2ba54 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bba6e5c751514b228bf86515aac3791a : SUCCESS in 3m 50s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/00c2eb31a0174853850d0a3f994fc965 : SUCCESS in 11m 47s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d4cc517f5f75489ca31c251f056b79d4 : SUCCESS in 6m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b35fc877a986457ab8f71b7ac32478c2 : SUCCESS in 9m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dde4ca18e4ca468fba0810980eac1159 : SUCCESS in 4m 47s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a96bf271e22649958a3808a5fa0e7ef4 : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bfd676ecba904398a6200e0ba599ca34 : SUCCESS in 7m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bf12620ac6974c8bb237dcaf11c956b2 : SUCCESS in 7m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5f2646955e494219b4dc121dd6bb652d : SUCCESS in 17m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d63c6818d5134943a8084f2f499d7932 : SUCCESS in 10m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0dd4f6a4ae89460396a8f5c27b4df056 : SUCCESS in 4m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6559d56c904748688553704b60db6436 : SUCCESS in 19m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/42cda577948d41a59e62b43bc90f4ceb : SUCCESS in 10m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c27bf9ad7cd34ceda7633e74e4c65880 : SUCCESS in 5m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6d40a69782cc464982240f40ba0dfc81 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d3dc1a44661d475eb35f421971de6151 : SUCCESS in 15m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4ede86585d934807a7d35b0289b2f75f : SUCCESS in 5m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7ad0fd53949943fe867d0aa390979215 : SUCCESS in 4m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/e7de608cbf21442da16d9e9348873165 : SUCCESS in 5m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e9d45a8ee89441f2928e059c51641445 : SUCCESS in 10m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fe0eb158218d4203b4788aa73d70d26f : SUCCESS in 45m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/45385b4e474247e39d40d5e536413c47 : SUCCESS in 1h 05m 18s","accounts_in_message":[],"_revision_number":1},{"id":"2068c2b577e7143e5e7b2d5967da774f28556d93","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-25 01:54:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8cd14a8ce56c35143037aea23ab2e4c220b66d93","revisions":{"8cd14a8ce56c35143037aea23ab2e4c220b66d93":{"kind":"REWORK","_number":1,"created":"2022-06-23 10:55:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/847309/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/847309/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/847309/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/847309/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/847309/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/847309/1"}}},"commit":{"parents":[{"commit":"03061a94bf2b6e856d5ea8436d7e8d54a8bbc4e4","subject":"update constraint for python-designateclient to new release 4.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/03061a94bf2b6e856d5ea8436d7e8d54a8bbc4e4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-23 10:55:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-23 10:55:51.000000000","tz":0},"subject":"update constraint for oslo.messaging to new release 12.9.4","message":"update constraint for oslo.messaging to new release 12.9.4\n\nmeta: version: 12.9.4\nmeta: diff-start: -\nmeta: series: xena\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Commit: Hervé Beraud \u003chberaud@redhat.com\u003e\nmeta: release:Change-Id: I55f662c1d1373c11863ffe9c77d4d37e96778860\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I079a8be093869e457ad021b0de1c46b7b7a98653\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cd14a8ce56c35143037aea23ab2e4c220b66d93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8cd14a8ce56c35143037aea23ab2e4c220b66d93"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
