)]}'
{"id":"openstack%2Frequirements~848209","triplet_id":"openstack%2Frequirements~master~If0dc814b74ab10eb9400a2a1fa6a220d454da755","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-07-01 15:13:15.000000000","reason":"removed on reply"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-07-03 17:43:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0dc814b74ab10eb9400a2a1fa6a220d454da755","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-06-30 02:53:00.000000000","updated":"2022-07-03 17:44:06.000000000","submitted":"2022-07-03 17:43:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848209-openstack/requirements/constraints","meta_rev_id":"09e8fb36e3dccaa90356fb7d198938a8674e64d6","_number":848209,"virtual_id_number":848209,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-03 17:43:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-03 16:23:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-03 16:23:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 03:05:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-06-30 03:23:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-06-30 05:04:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 17:38:51.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"4a636f4497dd84b71817da04be6e88f27e91058f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-30 02:53:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e844058598464a7c3fb9896052a9a5ab89763aaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 03:05:04.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d16e86d0a1b84ecf8f7d5a5e9cdb95d4 : FAILURE in 5m 26s","accounts_in_message":[],"_revision_number":1},{"id":"badc63d85024d679a2703b58d1e9527d493937a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-30 03:23:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dfe0c745670ded593b8b2d016d8efe6d9ec2a23a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-30 03:23:29.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"dceed0a67eb41128ce028888fac70b13e624fb03","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 03:38:53.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/946832b6a1364ece909589986e0e70aa : FAILURE in 12m 47s","accounts_in_message":[],"_revision_number":2},{"id":"bfac00382c91126e248c8aaba589b7b5b2448650","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 05:04:29.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/051cc80018ee4ff6826ab7c7c821949f : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4270f6d23c0246a1862111f193b81b94 : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0caf78be2ee54811bf631363c4005812 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2addcdb6ff50492797e2270aea953690 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9a11670f12574302bb58749bf3b8f3e9 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f0925bf08a9471bae54fd98202067a2 : SUCCESS in 3m 25s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8c958721fa3d469e9431a9c0abe11299 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85a75d676f6042a290169399cabdcbbb : SUCCESS in 9m 35s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d7890a51b90b47a983085cf4220378c2 : SUCCESS in 14m 26s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/3df26729628b40699da96144e169b3b5 : SUCCESS in 7m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f9ba17709b83412c97f2d8d1ed74cb72 : SUCCESS in 5m 57s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e5c38790bda444f0961fee8cdd387eae : SUCCESS in 5m 56s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/473bb171121e4dd986d0bb5d0cec8dcc : SUCCESS in 12m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/caa7f878815d4e9fac8a9ea109a09887 : SUCCESS in 4m 46s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/28214e55fb5146368c929f645d376463 : SUCCESS in 5m 21s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/37febc2439104ddf8a8682bf3eda3ea1 : SUCCESS in 8m 55s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f7526d4752a24e189ab1f2e7c4c2f2a3 : SUCCESS in 11m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/234e47f518084c919df4f0966abf6305 : SUCCESS in 7m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e19ab5515ca84ebd847b733eed883f01 : SUCCESS in 8m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/eaff351bb98f4465a4fa7279b4a75b67 : SUCCESS in 12m 14s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0cf10e50f2214ed2a1f611efa1df409d : FAILURE in 4m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c0f02ad0f1af4803a0092edefbb82b3f : SUCCESS in 10m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2c1725df41b64bf8b713e754d0f509c5 : SUCCESS in 4m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5f165dc2c1a84d6597c15e13adb4404e : SUCCESS in 24m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/19fcfe78f646407bb9fc15263f894f8f : SUCCESS in 13m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f8560963637942b3837e73ad815ee968 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1cb66742fc2948b98e5ddec1a2702a6e : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e17007912d20400d854850dc20e48fd7 : SUCCESS in 20m 24s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a7cdb6b04bea49d08a9d88d8af92d728 : SUCCESS in 5m 30s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/44c5c3ebd3b0491988dd467a652b9b17 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f5aa2df062f2430a9a7ac4f4d1a422fd : SUCCESS in 6m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/08ec63d638ec46f88c51893b936c3b1f : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f03aeac4ab464108abdc95e0e05917da : SUCCESS in 54m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/db00ddf45b584c669a64560e78117096 : SUCCESS in 1h 34m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ba037811228170070a9f21daeb9c2695fa5f59c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-30 09:24:30.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"556abd1378579d3d4f842875efaafa38578f3e4b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 10:23:42.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/dafd30fb160448ab8f8b95d9b3d5d4e0 : FAILURE in 13m 18s","accounts_in_message":[],"_revision_number":3},{"id":"07f9fab4ad8e9c78dedf755c3360fcf19317a42d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 10:25:48.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eaabccc3cf84dfea62f7f6b2ad8693f : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f5d7d8f1874ac99db90c3c0040b4e4 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fd24135021db48d883088df36cae5c5f : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d11acccd97249f9952f3ac30ea1510b : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06cdbeb91c8c45318bfd43dea27ea444 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/471f8fb2f64a4376b1f43cd2cda3a7ef : SUCCESS in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c7a9943715694cd08363690c7263551f : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/61e1a7ab94dc4dadb2ac81c979d4602a : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/40e8c744a56644afa750eed98cd425b6 : SUCCESS in 14m 06s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/8df3d4d8f9394c1480f9a4341a94f526 : SUCCESS in 7m 57s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/75a5f2f647f940f5b71df9cc087a4c87 : SUCCESS in 5m 42s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/714550e9d02145d2b23fad5f7db73235 : SUCCESS in 6m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0278ad5b5cc34eeb9b0131e875c93601 : SUCCESS in 10m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1e2e3a2423714b0f959e625575f46905 : SUCCESS in 4m 50s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0197a3777da045759cfa77c82b2e5865 : SUCCESS in 6m 08s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/60b0e6db3bc94a7b8459398ce442f43e : SUCCESS in 9m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3e27e635a1fd4391936e1f129c6712e6 : SUCCESS in 12m 34s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/23d43ff7f197457db169e644812ea6ba : SUCCESS in 8m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a996cd18337e426fa1108e5377a74d3a : SUCCESS in 8m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/afe7ad28b9194aaf9c5fa667ac2fbf8a : SUCCESS in 11m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2b184ec3e7204c5889a7b9780dbc670b : SUCCESS in 4m 04s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/363e2adf5b124f008c6248ad66177bd3 : SUCCESS in 10m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e65a49961bf741109dcbd57c6d221626 : SUCCESS in 5m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/98cb33f87f174ba69125fd6118d7d8c4 : SUCCESS in 21m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8074d130b4fa49a6b2bb2d8bbcf220c1 : SUCCESS in 12m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/724368b054e743c78e9dce89199c4ee9 : SUCCESS in 7m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cf2d98e4197c46d0b33c04a48a2306e6 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4dc2abee21a544f19de64ac8db645e19 : SUCCESS in 21m 09s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ce370c2867b3454592df114c18756f54 : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/b205c772f0b64d85b69bddcf08ce4f84 : SUCCESS in 5m 51s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8732af21d28642e1a6bfe95cfed0220b : SUCCESS in 7m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2cdbd4f498d647b7b2a3a0294d274d8a : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5447174a5bc0419eb5626376b5128e79 : SUCCESS in 54m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5bee741ae2ab47b194f308277f7925c5 : FAILURE in 14m 15s","accounts_in_message":[],"_revision_number":3},{"id":"e65b177d56675b2f45b3d01f307c4ec48470df30","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-30 17:38:51.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"77b4e965d62b1df028c11906948a45da1dfd7753","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 18:05:40.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bb02eaf759b84f12925d1cb4a42a1172 : FAILURE in 7m 07s","accounts_in_message":[],"_revision_number":3},{"id":"f304b74eeede92bd16aaf6339cf67f607bcb8b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 18:13:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b2abe4df6e04aa3a617a937c1f55b59 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0c474d615da4386948d3d1307932dc3 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a07e43d79a94ada9a8a083cb5a6d2a6 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a91561e12ca42559215e706617fec98 : SUCCESS in 4m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/abdac908fb004301a7764a9e1273d366 : SUCCESS in 4m 32s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a85d025c47cd47ad9e934ef7bbce2d71 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/988446ae436f47c0bcb7607956ceedf4 : SUCCESS in 4m 45s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9890fe5d2ca14752a19272321f3ef0c9 : FAILURE in 5m 07s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/fb45eb1d4e644d3db29b2ddd46700520 : FAILURE in 5m 02s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d8721fe1e21f404e84725efb85dfe1a3 : SUCCESS in 9m 20s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/52fbb3ac37884da6a1e22506a2df96d9 : SUCCESS in 5m 50s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/8f14074d51694075b8f9b809424f3511 : SUCCESS in 6m 10s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1ac667701ffd4346b1bfb55e7cfad5d4 : SUCCESS in 11m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8b9c277cc890474c8faeeca63c3637da : SUCCESS in 5m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e832cabb4df94bad8d8881c47c72a868 : SUCCESS in 5m 59s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/38836d250f494e29bf9b7fc4ec484678 : SUCCESS in 9m 30s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8d4a039198844f8994c8ced16ea6c0e7 : SUCCESS in 10m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6fc88b58bd0a4a78b1e2f818556ac7e4 : SUCCESS in 8m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8bc69d8343144864919a0eb3151f33fe : SUCCESS in 9m 14s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b59de248de1d4a6abbad65203e7cf074 : SUCCESS in 11m 55s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e052c6902cca47babed47dd2d40f46e8 : SUCCESS in 3m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c6e284302686435383a70a81091a3ee2 : SUCCESS in 12m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7dd258870fd042398d56d41a5956907c : SUCCESS in 5m 43s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f2a8b2f18bcc49fe96432c398db090b7 : SUCCESS in 28m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/645c5f73ee74459da6ed5c1dac175344 : SUCCESS in 12m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/89d6a93c04eb44e2873f29d221cbb5ce : SUCCESS in 5m 05s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/93b8fef2bf9f470db7b20dd538ef6a79 : SUCCESS in 3m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/76366790c4e946aaa28310d7f9565a27 : SUCCESS in 25m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2265bf8909ac4ec2932038aba06fee0e : SUCCESS in 6m 29s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5644ecf896c64ae6bb7562365c27a125 : SUCCESS in 5m 06s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/37e07fe288164807b5408ff851a3e5eb : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/20f8b2526df148a4bee4ad25ad632fac : SUCCESS in 11m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/44419f9552664515933a6483c4b8f9bc : POST_FAILURE in 15m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e703f39030bd440f82173d2d31def460 : FAILURE in 12m 28s","accounts_in_message":[],"_revision_number":3},{"id":"4b65a7e99c6a880fbdfb0a793dab98d97ac3d049","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-07-01 13:44:25.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"97a146df2d2f0f411a59292ef969d6ad6b894fbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 13:58:07.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f8ca816e29ec4ddea8826d0c5b3cd398 : FAILURE in 10m 59s","accounts_in_message":[],"_revision_number":4},{"id":"88172cd978e042ce11ba0ed0e55adfd516d3b4cf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-01 15:13:15.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"d2e3dbfe6cd3bc7f85d62ae22251dca8ba17aa9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 15:17:48.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8842c4fbb2cc4a2dbcc5a7e90333a4b2 : SUCCESS in 3m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fe1c7f71fe242afb4a126b60d53b965 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fd3259e384b4546bc977c8a2107f6a8 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/518786142b5d4d8bbd1125e0a1964ae6 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0608d8bda0246438e9d35ca8a930ddb : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b7fb155eb49e428c92b768209191beb6 : SUCCESS in 3m 40s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/29d569880c3e4ba993368bf8d855cf0c : SUCCESS in 3m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/64d49aa7f03944b58bedd3465e0d825c : FAILURE in 4m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0cb5c19f995b496f905b7096fb893d0d : FAILURE in 8m 26s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/5aaf8ac59a1e45e09365d5ea96133781 : SUCCESS in 7m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a8bb15006cdd4115b84eb887be302274 : SUCCESS in 6m 12s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/9d2503bf9db74745ba6ed3ea53b18eda : SUCCESS in 6m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e3ce7a8bf4df4daa8f2ac3ee973f8ec7 : FAILURE in 4m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a676d1eea0bb43b582450487e314b59d : SUCCESS in 4m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b227e752d77a49bb994c5b675ba5d9a3 : FAILURE in 3m 46s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1c078bfd2996430881f058116ab6fba3 : SUCCESS in 7m 28s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9f8ee14567834105b4b70c6fc0062f2d : SUCCESS in 11m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1cd7596b3a4b4392921c083f0213b8a6 : SUCCESS in 7m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/63754a30345f4b6a9b4fc15f61b8e0be : SUCCESS in 8m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e481841b85754529bd8b4f11d9d9b7a7 : SUCCESS in 10m 58s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e3de8d4f62254f2994a43d5fa7b41154 : SUCCESS in 3m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3d9040ac0d6446ecae7982fece6d2b63 : SUCCESS in 9m 33s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/05a166e256a24efb9a3fe6c6e052404b : SUCCESS in 4m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/be43bd7569c14acba34b00850b3ec947 : SUCCESS in 26m 21s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c850a3c159f84f8fa41a78372feb170c : SUCCESS in 13m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6754458c4e424e689f59ce0665e19b6d : SUCCESS in 5m 45s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fdbfb89b39534888af2728c8e71c4a67 : SUCCESS in 5m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f7d64bf31f134b93bf2f605eb1c162bf : SUCCESS in 28m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e9e9c429beb042c69932bf915a753dda : SUCCESS in 5m 48s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4e87d130d5cb4d0f86d121ed0f0f3874 : SUCCESS in 6m 03s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/142bb4c2c74845929bde8bb1aea4d385 : SUCCESS in 6m 53s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/94cd1191dfc0417b8e0b1d64efc9eed2 : FAILURE in 3m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28962c916bc44f69b9814fcff5449e1b : SUCCESS in 49m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/07f9459fc7d64790aea0ec7cbfe3a886 : SUCCESS in 1h 31m 53s","accounts_in_message":[],"_revision_number":4},{"id":"f00eb870130bc72ed17bd13cc7f94d439c124df6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-01 15:25:00.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/54a784c77c084438b660c4e3c07de6c7 : FAILURE in 9m 38s","accounts_in_message":[],"_revision_number":4},{"id":"bb3b6d8b6dec2d3e7962886b781052f4c8ed15dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-07-02 08:17:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"439b199c85b5e3a026a76aaac1460cbe220705b5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 08:27:32.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b1b6ab29c60e4cd69b8a0d9f9ff79883 : FAILURE in 7m 05s","accounts_in_message":[],"_revision_number":5},{"id":"289398392b07ca6a8c9f1311f7287c9f06635354","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-02 10:05:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/085f38b2407d49e0bb8de9fe55d4c4aa : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aab28d680335461ab773e49d4a933e14 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fce767322d074639ba3e65cfcec40f67 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c8cdcbd42fe481a83936dbf1af4cba1 : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fb71d651ca540139a4b66d69f9adea6 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67f276225cbf491fb5afb205aacb0af0 : SUCCESS in 3m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a778fa946ed844468cc405bb2e18f504 : SUCCESS in 3m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/85525821d71f4a3eb30da534808226a2 : SUCCESS in 7m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/166735ed00424cb3b97712c54c0957d2 : SUCCESS in 10m 21s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ba6ff8d38f7047e98eef9af665c56ac9 : SUCCESS in 7m 26s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/99010803d6894b939b78058a7dcf3a73 : SUCCESS in 4m 28s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/34116a009a7244b385b2de56942af892 : SUCCESS in 5m 00s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e6b555c767a449dfbc0f35d47b75bebc : FAILURE in 8m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8da1d6513d654119b3658bbce28af4a6 : SUCCESS in 4m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/598a36bfebc846d889de585f4c33822e : SUCCESS in 4m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b6160e286c684e5396ef33adfdc5bbd4 : SUCCESS in 8m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7adc4791a0a5435badf14fec870304c8 : SUCCESS in 7m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d5b08c9ed4d0499b9ee67eff4b537ac7 : SUCCESS in 8m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/37c326d1e8ea42e58af59da8d1122ac9 : SUCCESS in 6m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1356f976632343a0a98eb651caaa58a3 : SUCCESS in 12m 44s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ca004a168e3546e1a7f48ae1a583f3e8 : SUCCESS in 3m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0bb97383ad3a4601808f22296d7b0ae3 : SUCCESS in 10m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/265ecdd52b4c4ae7b1f624941617e6be : SUCCESS in 4m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a66a81619ae942cd9deb0bb97446a924 : SUCCESS in 26m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/651d4d3c3e6a4f2098930792f6b40d91 : SUCCESS in 12m 31s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/21f9585beb234d6b87c3cb72ddba52be : SUCCESS in 5m 36s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8f2921bdc66148419257ba367cf5cca4 : SUCCESS in 3m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2573dddade1a48438b5e347c01853a2e : SUCCESS in 18m 14s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b220cdcbf41f49c9ba40ecd78e42865b : SUCCESS in 5m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/665e7418893f40bbab45a90a380d9a5f : SUCCESS in 5m 07s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/1aff2390606c41cc9e31cf00d180e23c : SUCCESS in 6m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4d36e3f2f42b4ec18b81fe4b7d9f881d : SUCCESS in 12m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa0162e562a042f7a2c448cd7e5ede7c : SUCCESS in 34m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a426f27bc214a8983db4c7dc624b16a : SUCCESS in 1h 42m 35s","accounts_in_message":[],"_revision_number":5},{"id":"20d51f80f53744e940bfbfe188aadc307eb24d33","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-07-03 09:41:10.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e847b3847e335805a07e755be4b894d18de5db0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 09:54:49.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c247818486914dcda51b30468f28b27b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/63b22d21e63646eeba78c268c444cd8b : FAILURE in 7m 47s","accounts_in_message":[],"_revision_number":5},{"id":"c9016d7d9eec5fec97c61271364112dc72965ab8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 10:59:12.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239c780add2548c9b05d9ae3dbb0361f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e00dde0472e4ae592e99c6e06e23bdd : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f028d77d0ea84b6ca9078ac366d1c752 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4930d08e72c94848baa60f392c8656f8 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/417cd731db1045fd80513cc2962c57a6 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5befb6add6be42f3b1ee76a86bd08d66 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2840a49ab761458eb313dcca9b663fde : SUCCESS in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e508df4f6fcf409cbe5b849f3b801d46 : SUCCESS in 3m 46s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3307e0308630497e89e7f1916737312f : SUCCESS in 8m 45s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/2c91256139aa49e7b3118b8988360b41 : SUCCESS in 11m 25s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/49487f61fd9d4e4e97478e167d0e9018 : SUCCESS in 8m 27s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f6e006f2a8d741aab38bfa5269d3ae2c : SUCCESS in 4m 41s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/8fbfd68a81364f959347ead8a7f60db8 : SUCCESS in 5m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8541bd7f2f59476da6e2aa0dfc8af5ed : SUCCESS in 10m 01s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/837d37d8355d4b59af639fed7c6bd1be : SUCCESS in 4m 34s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/afa945aa2e434dfd9b46e97bacb129ab : SUCCESS in 5m 18s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d14e809daa4c4990aeeef2acdd8d60bf : SUCCESS in 7m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a6a2eab3242842f7b4fcf89a7135f7b1 : SUCCESS in 8m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db302f904ba541029073c155edc7fb59 : SUCCESS in 7m 19s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/98876506c7334ae38e9a88a0340c0fe7 : SUCCESS in 7m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/84e1899a33f84876999ef6c45689a4c7 : SUCCESS in 11m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9de2c28b04bb483f8882b912589ea852 : SUCCESS in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/632e8e17692e4bc798a781c45db3550a : SUCCESS in 10m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eb782f25594f40119e29ba404fc45267 : SUCCESS in 5m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e616105aa09640379a73dd67751442e1 : SUCCESS in 19m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/67247d5ceebf44e7a1ed479b562afc83 : SUCCESS in 11m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7e13d63c10144510996925fbc2033dcf : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ed1cb48ad3b5498fb4261ac8eddb95df : SUCCESS in 3m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5a1c27f02fad4321a6e9640e731e5d54 : SUCCESS in 21m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f314088b8122437faf9c20a07233cf03 : SUCCESS in 6m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c1f95534ca294702b6b44a0653ef75a2 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a082a52ff3a34888b48d864669ece48c : SUCCESS in 7m 21s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/25433479cd204ffc84cce46d74d0a2eb : SUCCESS in 11m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d2cb905abf94376bfd88a0717a7bd12 : SUCCESS in 37m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09a8aa830812436181fb99ec39d02fae : SUCCESS in 1h 16m 04s","accounts_in_message":[],"_revision_number":5},{"id":"6a774e6070a517886803e1cd35d755ace080ccc6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-03 16:23:03.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"d71f7cfe53e0d283ab03beac682822a5e723a50e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 16:23:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8fc3f24748a4f847fceab9b79d96e6153506f0eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 17:43:00.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daea554770ee47b4931dc3f1942b1fd3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18e65fa9165b42cbb675124bec10a552 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb1efe7d7e514345ab11b72413c37303 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f16c2cc1ffb24132921c65bcc6815380 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93c756741897419380ac630a071f0e74 : SUCCESS in 4m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/865b678fd944465eafe564bacba445b9 : SUCCESS in 3m 32s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f97baef3bc604b2e9f1f28c7af99d154 : SUCCESS in 3m 58s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce0b5b6b39314b2783a31461bd156402 : SUCCESS in 9m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/577589164547487ba548b18262dc3661 : SUCCESS in 14m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4516d079ebcb4d45a5755715b67b596c : SUCCESS in 10m 25s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/138c710a1aa34335b52f8ca69e408514 : SUCCESS in 4m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/42a4d2ec8be34224bdbda23f8b559715 : SUCCESS in 5m 43s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ec445bfb999141079863174c6a79d2ec : SUCCESS in 6m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9865b4454aad484ab71bbc5ee6b1a230 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b4225f12195643f7aa5f8d6be8a37b51 : SUCCESS in 7m 26s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cefd9d27a21a47b4865ecea99dbac77a : SUCCESS in 8m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/837107c13ace462fb6a4e3bc80250c24 : SUCCESS in 10m 48s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/aac8cfe935ab4c9996d02fdeb3300b81 : SUCCESS in 4m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/be6e221c56774ceba4b34b4accdb5270 : SUCCESS in 8m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/079431be80b2418dba75084583b4a3ea : SUCCESS in 5m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ca6af436b0074d2bbfe1ad0ed827a573 : SUCCESS in 17m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/36e35b54a523473da8e33eb74a839924 : SUCCESS in 11m 04s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2322f54bc39e4622b6033244f203deb7 : SUCCESS in 5m 46s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4adeb8a58e4e4f4b952bfe7914585eba : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/de0160419b7542cfa58a9ae3bd5a1e1e : SUCCESS in 17m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a3f05af907d14fd98cc4ff76477f567b : SUCCESS in 5m 13s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ea3c5ad609b0427f8869ed9e565277fd : SUCCESS in 5m 29s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3fd8858cfd32404d89e79d2aea915d97 : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e2c3b67884da4125b8f84bfc8c3ae570 : SUCCESS in 10m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdb4d040fe1847928f3c4b5adb30b336 : SUCCESS in 47m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/aa4b26199a45431f8c5236e61af4a3b6 : SUCCESS in 1h 18m 36s","accounts_in_message":[],"_revision_number":5},{"id":"c2e46763ca6fa906af335bb6305c248609e2a993","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 17:43:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"09e8fb36e3dccaa90356fb7d198938a8674e64d6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 17:44:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24016103ac3a4771898854d3c3203912\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99157f6ae2614709ac02fcb312e3e8d6 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"70282106c9d00df1ec0b6bfb830bf1858a2ebad5","revisions":{"7eac48bf99791a72349ec9163154e4183b841fce":{"kind":"REWORK","_number":1,"created":"2022-06-30 02:53:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/09/848209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/848209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/848209/1"}}},"commit":{"parents":[{"commit":"e592841dcac7f621b7d3dc492fc7292cb33f892f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 02:52:58.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 02:52:58.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0dc814b74ab10eb9400a2a1fa6a220d454da755\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7eac48bf99791a72349ec9163154e4183b841fce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/7eac48bf99791a72349ec9163154e4183b841fce"}]},"branch":"refs/heads/master"},"bd450b6a7990649d825d6e3cc814b82097d3b348":{"kind":"REWORK","_number":2,"created":"2022-06-30 03:23:26.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/09/848209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/848209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/848209/2"}}},"commit":{"parents":[{"commit":"e592841dcac7f621b7d3dc492fc7292cb33f892f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 02:52:58.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-06-30 03:23:08.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0dc814b74ab10eb9400a2a1fa6a220d454da755\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd450b6a7990649d825d6e3cc814b82097d3b348"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bd450b6a7990649d825d6e3cc814b82097d3b348"}]},"branch":"refs/heads/master"},"fb89c7699cf5194820cd7f52712ba4a9c2cda684":{"kind":"REWORK","_number":3,"created":"2022-06-30 09:24:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/848209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/848209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/848209/3"}}},"commit":{"parents":[{"commit":"e592841dcac7f621b7d3dc492fc7292cb33f892f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 02:52:58.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-06-30 09:24:30.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0dc814b74ab10eb9400a2a1fa6a220d454da755\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb89c7699cf5194820cd7f52712ba4a9c2cda684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fb89c7699cf5194820cd7f52712ba4a9c2cda684"}]},"branch":"refs/heads/master"},"43dccacc42cb40c770e5206d9f411f44b42a69a4":{"kind":"REWORK","_number":4,"created":"2022-07-01 13:44:25.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/848209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/848209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/848209/4"}}},"commit":{"parents":[{"commit":"e592841dcac7f621b7d3dc492fc7292cb33f892f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 02:52:58.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-07-01 13:44:25.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0dc814b74ab10eb9400a2a1fa6a220d454da755\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43dccacc42cb40c770e5206d9f411f44b42a69a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/43dccacc42cb40c770e5206d9f411f44b42a69a4"}]},"branch":"refs/heads/master"},"70282106c9d00df1ec0b6bfb830bf1858a2ebad5":{"kind":"REWORK","_number":5,"created":"2022-07-02 08:17:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/09/848209/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/09/848209/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/09/848209/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/09/848209/5"}}},"commit":{"parents":[{"commit":"e592841dcac7f621b7d3dc492fc7292cb33f892f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 02:52:58.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-07-02 08:16:42.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0dc814b74ab10eb9400a2a1fa6a220d454da755\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70282106c9d00df1ec0b6bfb830bf1858a2ebad5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70282106c9d00df1ec0b6bfb830bf1858a2ebad5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
