)]}'
{"id":"openstack%2Frequirements~848232","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~Ibfaae1890708438a8b03c961f55c88dc8c7acd17","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"Ibfaae1890708438a8b03c961f55c88dc8c7acd17","subject":"update constraint for horizon to new release 19.3.0","status":"MERGED","created":"2022-06-30 08:55:23.000000000","updated":"2022-06-30 16:37:48.000000000","submitted":"2022-06-30 16:37:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848232-new-release","meta_rev_id":"3e93ee6641f11a86a5649d0ad673a2997ddba3b4","_number":848232,"virtual_id_number":848232,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-06-30 16:37:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-06-30 14:33:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-06-30 09:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-06-30 14:33:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-30 09:25:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-06-30 10:28:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-30 14:33:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"48b0c1874d4b9f69d3110d9189622cd5084566be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-06-30 08:55:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae8ed56fadebb76b9b5b5f473a5cb918de499c43","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-06-30 09:25:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3f0f1f69d40761f4a1146bd9a6eb4942b9ef4380","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 10:28:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70069c60b5549dcaecf192458b47c02 : SUCCESS in 3m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/46173ed996274b2fa1de0c15865c52f5 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/523b07c5bd44496bb470fc3b36636ec2 : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de9be7d3506d46c8979247806511bbbf : SUCCESS in 5m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/099faec202b04f05bed06b1000b41304 : SUCCESS in 3m 24s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cc8635764cd64c658ff4edf3d174731a : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/7e343803c0cb4fd5a90645b27f0004a7 : SUCCESS in 18m 22s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce115bc54ae14f959c03696664e1e252 : SUCCESS in 7m 26s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/616e857b1ff445289660c9794026ef17 : SUCCESS in 11m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/1112c75f44154dc5ac3abd3fff451881 : SUCCESS in 5m 28s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/6b2de38dd24444168ee1dfc10e2fb7e6 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/87b98a67abdc451e8f2d22516ed3015f : SUCCESS in 7m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/854086f41fe34ffabfce615ada8d2c5d : SUCCESS in 8m 09s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/1b7c2371304f46408a69801e9219fa04 : SUCCESS in 11m 55s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/25eabacf312d45bc8fd1cc4d2e17646e : SUCCESS in 8m 13s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/7bdc607fa3d545758b33efdd1be18939 : SUCCESS in 23m 16s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/da191cd6c6e04008b5cab986362cce5b : SUCCESS in 12m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/614a499361764e2f815faef0232b0373 : SUCCESS in 3m 40s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8814a8b32b414696b56cd32c4d124b7e : SUCCESS in 16m 09s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/4b3f0c1f1ee3400ab2857016589d425c : SUCCESS in 7m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b7df2c092bfe4481916ce94ccd81287a : SUCCESS in 6m 51s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/ea5852b7354c422c8dd1b40f935335e1 : SUCCESS in 9m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/94c0010b3464469b82c54a6aa5a237dc : SUCCESS in 55m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ead7c0aedb5342ada104b78740048c88 : SUCCESS in 1h 30m 50s","accounts_in_message":[],"_revision_number":1},{"id":"d734f5f48d9351c31a1dc55bdbae341f899ea6af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 10:56:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/31408ae1365844c2953a13c4c62c98ed : FAILURE in 2h 00m 01s","accounts_in_message":[],"_revision_number":1},{"id":"fe9e1fdc940da7024e3c960c82fb882568c15858","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-06-30 14:33:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d54af218d5b7cabb8293059a276312f033329dc7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 14:34:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"af249ba84a252fb95baa07c0f68a84d653537d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 16:37:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c233592cfed448c80dacd765b19302b : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/effd01db33e2495e8a87d3fb475ed2da : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/74ac083f3970440e87de39a41f05b6c2 : SUCCESS in 5m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/22c85bbe5b3049868606c46169068572 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c9583bef157144ed8de12cd4902dc0fa : SUCCESS in 4m 11s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/0ea54355d4da4172aebac26087e4f175 : SUCCESS in 10m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d38cd7fa285445e39750f79b3aa802f1 : SUCCESS in 9m 38s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/b3c03f05fc2f4fa4a153efd59c659e8d : SUCCESS in 11m 45s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/247dc8fb7256488398f725d2810a9c23 : SUCCESS in 7m 31s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/cc2fe0bc4f6d430cbf8d9ad56948dcb2 : SUCCESS in 12m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/52097f9d973a4a8c847f00cb2ed758b3 : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/57eb089527084399be1c37e9cbeb4e0c : SUCCESS in 10m 01s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/e55a730eca114bfb9e7c6d648111d73e : SUCCESS in 11m 59s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/9eebc360c0e5457aabaf1dcd1d668961 : SUCCESS in 4m 42s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/c9e12620d45044438e42d1bbd4a724e3 : SUCCESS in 26m 31s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/6ad957f4ad13412ea6b4f97bcd8cb6e1 : SUCCESS in 12m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d78fb0fa27f84707a488cfeda91eaf81 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/03f2cf6b7eb74f54befeb5ca5f498680 : SUCCESS in 16m 14s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/ce5496cbcc10426eaed880cf5a591728 : SUCCESS in 6m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/31562ccbfbd744e2ab8faa88e5648920 : SUCCESS in 6m 04s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/1b39a806c7714e72accbeb9f3651cee4 : SUCCESS in 13m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/809991d032fc4fa1bd963d0e8b497034 : SUCCESS in 47m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/08aa84f8a53c408fb6e7d0a34bf1fcf2 : SUCCESS in 1h 43m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3e93ee6641f11a86a5649d0ad673a2997ddba3b4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-30 16:37:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41c274af44c8d463da38baa0722c1d12e9a87764","revisions":{"41c274af44c8d463da38baa0722c1d12e9a87764":{"kind":"REWORK","_number":1,"created":"2022-06-30 08:55:23.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/32/848232/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/848232/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/848232/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/848232/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/848232/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/848232/1"}}},"commit":{"parents":[{"commit":"a0f7bcc52735e8034b98f11964558e921e81f223","subject":"update constraint for glance_store to new release 2.5.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a0f7bcc52735e8034b98f11964558e921e81f223"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 08:55:22.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-06-30 08:55:22.000000000","tz":0},"subject":"update constraint for horizon to new release 19.3.0","message":"update constraint for horizon to new release 19.3.0\n\nmeta: version: 19.3.0\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Előd Illés \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I89887738a4da9fd1beefa210980479d2508c2227\nmeta: release:Code-Review+1: Vishal Manchanda \u003cmanchandavishal143@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: Ibfaae1890708438a8b03c961f55c88dc8c7acd17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41c274af44c8d463da38baa0722c1d12e9a87764"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41c274af44c8d463da38baa0722c1d12e9a87764"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
