)]}'
{"id":"openstack%2Frequirements~848526","triplet_id":"openstack%2Frequirements~master~Iaf0d98566d6993bba56ee990b95663e627de2778","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Iaf0d98566d6993bba56ee990b95663e627de2778","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-07-03 02:44:54.000000000","updated":"2022-07-03 17:10:01.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a4c48689769d2eef1611033fb5d65db00c7c952e","_number":848526,"virtual_id_number":848526,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-03 04:20:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-03 03:00:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-03 04:20:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0aca69be3878fb8b5a5d21d75e02bbfa5d173a4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-03 02:44:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a99033670b3fd8b99e0fb7a178ec396bfc22945e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 03:00:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/415092d6fcff4009a6d1a1589a11e11c\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/085b224dce3241be88d343aa77df7a67 : FAILURE in 8m 04s","accounts_in_message":[],"_revision_number":1},{"id":"acd84cb05bcc772b1923111e4e19f4b832310e98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-03 04:20:44.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d689e8e3d4fc4b899ede0ca1718ba5fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/796781244e144b03973a697c400d1c2b : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f562f3907cd4496bded1bd0e173402c : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eca46f3047f64f7183a4bf5c793616ed : SUCCESS in 5m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8b0c7d2f8784de48261058b6029cb22 : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d492cb492544fd09e07610dea75a762 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/42183c73496e49748242c562dde633dc : FAILURE in 4m 30s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d262fd7f9ad54f858a3820dc821cfef7 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0c606bbec1734e4aaf71a72f34f460c0 : FAILURE in 3m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/97b0993a1cdf4376b8be4d60db5fbc49 : FAILURE in 3m 58s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/728c7b108a73403098f8da107f9d14fe : SUCCESS in 8m 53s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6229484bd8a9475baa319ebd753a5755 : FAILURE in 5m 49s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/64569d2d7e1448b1a61f3601a9b09e64 : SUCCESS in 6m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b9e4513bb7ba4bf286fe169c7d398ff2 : SUCCESS in 10m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3459ddb7659f4cdca118219a041f0740 : SUCCESS in 5m 02s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/992068832ea44cfcac84b28723eb1a64 : SUCCESS in 5m 52s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0a98011cc4324d86b29c7c250eb82f0f : SUCCESS in 8m 38s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b6802bfbb52b4cd29033276f35d7905f : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1915b72b3ded460fa7b9dc2fe94b4822 : SUCCESS in 10m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3b157be7263a4507b44d0c98ca70fb75 : SUCCESS in 10m 24s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/78d992e1692a46debc49b97dc8a1d55d : SUCCESS in 11m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7bc1090e1fe3495489fb43ecd47a68d9 : FAILURE in 3m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0ecff838bc2746cbaf1f309cbf341d3a : SUCCESS in 11m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/da688c714484447089323f9e8613d59c : SUCCESS in 5m 39s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/01d6f5db58b54ac0a655517db618308d : SUCCESS in 21m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/163d6550a5a34471ab4ce5d3e2f59c52 : SUCCESS in 12m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6d00f18b5ee646f18ebf98cf63f62336 : SUCCESS in 6m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/451f5078d4954f0a8ea6d3b8e0dc6069 : SUCCESS in 4m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/081b4c873e474b9dbf35868a06af5a93 : SUCCESS in 21m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/57206f05d99e4ce39a5e3a9bfe58b6f1 : SUCCESS in 5m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/932725fe9a0d40568d6adaea7e4ee71d : SUCCESS in 5m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c96534bd8b524a19a33e3e99f9cf2d44 : FAILURE in 4m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1b9f012348f44cc6b16b06359c81e367 : SUCCESS in 9m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/95c82fd1ea0548d6a8f099304c07a8ea : SUCCESS in 43m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/192402cfae29469eaf5c2570dcd82826 : SUCCESS in 1h 24m 50s","accounts_in_message":[],"_revision_number":1},{"id":"a4c48689769d2eef1611033fb5d65db00c7c952e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-03 17:10:01.000000000","message":"Abandoned\n\nuse the other one","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"92e91c6aa70024d2a4966466a6856816f92e3240","revisions":{"92e91c6aa70024d2a4966466a6856816f92e3240":{"kind":"REWORK","_number":1,"created":"2022-07-03 02:44:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/26/848526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/26/848526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/848526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/848526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/26/848526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/26/848526/1"}}},"commit":{"parents":[{"commit":"e592841dcac7f621b7d3dc492fc7292cb33f892f","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e592841dcac7f621b7d3dc492fc7292cb33f892f"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-03 02:44:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-03 02:44:53.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: If0dc814b74ab10eb9400a2a1fa6a220d454da755\nIc0269dc272807f3a4801adb6c30e332eb73b1030\n\nChange-Id: Iaf0d98566d6993bba56ee990b95663e627de2778\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e91c6aa70024d2a4966466a6856816f92e3240"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/92e91c6aa70024d2a4966466a6856816f92e3240"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
