)]}'
{"id":"openstack%2Frequirements~848557","triplet_id":"openstack%2Frequirements~master~I2cd124c546b7af278284aef5e5ac745fa7fb563f","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-07-04 16:44:19.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I2cd124c546b7af278284aef5e5ac745fa7fb563f","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-07-04 02:44:56.000000000","updated":"2022-07-04 18:23:50.000000000","submitted":"2022-07-04 18:22:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848557-openstack/requirements/constraints/noclob","meta_rev_id":"9d26ef7a012b7629b3ab1b73084adf9951252973","_number":848557,"virtual_id_number":848557,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 18:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-04 16:44:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-04 16:44:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 02:59:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-04 03:08:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-07-04 04:36:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1266e02f817f987f8c76e698f2c1df6f6dfdd92f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-04 02:44:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d8690635b4ad23d1318e0e6b1c608abb81e79564","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 02:59:49.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d82b37b7aaba48ccaff6beb21d15d289\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/132042ecb1f44df6a005051b7edad618 : FAILURE in 7m 52s","accounts_in_message":[],"_revision_number":1},{"id":"106208e309a8d7eb09f53b578271515ff639acba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-04 03:08:20.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3ccd930c8b7e598282c1456ebfe1a642ccafa2cd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-04 03:08:32.000000000","message":"Topic changed from openstack/requirements/constraints to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":2},{"id":"8e062d727e4a7f1a36869ecb446619837ad43d93","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 03:20:48.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c9ca52e016b6447f83b5048f931fb855\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5339bafe382441bea5eb22f137e971de : FAILURE in 10m 46s","accounts_in_message":[],"_revision_number":2},{"id":"b1a8b03f3c7a5b77501c343039ddb5133b9bb908","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 04:36:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e38fba68c3d24aa8b3e9ff80bf105039\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fdb5e93eab64d5290dd984ef1ea31bf : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3b267a39a154a41a0b1e4c3af9bfd2a : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/186161ad8e38443dbb449aab86d7c35f : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/48fbb8f8c85e493b869c2c32dd8421f1 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a6ae8f085d8491cba66ac7088ad457b : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/91a5a2366fc948328009419a7842954b : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55bf8fae5c8d4f12b030bade426e32bc : SUCCESS in 4m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f0da2a8b09dd4ea99fed43a1552b8efe : SUCCESS in 10m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e6fc14eacfc242a5806b16dba909cdda : SUCCESS in 12m 33s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a24c7c677bfd4d84a1fce76bd6711077 : SUCCESS in 6m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9d37c4e40a024c50a1f2c024c9d193e6 : SUCCESS in 5m 59s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/9b8833803b70410e805fab4f6916ef94 : SUCCESS in 6m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/265c1e6edc3643e5be60d1e257891cee : SUCCESS in 12m 43s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8d172d4941604a70bea375d7afc838ef : SUCCESS in 4m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8d94841417434ac8bc6cfd27a681652c : SUCCESS in 5m 41s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7b04f9f3927043c58e2b7f5213cb948f : SUCCESS in 6m 51s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b8cefc7f03424cb3a9d0348ea5f4b3fc : SUCCESS in 9m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/577cf528d41546ce96198bb3ee5a7f74 : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/475e836948ff4122bd1e8281214a31ce : SUCCESS in 9m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/28904767151d4d89af00d9dd62ca643d : SUCCESS in 12m 03s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e051d50cb03842c790aa0c01d90f459b : FAILURE in 3m 13s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/37c8d03a748c476c8e9f7b3f9a2f0633 : SUCCESS in 9m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f36615b6bf9746d38b608e6d0a3c279f : SUCCESS in 4m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4a3d7fc2203447479b0ad73effe889b8 : SUCCESS in 21m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e90fb035df394c2ebddfd16407ea33a0 : SUCCESS in 12m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/789bc3a7bcc649679121af73ad2d11a4 : SUCCESS in 4m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ecaeb60d603a4ffdb5f99e559408846b : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c64a4fb2bf374820be20f4aed2b7544b : SUCCESS in 19m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/59c77771a4904b2fb170f6793d956e3b : SUCCESS in 5m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ce418df6e3bb4d5f955e30ee4b97492e : SUCCESS in 5m 20s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ec1e8cb0689c40bc8b388426f6752a94 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f7ebb3db254f42938d7bcf5e064ee4aa : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d6e5cc4767fa458b8cefca3c1f8631d2 : SUCCESS in 35m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d5276b7afb640b88461e4895e78a14c : SUCCESS in 1h 25m 49s","accounts_in_message":[],"_revision_number":2},{"id":"749a9c7a2d7b87d5a224a1bf4fede8d8b89d9c95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-04 05:22:23.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"d0afdaedc635811818cdfd9d45c6992d54376da4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 05:33:32.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d302257be2c54b96b12038e2518c0eb0\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/43462a2e48d94582bffac501225b3cdf : FAILURE in 7m 20s","accounts_in_message":[],"_revision_number":3},{"id":"72760d6f164f43abe34721d36a2a1b2111aa26e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 06:55:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d7f3f41c2c541f0a3bdc4c35e94890a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd12da74c6cd433ba94c61abef520043 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb6717f8b2414445a8a03727b01699c5 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd7b4f38f5a24105b997a0df0f0a788b : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61de50fb679e4e29a6e9d79ca630ad9f : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebd815a44d15434aa11b3cd10f620e69 : SUCCESS in 3m 23s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c18f8128c84a445daf08d634f712ecde : SUCCESS in 2m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8452f22ceee34ee9ac3f02f860e0a12d : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8f95c0749b884108a7eed089afaa4700 : SUCCESS in 8m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/60fb4ec17cd04f23a0ee34f22988f339 : SUCCESS in 12m 32s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e792450e63024b42a653182d55eb0015 : SUCCESS in 6m 55s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/00396f8abaa74112b128c5e458fe0bc2 : SUCCESS in 5m 31s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/7cf751838afd4ab5a6503863ed2d60fe : SUCCESS in 6m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e2a7761ebec743b898c4ab305bffff4c : SUCCESS in 10m 01s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/48968365a5e44809bc55c753ca2e0045 : SUCCESS in 5m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e3c6cd86e42b4e0daeffaf5d099bd13a : SUCCESS in 5m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bbb66a67a6ed4d15ae7dab8f9df21b3c : SUCCESS in 7m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/afb3ca2e75044de1b798f9e4a26f8e8d : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/5ae05c79a45e4024aa6e67558069d0eb : SUCCESS in 9m 02s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1d13e0ee7c9d4178a649210b45e05306 : SUCCESS in 7m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4b09f062ecff4ce5860d021a3ff4d105 : SUCCESS in 11m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d757bff018e2467a9a8c8f78db538c10 : SUCCESS in 4m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ee10ed5cd4d24cb19fa16542a60bc8d3 : SUCCESS in 7m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bab7bf36d96c44c4bf4b3982a097bae0 : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d1670e2df86844ab9cc13c770c1e65e4 : SUCCESS in 20m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0328d5b6bfe94936b6c1992d3c088340 : SUCCESS in 10m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3714123fdf314235a98f4bb389eac18b : SUCCESS in 5m 24s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6dc5f2ffad6c4a0a942ae00bedaea77e : SUCCESS in 3m 02s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3972f00ca4014ceba4ff60f699059eea : SUCCESS in 18m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/eef6b4948e6e458da87fb3afd5481e73 : SUCCESS in 5m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fcd0579af81a480e98b070ea28c53548 : SUCCESS in 5m 00s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/292dbb8b326b4531b30d78a36c10d65e : SUCCESS in 6m 08s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1eaeb344063142d9bfe4be7000431a68 : SUCCESS in 10m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fbdd6b944064492b8a9d970d80948d5 : SUCCESS in 45m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7eb22d8606bf4fbcb7d5d7c5280dee5f : SUCCESS in 1h 27m 54s","accounts_in_message":[],"_revision_number":3},{"id":"70fd43d01282a6f979d8e3c9ebefd55b0419ddeb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-04 16:44:19.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"367e08ed4417140d87887ac507421299473cee9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 16:46:17.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d7f6d75990df5c8eb977bcff72d92d588234f24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:22:23.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33e231cd16e143f8a6cd5428d7abd2fe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d03567788df140f2bcf809d707958c35 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2b2f0aa288140ada3fd3fd8bf7b372c : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5b2156c353834d4dadd8ff4dc464f8c5 : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68d04e5a6c18494b9c59a1ca71899c3a : SUCCESS in 5m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4cdce216de946d7bb0b5ee1a37c6ff5 : SUCCESS in 3m 50s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4b7041f94e0a498593a670bbcf25b34f : SUCCESS in 3m 48s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/647e368bf965441db21c2556c03e9008 : SUCCESS in 10m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/bc0c1636349e4f57a2f1d3ba4e2f864b : SUCCESS in 13m 21s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0484ea6902934670a521809f2d906807 : SUCCESS in 12m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/95486fa3cd914531ab64d0df2a89e295 : SUCCESS in 4m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/899f0926a5624878875c803116b62b71 : SUCCESS in 5m 08s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/9b0e44ec919c4ae19dd5e07724f1a4bc : SUCCESS in 7m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9ee1d43ed7e94864aef226f02e6efa1b : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e8b6676e73f7479d86ee606410f6e63d : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7b3297eca4ae4dd0b2fee9b388bd360b : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a46baf0ba8b84e10a69b804adbc41a7a : SUCCESS in 12m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/ae7bdb9ff96b48c996dff75a4f3fe86f : SUCCESS in 5m 01s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0c73311f9d6b476181a773df37aaca24 : SUCCESS in 11m 53s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/40d1113caa594575a83a3b0004488d97 : SUCCESS in 4m 51s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a17d4f8075fe45909136f30993dd1969 : SUCCESS in 24m 27s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1318d7585e744005ad6f1705c8bcfb3c : SUCCESS in 11m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/749d6e47ea6746f9be393273877070c1 : SUCCESS in 6m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/281584f8bd794982a76e2e6bb05674d0 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0692c90af0454e3e8908f45af555b3fb : SUCCESS in 19m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d744ab250b824f3fa4087339d848c4ef : SUCCESS in 6m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e0883e0480d24dcd88ffc754bbb33ca5 : SUCCESS in 5m 02s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a3c27262e4004691a00eede0c8a3f663 : SUCCESS in 7m 49s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f8e9413b53a84dd5b851851fe8a77b64 : SUCCESS in 12m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d8b497075ec94a3caa096b6520524d65 : SUCCESS in 45m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33cada8dc86f42068063d332dd0af1b1 : SUCCESS in 1h 31m 18s","accounts_in_message":[],"_revision_number":3},{"id":"772b1eb5c703ea2273271153b7ae48d2b2971e3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:22:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9d26ef7a012b7629b3ab1b73084adf9951252973","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:23:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/52fd9d07e76a438db861028313f56678\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6a065d73e864491a623167f3b36ba65 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1512e212d0e5dc24cacdf86e9671fb048d3af2f9","revisions":{"078903ba5cfc8606d6d55a84e0ca17f8967a754c":{"kind":"REWORK","_number":1,"created":"2022-07-04 02:44:56.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/848557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/848557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/848557/1"}}},"commit":{"parents":[{"commit":"70282106c9d00df1ec0b6bfb830bf1858a2ebad5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70282106c9d00df1ec0b6bfb830bf1858a2ebad5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 02:44:55.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 02:44:55.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2cd124c546b7af278284aef5e5ac745fa7fb563f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/078903ba5cfc8606d6d55a84e0ca17f8967a754c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/078903ba5cfc8606d6d55a84e0ca17f8967a754c"}]},"branch":"refs/heads/master"},"aa080d76e5fba57cee447128a2cd76f59d0f8dfb":{"kind":"REWORK","_number":2,"created":"2022-07-04 03:08:20.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/57/848557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/848557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/848557/2"}}},"commit":{"parents":[{"commit":"70282106c9d00df1ec0b6bfb830bf1858a2ebad5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70282106c9d00df1ec0b6bfb830bf1858a2ebad5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 02:44:55.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-07-04 03:07:39.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2cd124c546b7af278284aef5e5ac745fa7fb563f\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa080d76e5fba57cee447128a2cd76f59d0f8dfb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/aa080d76e5fba57cee447128a2cd76f59d0f8dfb"}]},"branch":"refs/heads/master"},"1512e212d0e5dc24cacdf86e9671fb048d3af2f9":{"kind":"REWORK","_number":3,"created":"2022-07-04 05:22:23.000000000","uploader":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"ref":"refs/changes/57/848557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/848557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/848557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/848557/3"}}},"commit":{"parents":[{"commit":"70282106c9d00df1ec0b6bfb830bf1858a2ebad5","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/70282106c9d00df1ec0b6bfb830bf1858a2ebad5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 02:44:55.000000000","tz":0},"committer":{"name":"Matthew Thode","email":"mthode@mthode.org","date":"2022-07-04 05:22:15.000000000","tz":-300},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I2cd124c546b7af278284aef5e5ac745fa7fb563f\nSigned-off-by: Matthew Thode \u003cmthode@mthode.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1512e212d0e5dc24cacdf86e9671fb048d3af2f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1512e212d0e5dc24cacdf86e9671fb048d3af2f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
