)]}'
{"id":"openstack%2Frequirements~848627","triplet_id":"openstack%2Frequirements~stable%2Fyoga~Ia7f81259e3c9db2d13afe77b2f7312f9a131ffd1","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"Ia7f81259e3c9db2d13afe77b2f7312f9a131ffd1","subject":"update constraint for python-manilaclient to new release 3.3.1","status":"MERGED","created":"2022-07-04 12:13:32.000000000","updated":"2022-07-04 22:23:14.000000000","submitted":"2022-07-04 22:23:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848627-new-release","meta_rev_id":"4095f1e8d758e95ed2c2ed3ff12b6052d9a0c6d3","_number":848627,"virtual_id_number":848627,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 22:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-04 20:35:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-04 13:56:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-04 20:35:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 12:39:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-04 13:56:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-04 14:16:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-04 17:59:07.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-04 20:35:06.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2b8bccf0d6e422350a20874c31deeda564118ae3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-04 12:13:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dd490776ad8ae012d23d50914f5397bd026da71","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 12:39:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/25831a20ff384e8fbf86863aec66800a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1d75394e0edd4661a6aa09cecbc076ab : FAILURE in 19m 14s","accounts_in_message":[],"_revision_number":1},{"id":"e3dedffa847549c12d4e8a1af8a55403278bc591","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-04 13:56:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c8e817c51cb85e4642aa6ef486a559ee2f002996","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 14:16:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5de1bf27534b419a85b3223cbfbc3cf7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/251819ad6ad9469381a4b8290be99972 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4958e6e328cd469999c5cf19b5414149 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/816daa2e84f342c7921b279d77a731d6 : SUCCESS in 4m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1d94e8ef4014b65a92179f04d22adc4 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8727b2f374834ac5aa50392431a4dd6a : SUCCESS in 3m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/c365514f764b44e49c9f50a57caab407 : SUCCESS in 8m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/121188e38e4541c8aa7b3e2233dc24ab : SUCCESS in 10m 06s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b15b2f8f58d34d17aeba869b73c88de4 : SUCCESS in 5m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8efafd280c1f42bd8b9e49b4f6fec73a : FAILURE in 10m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/876dd1ba359a4fdcb3616784e9bddfae : SUCCESS in 4m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bcb5d2ffc62444858e95ac2ad5f7de80 : SUCCESS in 5m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/65c355a43746414bb42ae025eb0bbcd2 : SUCCESS in 8m 36s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2d39f8ddf9654cafa9e81c7cb1c05f57 : SUCCESS in 7m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dc3d7ddee80648e2acd8d6e3beebb964 : SUCCESS in 8m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/65d9800685554a47a9e7ed574f9bf920 : SUCCESS in 13m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c1391f0e94424c4caa9a342305c6b106 : SUCCESS in 3m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4f627eec4b754a39903317c7980ceb39 : SUCCESS in 12m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eed046e3e5e14366ac1f476d2a9b1f77 : SUCCESS in 5m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3682dcdbacff41cca5bef8ad356296eb : SUCCESS in 19m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2b852796dfe5445a8969bbe0703cd22d : SUCCESS in 11m 36s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a392de011ff942a597eb7c67d97219a0 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/74b068e3808543ca9c752a5510925b45 : SUCCESS in 4m 05s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5fe79775a2c94d13992270cb76aecde2 : SUCCESS in 20m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/53f687c0ec2443bca5f3c69a1da36e2b : SUCCESS in 5m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6b91ac15504647af8e0fce5de5717b26 : SUCCESS in 4m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0422a9d4eac843b3908073da3a2ed8b9 : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/53b5e8045fa34f5390b3fa0611ee5e2b : SUCCESS in 12m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e5bc49e0060345c9812e30795b7b577e : SUCCESS in 46m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6d651dc962454f92b004e491c065ddc8 : SUCCESS in 1h 56m 04s","accounts_in_message":[],"_revision_number":1},{"id":"b94ab1d622225f8e81881890d8b028eb947ef0bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-04 17:59:07.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"9648f4c22d095e6ffde1aa37893996c8dc68884c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:12:38.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30fdbaa2ba48424cbd9dec5ae50d47f1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f670d198ff61490390426eee3ad58181 : FAILURE in 12m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6d28da49317eff2a8a1c065a0c815a07fd8f6d98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 19:47:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53b919e6aa084f949d47c977cd686b45\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ace1028026549dd8a5d0518d316046f : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a70d5707cc5b4a7caea31eb2c17fe26b : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cd90bb94d414754aaccf83730f73de2 : SUCCESS in 4m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/99623807695d4d1ea9b6684a468bc583 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6384f760cd5d49c494665e5ca1b532f3 : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/f896e4d8bf87483aa5f257c1b84305b6 : SUCCESS in 10m 21s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/65c2b4f10e68435ca46a445b5eacfdb3 : SUCCESS in 8m 51s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/4c987a04880b4884af985db1a9dbc8c2 : SUCCESS in 5m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9c2ad1c08a234134965513733f2679b7 : SUCCESS in 10m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0953c9118e0e4e929a47fef4ce2f5643 : SUCCESS in 4m 54s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/69de7f6cf1464e3da7f296c4bac0f921 : SUCCESS in 5m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9277218e85184355a59b4920ab12dc7b : SUCCESS in 11m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/66c4ff506eb04fe0ac04bc88a10ab289 : SUCCESS in 7m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f2fb713cec89446db325e93034fddddf : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/adb5ba8e40e54be0ad7a71952feedefb : SUCCESS in 11m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/be1060885ab54af5a879dab35012245e : SUCCESS in 3m 41s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a18e07e67d1244109435754ac6c29b2c : SUCCESS in 9m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e40eb0a34af54319b6bc774ac3172c48 : SUCCESS in 4m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6e1d36c54c8c43f6ad14e29875fb4688 : SUCCESS in 25m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/265892123b5e4959865990d2f0ffa7b1 : SUCCESS in 12m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f306e6a8eb2243e4b446da0187e6ad2a : SUCCESS in 6m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cd021645cd8c4a0896f9c23dafffcf17 : SUCCESS in 3m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5864112edb304e7388fa282b11dee2f8 : SUCCESS in 20m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/425f215fc1154e50b29c9b34d3819ec7 : SUCCESS in 5m 18s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/56e2a7e7b2d34d19b86deb30e480c7e7 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/067a384f8f03495d8c06f60c030e74dc : SUCCESS in 6m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4a342c4341834aa3a50d8be6d4ca59d4 : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/493841c035dd4a4bb90e66e83a737fe3 : SUCCESS in 53m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/608ca44bba4243568a91c68d69ae2254 : SUCCESS in 1h 41m 09s","accounts_in_message":[],"_revision_number":1},{"id":"04c81c3421b2faab4d459faf1b446d4e7000f9fa","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-04 20:35:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e0bc83a1833940d8062bb92d7803394f19c2b95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:35:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d2ceb5d08a4f5177691de9cead94cee21eba1476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 22:23:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d03bb908f79543bc90d4aae08e187856\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b035ebf975842dc86f81c511abd2c69 : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/31695ce61ac94d4cb66c1658045ee633 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f35e2a33358449b7abe0203b85fb282c : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/177707e8256b40159cfd112c3287f11d : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5113b72efd23408993934d8e87a12695 : SUCCESS in 3m 38s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/ba04fcd8d6f94944be06f1d5d53c7259 : SUCCESS in 8m 04s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e93db40278384fed84b7600738218ce8 : SUCCESS in 8m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b18822fa60e54be19fb7ef55cefa7ed1 : SUCCESS in 10m 24s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ab28d2773091498785ccce973aa8b383 : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ee43661644924965b2f42d035990127b : SUCCESS in 5m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/52276d625fe249dc81a1781289f9ed29 : SUCCESS in 7m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/f2fc386b828c42628476ac1cee88252d : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b83b77ce9774d9aa3f612713a707aea : SUCCESS in 8m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/77cf4601335240b6899c1e0a19ba56b2 : SUCCESS in 13m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/aee0b45def3d4c4bbe0bc21cb890dcf3 : SUCCESS in 4m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1794fa0af0284f4f8492131350af3d13 : SUCCESS in 13m 10s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cf8375c97ebb42b5931d59bfbea3b67c : SUCCESS in 5m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b348c87b1bee4008aaa64c3719c17d3c : SUCCESS in 25m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/fc4b249e92644d9699f4a35058b6b0ab : SUCCESS in 12m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/97069a32581a414588fd664ec80cd7c8 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6d5250d0bf542d2a696d6b3eee5ed9b : SUCCESS in 3m 16s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e62c4339b9584974ab3095eca56d6b71 : SUCCESS in 17m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/8da3cdd8efc845da9bc0fbdd247f053f : SUCCESS in 6m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/58859751b1b745f59ff5e4edca438e5d : SUCCESS in 4m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a1cc47ac24f148b793f98b59c0ce62ec : SUCCESS in 6m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7cd5f0f20600453596222f0470229644 : SUCCESS in 11m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5fee201eb1274fc6924344ce66bda5ec : SUCCESS in 55m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e9cb1ad0c46d44e4b0f977dd731fa595 : SUCCESS in 1h 42m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4095f1e8d758e95ed2c2ed3ff12b6052d9a0c6d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 22:23:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fafd055d31584651a69a5bcf42e5da96525df037","revisions":{"fafd055d31584651a69a5bcf42e5da96525df037":{"kind":"REWORK","_number":1,"created":"2022-07-04 12:13:32.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/27/848627/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/27/848627/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/848627/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/848627/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/27/848627/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/27/848627/1"}}},"commit":{"parents":[{"commit":"c6687011d34c468e224feede6730640cdce0a717","subject":"Revert \"Revert \"updating gnocchiclient\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c6687011d34c468e224feede6730640cdce0a717"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 12:13:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 12:13:31.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 3.3.1","message":"update constraint for python-manilaclient to new release 3.3.1\n\nmeta: version: 3.3.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Commit: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: Ib4319d2c8e138073d889bbede29843c7bc014223\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia7f81259e3c9db2d13afe77b2f7312f9a131ffd1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fafd055d31584651a69a5bcf42e5da96525df037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fafd055d31584651a69a5bcf42e5da96525df037"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
