)]}'
{"id":"openstack%2Frequirements~848629","triplet_id":"openstack%2Frequirements~stable%2Fwallaby~I72c20ae7a436a868b6ae1ff1c4ed40219ed13fe2","project":"openstack/requirements","branch":"stable/wallaby","topic":"new-release","hashtags":[],"change_id":"I72c20ae7a436a868b6ae1ff1c4ed40219ed13fe2","subject":"update constraint for python-manilaclient to new release 2.6.4","status":"MERGED","created":"2022-07-04 12:16:04.000000000","updated":"2022-07-04 19:35:32.000000000","submitted":"2022-07-04 19:35:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848629-new-release","meta_rev_id":"8275faa1658710bec25abd41fbe2ae52e95a0c9d","_number":848629,"virtual_id_number":848629,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-04 19:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-04 17:58:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-04 13:56:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-04 17:58:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-04 12:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-04 13:56:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-04 14:09:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-04 17:58:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d9923c18383b08137aea0d69e6be682b9a63e877","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-04 12:16:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"461410023d5036dab09be75310e494b303717376","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 12:39:00.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b37b7ccac2c642348d4e55e0a40a32d7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2b7b5920e1dc44978899128a70f7a41a : FAILURE in 19m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7d233dfa62f4ff65a7c4ebdbb0168bd64c6b91b4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-04 13:56:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cf7b5069eada976c9ba271341aefa4fb645d79f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 14:09:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9b2c3b26bc54b288213569e9c7f9f1f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a742f4c07ab34c45af532ea5c7ee1602 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f38b24b1db341d3be3d80132c5d311e : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd4eb9f14f934d228b4a03ea0d143e47 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0997281502304f30bd56bdfb31079f4c : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eebfccc5c08644af848b37e7d780301c : SUCCESS in 3m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b150f3edb6b94b6dbad6da230bcf55ea : SUCCESS in 4m 01s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/96bdd4d0fde8401792094a60995c37aa : SUCCESS in 8m 11s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ce78b2a1d95e424aa1e734c1d5042b53 : SUCCESS in 8m 13s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/abf36c6167634b959ffa00c4f84226d4 : SUCCESS in 8m 43s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/a559fd74e0ba4952999d1c81bb5fc477 : SUCCESS in 5m 22s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/14539167e039449d83c5793055b56b70 : SUCCESS in 8m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e0bb85df2bdd4e97a997f411500a64f4 : SUCCESS in 9m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2481110025f14290aff334fcaa7ca4f2 : SUCCESS in 8m 46s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/753737d7913a4792b380d6f46ad1f3d8 : SUCCESS in 25m 39s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/8a0679d72f634b93a1c8f4d1646892df : SUCCESS in 4m 41s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/077332ff8e3b40259c4817b1f1ab75da : SUCCESS in 22m 50s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/199b55d1c4ac46529efa59b12bf62940 : SUCCESS in 13m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/efffae2aa337456db193e80d0ac656ad : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/068a8c5c24bd4910b64d025dc3f44b9f : SUCCESS in 20m 11s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/7918f7718e824ecb8e662b2081e618fd : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5902e0f47e7f4d9d9704d60b6eb51f4b : SUCCESS in 8m 01s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/850fc6d1839c497b91c32c1f81fe16bc : SUCCESS in 13m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0699bb349ffb4a7ebb405cfd155cce71 : SUCCESS in 52m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67ad913b72c244ab94363d5d572f12a5 : SUCCESS in 1h 47m 05s","accounts_in_message":[],"_revision_number":1},{"id":"439f48d6062e3344710dc4ee3796e29e1164bd87","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-04 17:58:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"292dc4cfa7c517dcd774a11e956d2f500d8b5ba2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 17:59:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b16848996ab9f3c73dae6b9ad201820e8d7844bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 19:35:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d86917ff938471f8694ffff67abe23a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f4604a862674d66a4d152c5036ac7da : SUCCESS in 3m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c08269a34a44f0d82b428e368805c4c : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/956d9fcead744fd7962c8930e4fef069 : SUCCESS in 3m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aabf021c02f540b1a4ebb3ceba773508 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0e3c18460e734aa298422b7a40c45cdb : SUCCESS in 3m 42s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/271cdbd3f6d34381be647547595bee56 : SUCCESS in 9m 24s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/42cb0f8987db42f5b12e7dca9497cd7a : SUCCESS in 9m 23s\n- cross-cinder-py36 https://zuul.opendev.org/t/openstack/build/5ec9029cba974abb856f9cc1c311aba7 : SUCCESS in 10m 14s\n- cross-glance-py36 https://zuul.opendev.org/t/openstack/build/14b9baf16750499291328ffc1995e3c2 : SUCCESS in 5m 09s\n- cross-horizon-py36 https://zuul.opendev.org/t/openstack/build/3fc7be1e7fdd4306b4f881cd80297f48 : SUCCESS in 10m 57s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b87872e3063d40b2a879874978a7cbec : SUCCESS in 9m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4221ac80de0245238bac0ad37f96c18c : SUCCESS in 7m 47s\n- cross-keystone-py36 https://zuul.opendev.org/t/openstack/build/53468fcef77d41b793223c181a71d7d8 : SUCCESS in 12m 45s\n- cross-masakari-py36 https://zuul.opendev.org/t/openstack/build/c1763d5141ab4777a2a2cfa72ab9261b : SUCCESS in 4m 33s\n- cross-neutron-py36 https://zuul.opendev.org/t/openstack/build/e6cdc02a687a46a6b0409088e8a12d57 : SUCCESS in 28m 40s\n- cross-nova-py36 https://zuul.opendev.org/t/openstack/build/8e758a4b6e86437aa026d57b273d35bb : SUCCESS in 11m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5b95502e21284e8ba51d1834063a558d : SUCCESS in 3m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26816514b2c74e038e170049229955b0 : SUCCESS in 18m 45s\n- cross-octavia-py36 https://zuul.opendev.org/t/openstack/build/c82a46a431154183a1e50e00619c10d7 : SUCCESS in 5m 33s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/4e6b01abd0ae4716a12c36f8ed7d653a : SUCCESS in 7m 27s\n- cross-swift-py36 https://zuul.opendev.org/t/openstack/build/cd08bf8928c34c62b2a7a6ff7753c762 : SUCCESS in 15m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9013adb940d24601be67536a7c03e3ab : SUCCESS in 49m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73f02e0ed2384319b5fdca56fffc28db : SUCCESS in 1h 35m 03s","accounts_in_message":[],"_revision_number":1},{"id":"8275faa1658710bec25abd41fbe2ae52e95a0c9d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 19:35:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b2e63902a44f9ff0c5fb1d04f353d3596f9090d2","revisions":{"b2e63902a44f9ff0c5fb1d04f353d3596f9090d2":{"kind":"REWORK","_number":1,"created":"2022-07-04 12:16:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/29/848629/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/29/848629/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/848629/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/848629/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/29/848629/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/29/848629/1"}}},"commit":{"parents":[{"commit":"41c274af44c8d463da38baa0722c1d12e9a87764","subject":"update constraint for horizon to new release 19.3.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41c274af44c8d463da38baa0722c1d12e9a87764"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 12:16:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-04 12:16:03.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 2.6.4","message":"update constraint for python-manilaclient to new release 2.6.4\n\nmeta: version: 2.6.4\nmeta: diff-start: -\nmeta: series: wallaby\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Commit: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I230a7e2dae91f91a37117cc63c6e2840f7c536ed\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+1: Douglas Viroel \u003cviroel@gmail.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I72c20ae7a436a868b6ae1ff1c4ed40219ed13fe2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2e63902a44f9ff0c5fb1d04f353d3596f9090d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b2e63902a44f9ff0c5fb1d04f353d3596f9090d2"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
