)]}'
{"id":"openstack%2Frequirements~848682","triplet_id":"openstack%2Frequirements~master~Idb96b8fefd6424145ba8cc3a8ef4e6877377942e","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"Idb96b8fefd6424145ba8cc3a8ef4e6877377942e","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-07-05 02:48:12.000000000","updated":"2022-07-05 16:30:46.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f9126781f43e1798713d5d95b2eaa474f01bc30","_number":848682,"virtual_id_number":848682,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-05 04:16:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-05 03:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-05 04:16:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9c5c0455e362a087ca8afdcafbff7360f398f299","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-05 02:48:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5337a79cf2ff6a325eaeef731140b4caa0b5447f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 03:02:56.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e659e36b9d784ab8aa2f7e3e6b787723\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0be2161d8a8e4f368b1bffbe940d8a8d : FAILURE in 4m 38s","accounts_in_message":[],"_revision_number":1},{"id":"69cf4987247c0e38dc40651df0718a055cebadb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 04:16:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a06138bda5043a5b8878842e9a67a79\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e43a3cb856a485db9a8c71169a2ff93 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b61043710ea945249c10ad63240981a2 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f153eb235da40b2961ad65039370c84 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0482c889c92d4abfa33248f2f0edbe0e : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4c44e2c203d4dd8886897396a9c79d7 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7571ffcae3ed493f8f23e452db060f5c : FAILURE in 3m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/24d2c645c6b34537aeab39dbf43d7901 : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d42994a58a0b466b9d98aa1b81fd8e0a : FAILURE in 5m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/90e338bc994b4130a6c8f9cdac6f6113 : FAILURE in 6m 43s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/9f0eb36f7fe84694a8327aa47a9cbb2d : SUCCESS in 10m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/3a7639baa513431ba4cef1b843ba206e : FAILURE in 5m 19s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/abb5ee4ec30c4aebafc3cf22c0f5af6f : SUCCESS in 5m 38s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/72fed28a9a8c48a8bbe84a5bd60544ee : SUCCESS in 10m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a244ea6935eb4ade85c3b1489884b2d4 : SUCCESS in 4m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e63c4925ec67410f99d95ff6b3cd151b : SUCCESS in 5m 33s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d6c3a93ec62641d693b903c249a1d80b : SUCCESS in 9m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b024e50389d54e43980550ddaf866f62 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db0eb773ad134d03a21a7db40eb4fb63 : SUCCESS in 9m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9df924ca6c1d4d39b6271eed3a8dce22 : SUCCESS in 8m 12s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/437c3c17b5d9404fb4413a360598937a : SUCCESS in 12m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c7d90f8637724a198ad895e8e8db1ce5 : FAILURE in 3m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8a7c369c140641a58aa706ff0c658606 : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/10543dae2e0648bc932eab06a9d2de32 : SUCCESS in 6m 15s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7c80832664bf4a51b1fea9a794761d98 : SUCCESS in 20m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0afa7c7eb78c41d58613d793603208d2 : SUCCESS in 11m 57s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3003001c7b9543f3a2a00731d096c5dd : SUCCESS in 6m 11s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7e06d05e952245b4b7f3d89b6beb2411 : SUCCESS in 3m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f95127d1517b4159b09e5e658ae72604 : SUCCESS in 18m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c97391ad40c343e0aeab2b1963a69bc7 : SUCCESS in 6m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/aadf11c5f9744440ba6f6841b4d3ce35 : SUCCESS in 5m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8b3fe0caeca749cdb0a2862d597ae326 : FAILURE in 5m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/afd94522b5d74af693f1cf650d0bb06f : SUCCESS in 14m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/faef8fa5ac4a43b68ae65223783dc036 : SUCCESS in 45m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f03094813fdf4569b0b5919b13b204f7 : SUCCESS in 1h 19m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5f9126781f43e1798713d5d95b2eaa474f01bc30","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-05 16:30:46.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"08acdb40518627e42bf5370948c38ae280dcf8c7","revisions":{"08acdb40518627e42bf5370948c38ae280dcf8c7":{"kind":"REWORK","_number":1,"created":"2022-07-05 02:48:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/848682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/848682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/848682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/848682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/848682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/848682/1"}}},"commit":{"parents":[{"commit":"1512e212d0e5dc24cacdf86e9671fb048d3af2f9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1512e212d0e5dc24cacdf86e9671fb048d3af2f9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-05 02:48:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-05 02:48:10.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: Idb96b8fefd6424145ba8cc3a8ef4e6877377942e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08acdb40518627e42bf5370948c38ae280dcf8c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/08acdb40518627e42bf5370948c38ae280dcf8c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
