)]}'
{"id":"openstack%2Frequirements~848797","triplet_id":"openstack%2Frequirements~master~I9c52d5974408d1712bbd7988608bdfb4356e0d4d","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints","hashtags":[],"change_id":"I9c52d5974408d1712bbd7988608bdfb4356e0d4d","subject":"Updated from generate-constraints","status":"ABANDONED","created":"2022-07-06 02:36:45.000000000","updated":"2022-07-12 17:07:41.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fe38353b22d0b591e86ca80198684bebe09dbb38","_number":848797,"virtual_id_number":848797,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-07-06 03:12:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-06 02:54:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-06 03:12:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"601d1b952604693cfe50ca23bdc7abeb33feb5d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-06 02:36:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cceaf134eb2f714fddc82ae683a6d17dca82ba4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 02:54:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/308e353b18414aa59a37e1ca25011b5b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f85b762a0f664ed99d9bff3383a93812 : FAILURE in 10m 57s","accounts_in_message":[],"_revision_number":1},{"id":"325664d4bd2de9c42129499428bf4b746e05cd9a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 03:12:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45a1d81434d145c08975e0c0f66d241a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4899d12d5cdf4aeb8487e638153d4159 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d0d1475bf25498ca7f12575247c7831 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a491d42767e4cf39fe546e1f365f468 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cace2049a1b94dc9a523fc25ac154183 : SUCCESS in 4m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a352162fc8d140b185859f672a92edf6 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2614866d62bf4d72b66875b7dffca7a0 : FAILURE in 3m 37s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/51ec445944f8438593d272a50c2c693b : SUCCESS in 4m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bab8e6ea211046f5a54950f24673fdb5 : FAILURE in 4m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ef9cfe9dede64915a17eabbcf1a10d07 : FAILURE in 4m 23s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f40d7681fb714f9286ce227533d418f5 : SUCCESS in 9m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c0d4bbec8f9249d582f42026e9959915 : FAILURE in 5m 14s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/7e3832baed424acba307ea9245d774b6 : SUCCESS in 5m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/af892c2638404af18f1eb3dabedbb139 : SUCCESS in 10m 08s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3fa963b9c3ee40439f1fe4783022f144 : SUCCESS in 4m 49s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3d029de3d1104d1e8e4d1cda7e8db874 : SUCCESS in 5m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/135f10f0ae2d471abff91e351659150a : SUCCESS in 6m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/358894dc1093430ea5abce04da125cd4 : SUCCESS in 9m 15s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b2f50d0bd3724592a4c9ee3709f947d8 : SUCCESS in 7m 44s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/858d71a26fa64a81a27a0dc883665795 : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/53ead7f5df7641bc9550ce3096891942 : SUCCESS in 11m 10s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/44362868dd5149f9b78200a3f6fb1a1f : FAILURE in 3m 39s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0f4a48e600d74ad5876435148bd80208 : SUCCESS in 11m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/35dd3f1912f7436286de4f744d1a7bb0 : SUCCESS in 6m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3ee0519b810d481ba73eb947f63ed145 : SUCCESS in 23m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5f15919f20a2481191b12531a0910da9 : SUCCESS in 12m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4d14a13f2f1643e881f93e163f2eb7aa : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2926390cb6d84f689b990375cd8b212a : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4ce3651ce217448baea43735f75e83e2 : SUCCESS in 17m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2abb3d870e0a437286c39f99a691586c : SUCCESS in 5m 34s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0b8114036797429392f7ec0a49d6d70a : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be1b322054cd4bf5a1a9abba6405f09a : FAILURE in 4m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/441bfb30c1dd4e779bef28dd65e68200 : SUCCESS in 11m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/664b5c4c73a742f68257979a257885e2 : POST_FAILURE in 14m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c202e3609c4421181fb2ab733a369bd : FAILURE in 24m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fe38353b22d0b591e86ca80198684bebe09dbb38","tag":"autogenerated:gerrit:abandon","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 17:07:41.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0061be7e4ca1800fe407bd3522197b4eec4776f8","revisions":{"0061be7e4ca1800fe407bd3522197b4eec4776f8":{"kind":"REWORK","_number":1,"created":"2022-07-06 02:36:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/97/848797/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/97/848797/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/848797/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/848797/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/97/848797/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/97/848797/1"}}},"commit":{"parents":[{"commit":"1512e212d0e5dc24cacdf86e9671fb048d3af2f9","subject":"Updated from generate-constraints","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/1512e212d0e5dc24cacdf86e9671fb048d3af2f9"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-06 02:36:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-06 02:36:44.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I9c52d5974408d1712bbd7988608bdfb4356e0d4d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0061be7e4ca1800fe407bd3522197b4eec4776f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/0061be7e4ca1800fe407bd3522197b4eec4776f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
