)]}'
{"id":"openstack%2Frequirements~848940","triplet_id":"openstack%2Frequirements~stable%2Fyoga~If00d75bd379c6204590a62fa030ce551f163d826","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"If00d75bd379c6204590a62fa030ce551f163d826","subject":"update constraint for os-brick to new release 5.2.1","status":"MERGED","created":"2022-07-07 10:56:51.000000000","updated":"2022-07-13 22:26:45.000000000","submitted":"2022-07-13 22:26:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"848940-new-release","meta_rev_id":"544c8bc9b184b3aee5cf80cb4246cf6da6667035","_number":848940,"virtual_id_number":848940,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 22:26:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-13 19:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-07 11:05:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-13 19:15:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-07 11:05:54.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-07 11:13:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-07 12:28:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 11:56:23.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2022-07-13 19:15:51.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"0ac2a6daabd3fd1ff47c331c58e1103f5507359d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-07 10:56:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9280e78160cf6a44622e67555a18d1ea828a88db","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-07 11:05:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"15379bbbb4fe2c85fcf5c696909f717f800c37f4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 11:13:40.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9b264a08f34d2ba880b828477c18cc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/bbd103725e6943af85ef98685baaef74 : FAILURE in 14m 51s","accounts_in_message":[],"_revision_number":1},{"id":"9eb3e38fac3978c0516d436d5cb659a613a0aaf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-07 12:28:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/773858a6d084496b9f7f1286e2060039\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/524c3e6e36e44a61a0f16de1bd11cc6f : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/de9a182a2eba42d9a06931d44cc4ab90 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90149b295ac44036a6414237259b6b29 : SUCCESS in 4m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3aec8bba8de644579a67c0483a89e013 : SUCCESS in 3m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/98b7180832a94dc79dfb88e9dc133b4e : SUCCESS in 3m 48s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/a88f8218cb66491ca53a2055f3e03ab4 : FAILURE in 4m 16s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d965f61dec4948da8e84780f93352b6d : FAILURE in 4m 15s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/75e4b029670a4073bdc9eba146911ccc : SUCCESS in 5m 12s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e662a202f7184dbc9a825d5b8b5b3068 : FAILURE in 3m 51s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7fbb9839553a4f90a12534c9ecda593c : SUCCESS in 5m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4ebcb27b5c974450bbab7b9772951e9e : FAILURE in 3m 11s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c29a05d8a46646628b4863b18797604b : SUCCESS in 9m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df70bf10080f46fa8f5f567b900ebb30 : SUCCESS in 7m 13s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/73f858a691ee4c47896f3ba4650f4cfa : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5502523674654c40a39abc75bb95b250 : SUCCESS in 9m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/c2ffaca45c0b45c7b0e24fb00e3eeb8b : SUCCESS in 3m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e6fac1606ef94c5893ba66b0f6b4bf65 : SUCCESS in 9m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b41387231dbb4d49ae861732fb6d0daa : SUCCESS in 5m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8add83bb1a114549a15b9ee1410e6672 : SUCCESS in 23m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d448f6454b1b43b69772821a59068b10 : SUCCESS in 10m 27s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2c0b29849c41470d8e0068ab8092dd9a : SUCCESS in 5m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b762442fe3474dbca25d8f2951a22fe2 : SUCCESS in 2m 45s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f35b9825ef624f66914ace3c27f5990e : SUCCESS in 19m 04s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d22be80617384d7fb7260f567e486f14 : SUCCESS in 4m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7df54ff6e9e9461fb89f50d5446da925 : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c2eb36f925674859bc0864ac61e049df : SUCCESS in 6m 31s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a8e8faeea5974702a62217088e58546c : SUCCESS in 12m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c85c17716d049e7a107faffefac3072 : SUCCESS in 53m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/badd6a2e058f40fa955ed4d18fdbb3c1 : SUCCESS in 1h 26m 49s","accounts_in_message":[],"_revision_number":1},{"id":"1efc300e65429f0a1ece1386285b49aa199d66cf","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-07-12 11:56:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bafc384452ccbff2e575f7ee394229525fed59cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 12:10:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c67c5eecc144b4588af86aeedaf2cfc\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d9f699b97e464169a281dbcde96a7fec : FAILURE in 9m 59s","accounts_in_message":[],"_revision_number":1},{"id":"99904e3793f45a835e70f47aa4f4e898078a5f04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 13:42:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5917d31771db479cb56382f6edb0199c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0687783f5db4448ca71448d2356da415 : SUCCESS in 4m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33c09599386c4e23b9bc8849f839aaf3 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9faf24ac8265435baf74c85521f2f866 : SUCCESS in 5m 03s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f84231be9c904e709a663a68205bcc91 : SUCCESS in 5m 06s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/9d59a2666ffa4e6f9c1ea2ad641120a3 : SUCCESS in 4m 26s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/3922f7608fbb4b69b1651182ecc541a4 : SUCCESS in 9m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/eb46ea55e2344391a043af892d1e69b4 : SUCCESS in 9m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/90f273537b1746de8a51e8475c1f7786 : SUCCESS in 5m 54s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/fd7dbf6adb7a4871a35c769827cd243e : SUCCESS in 10m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/da0aba3106124a8181bafcc5c5aba933 : SUCCESS in 6m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/da82801aed0045ec95f78cf6fac66b2d : SUCCESS in 6m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/f18aedacedef4de6b17a8fbd5f24535c : SUCCESS in 9m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/008265d8e03c42dbbe6a35f02780f011 : SUCCESS in 8m 16s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/062cb11a1c2544bc85f3a865d32434cd : SUCCESS in 9m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0293173385184f9fa9d45b8ecbb528ef : SUCCESS in 17m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/19e4afcf36a54dd6b9c8dc7b74f42e94 : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8967528cad15413686594a9ecf2f7140 : SUCCESS in 11m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e4c0c61b12be44eaa5cb5d9bc05c8b3c : SUCCESS in 7m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/1a095aa7f748401bbcd81f9c4b043632 : SUCCESS in 24m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7ebedafd3a644c8cb538f3dd10e52bbe : SUCCESS in 15m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d014349d12c149848ae7d3dc3f513903 : SUCCESS in 5m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0dfcd8ff756a4975b71a9f15e9caa086 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2394f53636bc4affa1a6d5fbf4daa3cb : SUCCESS in 25m 25s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/819feef88d824e32895c6126fb53e49b : SUCCESS in 6m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a99ca6e19b3a44629b1fadbbb657db21 : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0049f51e7b5e4478ab67a33a9ae2119e : SUCCESS in 7m 44s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e0997bd6521f4d4b870669466c4d291d : TIMED_OUT in 40m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13b539c52622496ebebe62e29b51ad8f : SUCCESS in 53m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1d3ed5d145e4410ea3fdbb08bbfb2854 : SUCCESS in 1h 44m 40s","accounts_in_message":[],"_revision_number":1},{"id":"c0b8b94565d7d6517a7dfbaf20534a00d2162c4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-13 19:15:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d7c53beceef7ca8d5bd11f9558437779c600beee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:29:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c48a33a13c448ca991e57c65131970f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/3dae436f67354e38b0d9fd202b012136 : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":1},{"id":"63479e9b63c268cc9870ef8cbfb985d222d7afa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:47:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0153efb17194e57a4f20a8e48d31926\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/212586e82f164a7d81fa6de4fb852c5f : SUCCESS in 5m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d89df06c03f345d6bfef1514f2ede52c : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/880b2780821847ceb0d55bd2eb92bea8 : SUCCESS in 5m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d161b9d7d89743d1b173b15525deff2c : SUCCESS in 4m 49s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a355a3a2990d462b969f145535180787 : SUCCESS in 4m 55s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/d23b8b979eda4857bea8d955ea9bce9e : SUCCESS in 8m 53s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6ac348d16c494065b1a3c3e06fcc682c : SUCCESS in 10m 02s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6546ca8d7edd492394621aaabc74102a : SUCCESS in 6m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/99007a0ccd784562a610a44e64ae9cd3 : SUCCESS in 12m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7364926e49af46059a23953be92eb420 : SUCCESS in 6m 18s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e4408113e1e54b0abbef5b76ecafb871 : SUCCESS in 7m 16s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/291454424d77428e80d15978af8a2fdf : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/79c8ceaaa2934f71947f490d2078caf4 : SUCCESS in 11m 03s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7f557061ee234d2e8d2048dc16b4851b : SUCCESS in 9m 50s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b38962741aa049039ec2083fd0f64ee2 : SUCCESS in 21m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/96979f679b6d449e82995115bd464fee : SUCCESS in 4m 46s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b801371c764b430dbbef0329f683fd52 : SUCCESS in 11m 24s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a8604cffe54d4acd9256e4c6f1dccd9d : SUCCESS in 6m 26s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a8e775d62b4140268746a85e71f8d5e7 : SUCCESS in 20m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/76f9caad3f844e7da9ab6670fc6e5c8a : SUCCESS in 13m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/5924bde467ca4e3fb91091112182380c : SUCCESS in 6m 35s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/72b2908dfaab4b96a6038ac9717031a7 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/98c42c13b6d9405daeac10f3a80a1c44 : SUCCESS in 19m 22s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/634d5ff5cf8f4c69b57571401c4821de : SUCCESS in 7m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/2782d398bad64668bfef6c5bef378e0b : SUCCESS in 6m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/887ca72673e842dba6d5c12e1abe46b2 : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/69730e726294466d96b1d3f9a71fa9ea : SUCCESS in 13m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f11e198872d7478fa9a8522ebc87205d : SUCCESS in 51m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30cfc85d022f4f6586fd2286ed830922 : SUCCESS in 1h 29m 24s","accounts_in_message":[],"_revision_number":1},{"id":"62252899b38d27d802deaaf40c134dec8da0df7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:48:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4f3b13dfd04386641cf7b65fecce58bfab69e83a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 22:26:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/268f9533a18b4dfcaf7e2493f1e1644f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2c868a856bd4ad78f8b7fa56b7c1f50 : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1876a388610f48e2a22fa71e138cab61 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4d83e6e7f62497ca8efc4ab073bea62 : SUCCESS in 6m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ba027fd57dd24122b87b2145b8b92308 : SUCCESS in 6m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/644bd7ff01bd4eada710a50604c9c968 : SUCCESS in 6m 21s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/8d6fd395d5814cf6a37e4b15e3a7c852 : SUCCESS in 9m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7ec02830d1aa4bf6a0b8e06b6c954936 : SUCCESS in 8m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/dbafabf0428d434f898616602486eb02 : SUCCESS in 12m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0b0d2eb926f242c1939fcf1ee9b60705 : SUCCESS in 5m 48s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bb1d60581ee64dd39e7f7265e4319964 : SUCCESS in 5m 57s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/52ff4a7e99fd46d58956ccdce4123e32 : SUCCESS in 10m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/182aeb2db9d7422392c4f6b869466c1b : SUCCESS in 11m 45s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/1da4d70bb2d64997b1964dc1d7b4f19d : SUCCESS in 9m 10s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ab16baaef9354fa390fbf11800b6fae2 : SUCCESS in 9m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/66fbfc7df85549a6bc12080fe7e4fecd : SUCCESS in 6m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b96e079153b94305836d107a59358d81 : SUCCESS in 11m 52s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a7c81264285f47148be7a4217557017a : SUCCESS in 5m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c5d6b0514d714b0191783099bbc5755f : SUCCESS in 22m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f28a2777bcd64c6b8b171942cd25f357 : SUCCESS in 13m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ae97fac9d0674288b7bca665f6e9f68a : SUCCESS in 5m 34s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/039af55bc68746baa0b99cd2ee86d7ae : SUCCESS in 6m 50s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7d8996feacea4084b3307fcec4c22aac : SUCCESS in 19m 20s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7785f833bfd84b8dabdf3ea2406ba2fc : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3d7906ce6cc94ef9b2725a9f9da5e1bf : SUCCESS in 6m 09s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/9b189e5f3f49447ab0948b6ba0057988 : SUCCESS in 9m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/15b8f7babc00431aadeb57d0b7064f55 : SUCCESS in 11m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7024be59ce94cc487daf6937015b5d6 : SUCCESS in 54m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2eab80ee1a274ef0a213744252ff5652 : SUCCESS in 1h 37m 09s","accounts_in_message":[],"_revision_number":1},{"id":"544c8bc9b184b3aee5cf80cb4246cf6da6667035","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 22:26:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c70f9032bb885e861656605375324b8d016f23d0","revisions":{"c70f9032bb885e861656605375324b8d016f23d0":{"kind":"REWORK","_number":1,"created":"2022-07-07 10:56:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/40/848940/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/40/848940/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/848940/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/848940/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/40/848940/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/40/848940/1"}}},"commit":{"parents":[{"commit":"fafd055d31584651a69a5bcf42e5da96525df037","subject":"update constraint for python-manilaclient to new release 3.3.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fafd055d31584651a69a5bcf42e5da96525df037"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-07 10:56:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-07 10:56:50.000000000","tz":0},"subject":"update constraint for os-brick to new release 5.2.1","message":"update constraint for os-brick to new release 5.2.1\n\nmeta: version: 5.2.1\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Commit: Jon Bernard \u003cjobernar@redhat.com\u003e\nmeta: release:Change-Id: I39dce8bc94c66051ecef94b0177d367c85aab05c\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If00d75bd379c6204590a62fa030ce551f163d826\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c70f9032bb885e861656605375324b8d016f23d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c70f9032bb885e861656605375324b8d016f23d0"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
