)]}'
{"id":"openstack%2Frequirements~849284","triplet_id":"openstack%2Frequirements~master~I43559379bc71c990361af24b5e2543b426ce84da","project":"openstack/requirements","branch":"master","topic":"uncap-warlock","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-07-11 16:53:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43559379bc71c990361af24b5e2543b426ce84da","subject":"Uncap warlock","status":"MERGED","created":"2022-07-11 08:55:09.000000000","updated":"2022-07-11 16:54:32.000000000","submitted":"2022-07-11 16:53:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849284-uncap-warlock","meta_rev_id":"782f4d0a86a7a3f3853ea79d7dc4d4c02719dffa","_number":849284,"virtual_id_number":849284,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-11 16:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-07-11 11:04:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-07-11 15:29:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-07-11 15:29:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-11 10:20:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-11 11:04:32.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-07-11 15:29:20.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"35f7607eaaa28e3a8e5bf147aedd771bf34d41a5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-11 08:55:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8be6d3b92c24a35801fd8554ea61d2a8e982ef61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 10:20:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f406fc5df64147d88e112bebc3c925a1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44292a894fbd4c0f8a1b41c2b35009f0 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa9c2d0a8b3d4bbba8a3b855ad57ed92 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4fed9e03bd746edb18566bb778522c3 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c28e3711601f4bffb841662c53b440cf : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/376d53442cc9446988d531f21b49c403 : SUCCESS in 5m 13s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f5020d89e5a941fcb8aa655c7fd318b9 : SUCCESS in 4m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a87c0c9ff66d49999b7a886e202aeed4 : SUCCESS in 51m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/726d3efe77244f29b657e27c0241e845 : SUCCESS in 1h 23m 35s","accounts_in_message":[],"_revision_number":1},{"id":"407be8d3145a05a383301ee13743ea0e59bf2dcd","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-11 11:04:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"502f82bb62e3469468b57ee676b7cc19066fbb8e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-11 15:29:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b5c42d9d17fd9b8297217005b8367a277d0a30fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 15:30:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cbb306fc5711d3e5a60ff4d7865609cb5b86e9c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 16:53:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7634b62d298144239c32911de8d11025\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/288dc03665bc4b7ab199b025fe8ab780 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44d41f2faddb4f858baca0e9ad1735ca : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd084173c4a84395aa742980e7714419 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a7a8351cfdfa4f4985478d41e167291c : SUCCESS in 4m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/315f94ef5a6046669dbbd0b11ba598b4 : SUCCESS in 4m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/132000c7439e4d09ad13afad451c49cd : SUCCESS in 47m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73ebc9a9da7240de9b693ae12bc1028f : SUCCESS in 1h 22m 34s","accounts_in_message":[],"_revision_number":1},{"id":"ecde5cc208e21739b44260e2f983c374b6836273","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 16:53:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"782f4d0a86a7a3f3853ea79d7dc4d4c02719dffa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-11 16:54:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17dd08b303db4ccaa5651ae74effd84e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a719f139d9406292cf4370fc58c8b5 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013","revisions":{"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013":{"kind":"REWORK","_number":1,"created":"2022-07-11 08:55:09.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/84/849284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/849284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/849284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/849284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/849284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/849284/1"}}},"commit":{"parents":[{"commit":"e6a484bf97fa30b8bf69fb397970e382344f7b6f","subject":"bump dogpile.cache to 1.1.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6a484bf97fa30b8bf69fb397970e382344f7b6f"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-11 08:54:46.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-11 08:54:46.000000000","tz":60},"subject":"Uncap warlock","message":"Uncap warlock\n\nwarlock recently had a 2.0.0 release that added support for jsonschema\n4.x [1]. We no longer need to cap this.\n\n[1] https://github.com/bcwaldon/warlock/releases/tag/2.0.0\n\nChange-Id: I43559379bc71c990361af24b5e2543b426ce84da\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
