)]}'
{"id":"openstack%2Frequirements~849447","triplet_id":"openstack%2Frequirements~master~I2b7262f548ec62c34636f47fe6e94c44510a9cc0","project":"openstack/requirements","branch":"master","topic":"exclude-textfsm","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-07-12 15:59:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2b7262f548ec62c34636f47fe6e94c44510a9cc0","subject":"Block latest release of textfsm","status":"MERGED","created":"2022-07-12 04:57:06.000000000","updated":"2022-07-12 16:01:50.000000000","submitted":"2022-07-12 15:59:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849447-exclude-textfsm","meta_rev_id":"e876de450ac500506db2b065dae0e79f67429d70","_number":849447,"virtual_id_number":849447,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 15:59:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-12 14:08:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-12 14:08:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 06:14:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:08:00.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"939ba81280d74e5d9c4cacd0660fbbfb9b81386a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-07-12 04:57:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62846bde722c7dadf452da0affde15ff64672611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 06:14:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51c09a3136ea422b9e5e405fee0bb9f5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db486a02acc440b8b27f447d7363743d : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d97838a09c4531bd457f3ea99f1c86 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5408b04572f4c5eb3d5786a83609079 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc65392ea48b4a9da3c0dc46eb1a6d5b : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36afb3ec87d64791a868721e7f7fcd15 : SUCCESS in 4m 21s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14b8ae722ff0455a914e7440aa96e12f : SUCCESS in 4m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fb571b1f741c43a0af8531a717fcd515 : SUCCESS in 44m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ea4d9b8366e4cbbb4a41ffc8c96e190 : SUCCESS in 1h 16m 09s","accounts_in_message":[],"_revision_number":1},{"id":"12e5a8a5095fed9a03261dd95f5704745d8296e0","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 14:08:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b27aa563380b6327c5551a0dfbaf9614df4a4ff2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:09:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"478788734bb0a92a26717a6a0c9aa8ffe8412ab2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:59:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c79080cb1f04ebe810e19af788873b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fcf81f0e68945d690ab7f9a0cd63328 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf03a1745a8b469b8eb14470aa3cf541 : SUCCESS in 3m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/241e6ac530804344a1b56216446c0067 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97fad9d0121c486e821c7915e7154c9c : SUCCESS in 3m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/82dc3c0c44c740de8345a00758333b79 : SUCCESS in 5m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e649613a6c754dcea3a8097e7b458e85 : SUCCESS in 54m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/934359dbec6d4a5e9a042b5b7f7710ac : SUCCESS in 1h 32m 45s","accounts_in_message":[],"_revision_number":1},{"id":"287f663f111d5b4a5776c07bc51279f6d80a7a3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:59:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e876de450ac500506db2b065dae0e79f67429d70","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:01:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebce95aa6a9f48ae980351e07916c21b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6437b7846d55496b9af8543dea0fb657 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"388095bd38c907aed97ffabf45c6d52ceea7dcc5","revisions":{"388095bd38c907aed97ffabf45c6d52ceea7dcc5":{"kind":"REWORK","_number":1,"created":"2022-07-12 04:57:06.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/47/849447/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/47/849447/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/849447/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/849447/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/47/849447/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/47/849447/1"}}},"commit":{"parents":[{"commit":"e6a484bf97fa30b8bf69fb397970e382344f7b6f","subject":"bump dogpile.cache to 1.1.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/e6a484bf97fa30b8bf69fb397970e382344f7b6f"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-07-11 07:47:26.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-07-11 07:47:26.000000000","tz":120},"subject":"Block latest release of textfsm","message":"Block latest release of textfsm\n\nThe latest release of textfsm is broken and not installable[0].\n\n[0] https://github.com/google/textfsm/issues/105\n\nChange-Id: I2b7262f548ec62c34636f47fe6e94c44510a9cc0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/388095bd38c907aed97ffabf45c6d52ceea7dcc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/388095bd38c907aed97ffabf45c6d52ceea7dcc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
