)]}'
{"id":"openstack%2Frequirements~849477","triplet_id":"openstack%2Frequirements~master~I86033450b120b87f1709b22050c16b79fb378a75","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I86033450b120b87f1709b22050c16b79fb378a75","subject":"update constraint for oslo.config to new release 9.0.0","status":"MERGED","created":"2022-07-12 09:26:00.000000000","updated":"2022-07-12 23:17:16.000000000","submitted":"2022-07-12 23:16:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849477-new-release","meta_rev_id":"9b55a02c74a1d604457c6148d111986706d09dfe","_number":849477,"virtual_id_number":849477,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 23:16:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-12 16:55:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-12 19:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-12 16:55:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 09:42:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 10:45:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:07:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-12 16:55:37.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-07-12 19:58:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"7a62150c49894f8375731869d60570a9026c5c4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-12 09:26:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a1afa701d95d5822d3980b66d8fd7dde5628f4e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:42:57.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9db1eaf88d404ccaac1d5be517e83dda\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/eca5d84b71e6445f87cbde7eafbd6bef : FAILURE in 11m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2dc2ae8015e784f8aaff88c138be1705d18754a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 10:45:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f23f204e55b9472b816984f67079f25a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa0c065215464835a253f94c1e1244a3 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7a85b9ee2874329b2d3927aa89c0c79 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/892e102796a448a69f64f2e6bdc8b843 : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e62e41116e614357b0ec8f14970e5639 : SUCCESS in 3m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47c3d74fad5a4f74a387f778e1f4cce1 : SUCCESS in 4m 09s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/52e78294d0084ad4b488eef10f65cc28 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a2cc39e426174b7f8016bf9d15d1e483 : SUCCESS in 3m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/de0eba328194429fa667f6bbf428a069 : SUCCESS in 10m 19s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e5bc4908e17041c2aa87ff6d4f640655 : SUCCESS in 14m 37s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/3ba4e1f66d7844ec96f7463ca86d5b58 : SUCCESS in 7m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cf7cc4e20cc64250a6d0f33c3759ab7d : SUCCESS in 6m 03s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/26026d35713743bf962c737b480d3fda : SUCCESS in 6m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/940c1070aff74db6b35edb7b99bee924 : SUCCESS in 11m 13s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/be1d62c21ec9451ebef91b221ecf1b2d : SUCCESS in 4m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/277d79ebc0464827ad3539bea7758515 : SUCCESS in 5m 44s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d802a49e2ff14554a805983c94f54ee7 : SUCCESS in 8m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0cba7c0169164af09c5ec5729cd4b02c : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/df0afcf598594280a885c20e2b648b0a : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7bcb48ab869f41fe87e9e14fdf76d61d : SUCCESS in 8m 04s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/9299856516834717b7f2525d9087b4fb : SUCCESS in 12m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1d2ce45d76de4a8fbff5136b688723d9 : SUCCESS in 3m 26s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8ef87d1f39db4977991d26d654b657b5 : SUCCESS in 10m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cb7b7b0470544d6581030589b81b5ccc : SUCCESS in 5m 41s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/059928af3af64817a746b423df97e20f : SUCCESS in 21m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/120ca10d11944adaa23c8c19b7fc693b : SUCCESS in 11m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e443fe9bc4a74e69a7cb79bfaf438470 : SUCCESS in 5m 59s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/e4caad04009e4050be22438a751b6246 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/24a950db58374927ad955361c6a28a6d : SUCCESS in 20m 03s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1c226e45b8474995b46e915ef4633a48 : SUCCESS in 6m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5a7cd8b48a124ad0a93d40396b437fe5 : FAILURE in 5m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bbd82e4b14644bdc8c7c9267f00eb721 : SUCCESS in 8m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e9981af2d4fa423482a7ecf1a5cebed0 : SUCCESS in 10m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0eef5b9d6aef4fd9999a0b111f327d3d : SUCCESS in 55m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96ce5c9f3e22439a9304cf3aae8ce008 : SUCCESS in 1h 14m 09s","accounts_in_message":[],"_revision_number":1},{"id":"88a635a087ebd0b609f0eefefc614f4bb8f4dd4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 14:07:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"12e3000cb256f2be27a73bee7f27399e388a0f86","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:21:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec185dc1b0164e2d8c06cd8b34d854b1\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/32e01e5d7a434ba3b9d91bb56085bbf6 : FAILURE in 11m 27s","accounts_in_message":[],"_revision_number":1},{"id":"e1f4af23cf68b6c7c29d4c0f9ade2675d8a7cb6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:17:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6a08af182e0425d9abbcc54200f65ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a967620d31b840f88eb1900832dd38e4 : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/912796001f9e4103be604509cc8fd8d4 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4528f1e2aa9b488fb1341de0873b32ba : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8512164b93c3447198fb676bb4f262b6 : SUCCESS in 6m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de6d015a3ff9406aa0fe93409f1b65de : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1125f9a8edb49bd82adad3eb3d4f874 : SUCCESS in 5m 05s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/f391a3f0eeda4aba8c63c21722d105c6 : SUCCESS in 4m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be929655dbb348449afa30043152823e : SUCCESS in 9m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d8d9c0bb10db4d8ea438536a4c65f281 : SUCCESS in 14m 22s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/426e88756158443ba4baed7df9dea2de : SUCCESS in 9m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/a8df56c2d5aa4d31be7b0d93805f0fd7 : SUCCESS in 10m 22s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/005a125bac8a49bab3a26457752815c8 : TIMED_OUT in 40m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/449b63e54b434b73a02f43c06d5d300a : SUCCESS in 10m 57s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1880034c211940d38110e8e080e9a014 : SUCCESS in 6m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8fe750d46d6d4d95a80eb81faf0d2a4d : SUCCESS in 6m 25s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3b28e7ffd8c7499c8406bdbbe08d7181 : SUCCESS in 9m 31s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3ccd0bc40b0a40e2b3548665bf6a07f0 : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/b3bdb8605e4d404ead87a9e28409a091 : SUCCESS in 9m 18s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/76beea64ca584a3a9d8a2cc410569225 : SUCCESS in 9m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7870f9264b234dc2ac15869719aebf10 : SUCCESS in 14m 31s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/da889531782449e6a17e059e71c6e2bc : SUCCESS in 4m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/6bb8bf6c704045ed8acee55ff7633c76 : SUCCESS in 9m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d8a7121639e94f9aa60a069915b99f02 : SUCCESS in 5m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/51e24fb8f11a415d950bafa4d5f26a46 : SUCCESS in 21m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7b2505fdc8ad4297876b29e12057185a : SUCCESS in 11m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/feabdb076680409e994fa9a13ade5d96 : SUCCESS in 6m 41s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7c3c5214f3c249c793ebefdd52df9f79 : SUCCESS in 4m 56s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef06cc72d4fe431394d20997cdd1ee96 : SUCCESS in 18m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/816b837683b546ad932027475fbed547 : SUCCESS in 6m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/384b14d8f88f432a9ba6bf612eda6523 : SUCCESS in 5m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/926d3dd2df3144fdb38c22f1c051e6fa : SUCCESS in 6m 24s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/7a69a928ee764e2a82a4eb546fe9172b : SUCCESS in 10m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c30fc4b168241c0b58292814ef0409f : SUCCESS in 44m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c62582f62ae44e5e98902052f126e7be : SUCCESS in 1h 25m 14s","accounts_in_message":[],"_revision_number":1},{"id":"f1f5c77e5f3d8d85cc9d72588239e49bba001d37","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 16:55:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b1af4e66f21ccd407dc0450b4dbde5a87246e01a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 17:09:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d44c7a5ab3f14049bdf1b2257b51a5c5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a4ddbe56882c4987af710c4f0f3d0a86 : FAILURE in 7m 03s","accounts_in_message":[],"_revision_number":1},{"id":"f0009953ea9aef52851c12e6a822a49092c6b07c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 18:36:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b0c958139fd4d7cb23c9fcfe5596263\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ccb04ca16b84a3ca0eebc33ec3d2999 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/534868e8a64b43c5a537b9654c1037a2 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf15ecb02c74cc5ab7c808bb3387887 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bccc29673b184cb4a05b410044a46bb5 : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2069bca307ef46b894f41ceee6fa47ad : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fdf2741befa242929a98a4553edba093 : SUCCESS in 7m 18s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4f3f41d742bd49e6b48b0fc06ad994a8 : SUCCESS in 4m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f29eaa5814d84693a146e9dcf32757b7 : SUCCESS in 13m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4e46fb644bf74a34b605d86fdee5cb2e : SUCCESS in 14m 45s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/026773bed3ae41b99610773a53f19e1d : SUCCESS in 10m 21s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b66bc51cd1814835b00a407a5deaafa2 : SUCCESS in 6m 03s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/c89fc1ef88944103910d797f8e27b72f : SUCCESS in 6m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/771674d4969c4d20be03e04ec972593d : SUCCESS in 9m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c91951149be249b1b0c9a8e3507fad6b : FAILURE in 8m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/08e081b44d2b42e6b092bcf046f860e3 : SUCCESS in 6m 10s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/13ef608f5b384480a6e233053da7a815 : SUCCESS in 8m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e39a94a9d7d94a1389e29343119ce602 : SUCCESS in 10m 04s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/afccd9221a0d45b5b9b63b29a4a95eab : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dd2b281971f147519cc0e84ffe784034 : SUCCESS in 7m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/827210066f8646d598c9954bf4863c9f : SUCCESS in 11m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/21a463f1e94146199af8941734c0710f : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/af93f0e7501d43e1a687bba6735a0aaf : SUCCESS in 8m 21s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5ced46a374384b1c9a4a5abfa57a61e7 : SUCCESS in 5m 49s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/22a943e580994581b18e3a0dd29a9ca8 : SUCCESS in 23m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1793c3a16fc4400aa1f7b6462ee96d1e : SUCCESS in 11m 14s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/033c77a324bd4b03bc1ec8524a551cc1 : SUCCESS in 4m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c4615191687470eb6c8166f1535c558 : SUCCESS in 3m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/370bd66ded8644d4bb3464ec058c78b6 : SUCCESS in 20m 36s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fc0a4ce2db60488eabade980f9beeb52 : SUCCESS in 5m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6fe0dd031318495b8f26d758a1969b87 : SUCCESS in 4m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c06c6cede4ea4438bab91fdbf199e05b : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2662a098ccd240a2b9c8f92a681de61d : SUCCESS in 10m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/62a59d032ec14ab1ba56571788297b49 : SUCCESS in 48m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/75c91b54bdd34f5e921f82bccc1758ff : SUCCESS in 1h 34m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f2dfbede172765289c161530ec4ee84ea349fdd8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-12 19:58:42.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f01575cb133a6acdee2d82344dd632787c41b5c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 20:09:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a478fe18ebd44d90b014806191c001ae\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e828365828ae4fbc975d03b522fb1b1e : FAILURE in 6m 46s","accounts_in_message":[],"_revision_number":1},{"id":"03ae8ca808abac9a61f80a98efb93007ad6b3482","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 21:47:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44c94b16e5e743dd8e9bd95a45bc3bd1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc9d3e18c1b84084bce450bcfa95b220 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e5187e6512a4ef685da7acd267ab781 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48f5335acdce4033ae4185ae0e437ede : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/80a02af2f5c143799135cc8bbc479c71 : SUCCESS in 7m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5631605b12804938aa68c1fef143578f : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/552fb4ce0d5849e587156aadbe58e194 : SUCCESS in 3m 53s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/77e5bd1d7f4a494fbf62fcdcb9bd5d30 : SUCCESS in 4m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f5d35950848c42f48cdf93d5747df2c6 : SUCCESS in 11m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9cda5f481f5b46518a8ebc894a413677 : SUCCESS in 12m 46s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/1c2ad68267b54124bc43298b88b31934 : SUCCESS in 10m 13s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/6826c9490a544b73a43047f0e269e184 : SUCCESS in 9m 24s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/643ecfd43f2b40e4af5aa48fe79cc646 : SUCCESS in 8m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/155b8760a0604de5b3d56ef1da0e49ae : SUCCESS in 9m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9ef042d5b3234573bc4186aa8e19a3dd : SUCCESS in 7m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b1eed4a14bda4420981f346bc63db5a8 : SUCCESS in 5m 04s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/428b04f43d1d4abc85f10799c4f2bfa7 : SUCCESS in 9m 09s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/5e6eb5054e5143e596493d21f8618196 : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/05fd4ebfb630417b9261d903b906776b : SUCCESS in 7m 53s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d9d9c6c32f22454dbf921300c8dd4754 : SUCCESS in 10m 07s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/979673a3755f4f29a4e4f2282cc25216 : SUCCESS in 13m 11s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/93e9d649adf941c2b384a70200d34c3a : SUCCESS in 3m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/35cf1630860a4f77a8727c235d109c03 : SUCCESS in 11m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2ea500a2e8f04689a7a393b2bc5772ce : SUCCESS in 5m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eb2f465c1e954b34a3ad28255b87abf2 : SUCCESS in 26m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0253d561e6654afd946d8404503b2055 : SUCCESS in 12m 52s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/793bfaa8774449e49e3da132a9d10050 : SUCCESS in 5m 16s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ba304c33e38b4f1ab05e611d29afafef : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2ba05bf199fa415e83513ae6e1d98db5 : SUCCESS in 21m 21s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/dc24f4ec3f524841a2981f371775deda : SUCCESS in 7m 53s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/bb7bf77322d847f4a499c6831bde1dbd : SUCCESS in 4m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fee38b2b094b4f26a8d85803a75d24ce : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fcb0f2be542d42008062703f39057335 : SUCCESS in 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1ed073182834fe2bee85fdde22e65d8 : SUCCESS in 39m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/212888c5a7fd40b5ac11dedd867fbb20 : SUCCESS in 1h 46m 38s","accounts_in_message":[],"_revision_number":1},{"id":"67aa38fe9876d3b3cd934d8b1a93688a31cf71b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 21:48:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6ce73603dad9a800c51277353a48e8eb78e6b001","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 23:16:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a832bbdb01494862b7e306319362feaa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1a04f6b4d4e4b0cb85976aac02375d6 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c11bd4b4d584a5394d18bb346c1630b : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d97ec234bfb4846a3e9a2384532a443 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb89326a86a44f80a7aa31ec745b9af7 : SUCCESS in 5m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d387b954be54475d9205dc36a28a60f1 : SUCCESS in 4m 22s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c09b1cc951574c7ab9ee2e17c1ca2a5c : SUCCESS in 4m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d48e73be9e5c4d9a8f50e9db14c1635b : SUCCESS in 8m 47s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3c1732dcbe0840a1b5be8390fbf70ddc : SUCCESS in 15m 18s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e9e3442ee2634a89a330568cd96600a1 : SUCCESS in 11m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/84b5842d7e8a4370b42a39c9efa39c2e : SUCCESS in 5m 36s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f3f7f98971df4a4f8a887d87531add28 : SUCCESS in 6m 25s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/568bea93b45648369e5a8a6d9c763b8b : SUCCESS in 8m 26s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dcee3d632d8b42459ebfb89f4e40db4f : SUCCESS in 10m 22s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/150dd6f3b9bd41a0af0e584adaba387a : SUCCESS in 9m 24s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/dbe9e9fd563948ea957da091ae76a5d6 : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f791314a7da447f78f1fd154ad5dbc15 : SUCCESS in 16m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cea937c446c44f9892339bc5098b4485 : SUCCESS in 4m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/90117c0a6eeb43bbb14f05edfd290665 : SUCCESS in 11m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/898726f10d4b4a5a971d1ec9ea0b236b : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2b65d3aabc7840f4b755cbf9d98f5f73 : SUCCESS in 25m 32s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/30f02776a36341bab52ef77c1ca5b4a3 : SUCCESS in 12m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ac878bf2e4bd406eb0f55a7855e6c760 : SUCCESS in 8m 14s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f3bc74eeee6b4d14b71026450d773618 : SUCCESS in 3m 53s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/14a6219995ec4a6caebb44ba18a8f953 : SUCCESS in 21m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/cc46fad4aa5640ee92334c92948c9838 : SUCCESS in 5m 51s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5f2e588c19a147f48fffb257785f1c87 : SUCCESS in 7m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/3f62fd85e16a4fd198b6fe2c22b94ffb : SUCCESS in 6m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2cea6ad8ebfd49f6ab031337aa22606b : SUCCESS in 12m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bf14fcc1242940d386e7abe0668ae1a9 : SUCCESS in 43m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/24187cda15fc4f389fb466b4f33d49f5 : SUCCESS in 1h 27m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6051a45ffac0c2f8d38a87c093b693e049cbcb0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 23:16:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9b55a02c74a1d604457c6148d111986706d09dfe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 23:17:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/668c3cd13611448e95d4f36af97e6026\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17d2831997b34d40893c661ea5b56474 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50943f704a6c7703231a308de9eba25e7a6564f0","revisions":{"50943f704a6c7703231a308de9eba25e7a6564f0":{"kind":"REWORK","_number":1,"created":"2022-07-12 09:26:00.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/77/849477/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/77/849477/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/849477/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/849477/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/77/849477/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/77/849477/1"}}},"commit":{"parents":[{"commit":"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013","subject":"Uncap warlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:25:59.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:25:59.000000000","tz":0},"subject":"update constraint for oslo.config to new release 9.0.0","message":"update constraint for oslo.config to new release 9.0.0\n\nmeta: version: 9.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib5e588be618d9bcac136b90d28d9af806a2d58ec\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I86033450b120b87f1709b22050c16b79fb378a75\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50943f704a6c7703231a308de9eba25e7a6564f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/50943f704a6c7703231a308de9eba25e7a6564f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
