)]}'
{"id":"openstack%2Frequirements~849478","triplet_id":"openstack%2Frequirements~master~I98282aa28c399cbd78ad7636fa20644b4d477961","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I98282aa28c399cbd78ad7636fa20644b4d477961","subject":"update constraint for oslo.serialization to new release 5.0.0","status":"MERGED","created":"2022-07-12 09:26:06.000000000","updated":"2022-07-12 19:41:06.000000000","submitted":"2022-07-12 19:41:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849478-new-release","meta_rev_id":"2e699fe17a3be8a6b1bd36f100ae7df7f7d03340","_number":849478,"virtual_id_number":849478,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 19:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-12 16:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-12 16:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 09:42:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 11:00:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:07:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-12 16:55:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"048b472eef5888d0e0c9e4e63bfa41f36ab92996","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-12 09:26:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bfe26c29d3433fc93d1d4433dddd6a5324fe7d0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:42:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f03064bca3514de2bd7776309c4a3398\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d531479475a94df7bb65f59f5a64bb97 : FAILURE in 8m 13s","accounts_in_message":[],"_revision_number":1},{"id":"aec38cd3e8a375a17e4dd45b5f22a7fede8efc97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 11:00:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ad18b0e8a1354d43b9dae34a8cb73488\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a44283010e44eeca68192f5c2749914 : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71c49c1a78634aeba2a5af973bc12145 : SUCCESS in 10m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1915673003341f8aa61f9f0223799b0 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8b67bef35a44b48acfb9ba74110f5dd : SUCCESS in 7m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78d5a3215ac948c0ac956afa5a4e5380 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1b72df01d59447bb92eefe34db813f5 : SUCCESS in 6m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/030e2d02cf7f4a5486a0ed391c5aaf89 : SUCCESS in 6m 17s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/29f1a51c923e411c95c8ac5ee249a176 : SUCCESS in 8m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/02c6f468ce1f4e2fa702e4b645f2d05f : SUCCESS in 12m 49s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e17d10b05d914f40945ec9f4eaa01d03 : SUCCESS in 7m 52s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/35842615995b4ed185ccf4687c2a5cc2 : SUCCESS in 5m 21s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a5ce0cee93924078a8f27ff6b1269518 : SUCCESS in 5m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/45f67c454dc74018b571508d7842bde7 : SUCCESS in 10m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e962191a9c50448cb881b4235719505b : FAILURE in 3m 00s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/86e8a8a3539f4dda94b04c30eec79378 : SUCCESS in 8m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a5d35c42a2bd41928cdd539ddbd0835a : SUCCESS in 11m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/893fb93bfd0e4752b832f351e00e77d4 : SUCCESS in 12m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6577849591c64dd382fe823068cf2073 : SUCCESS in 12m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3838b11d62654881bbe8a4bdf2804a21 : SUCCESS in 10m 42s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e6092071a3a449c9b51207632d73c28c : SUCCESS in 13m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bd1eb984123f4f95a133c96a3a3a1efd : SUCCESS in 6m 51s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4923018a5b8b4e0194f09c0c308c9e0b : SUCCESS in 12m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/300298a306eb4eeea7e32e4d35da357f : SUCCESS in 8m 07s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f18c1f873f414d7c865338b49dc74eae : SUCCESS in 24m 05s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b4a1f43b1f024df6ba29a0d33278e425 : SUCCESS in 14m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2fd95efe51a2415fbd52360ab60897f8 : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1f4d088cf2ed4e149289403349824f4a : SUCCESS in 7m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/11120edcf8fc47ba8a4356fab193908e : SUCCESS in 20m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/2ab8fd1ef47b4734ac412e61d99d0a61 : SUCCESS in 8m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/76647ac46b7748609ea898d8a3a11bba : SUCCESS in 7m 30s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/56c66f434bf0455a8ebafc1e83d1d584 : SUCCESS in 10m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e3bc444149b04aa882adffca22d7f7a9 : SUCCESS in 13m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6246d4b01ee445f948464693e11dee1 : SUCCESS in 54m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/06d7115e00464bfbb8a05b3c8ca88f44 : SUCCESS in 1h 32m 32s","accounts_in_message":[],"_revision_number":1},{"id":"bfc1403bc2acd2c37949509f820b479c183b619e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 14:07:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1b066b367afe096b094201eef584f5809bb4b921","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:23:06.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adbf4abeac2b484a98cb0610eff605d2\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/8aa20c017f7c4264ac62ad7b86de7784 : FAILURE in 7m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1432727ab93c0fa0f3298574930d59943523d627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:31:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7edddee982f24995ad6eec94fb861656\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2fdd599877149fdb7c23014da943a10 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/694e4d331db34df7b4636afbdc964288 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d2e6fc11ced4bd393913cce0e74dbfb : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5be9ddd4ff504d32b566187c7b9ecb23 : SUCCESS in 5m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d355c45e81ae4576a8bc32f29ec766cc : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7590ba519080426ab1e2ee1b53bc2e35 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8535b3115a3549399235d17942916918 : SUCCESS in 3m 49s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b7fc5864c5ee43e1a4e9cb32b7f9ba56 : SUCCESS in 10m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aab1baf9002a441496ec256d749365e9 : SUCCESS in 17m 00s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f4ab989e55a0411d960b543663c6a3c9 : SUCCESS in 10m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e1dcff4279a94b0d8721bb1cfea47c76 : SUCCESS in 6m 27s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/08f97036bbd0425490d3f5a35987f175 : SUCCESS in 7m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/adc8710833d343aca3ae0d06ae0bd971 : SUCCESS in 11m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c39d03eaf77e45688ccee4aeb7f297b5 : SUCCESS in 5m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/83722ead7731406d86cbb52c4bbdb698 : SUCCESS in 6m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/527751a3db7340418f6dba6cf033de3a : SUCCESS in 9m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/31ffc117ad694eb48113f52a59d0f12f : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a0429e069b1e4a80b709d9b9354b3662 : SUCCESS in 10m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/237764c9b9ee452ea7494fe47003e6c2 : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/79e55804291642e9b4d7738b994a4b38 : SUCCESS in 12m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/37aef85cc0fe4a748a43731d59c1ab34 : SUCCESS in 4m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/28c91780e85a4a509b1d8705957e6bb3 : SUCCESS in 9m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/bee006643258432e940cd037b07039e3 : SUCCESS in 6m 02s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a6ef8dc46e6648d9a4d8fd0a7966ff99 : SUCCESS in 29m 24s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/2798a4a030b24dafb4c8f97b2a825496 : SUCCESS in 13m 26s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ec1dd2aa8a6b4034acbc39c23f6016d1 : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8c789f60b6704a4fbcdad4199fc870d7 : SUCCESS in 3m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/aa41dd2c068e4951a798edf2315b854d : SUCCESS in 19m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/245682d1b4134736ae8cf12e415d2912 : SUCCESS in 7m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/08f27d562a764397a073ed5446882790 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f78039c7fe724e0bbf3df4b139797172 : SUCCESS in 7m 50s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2ac8cd29121d47ec958defa9e1caa2cb : SUCCESS in 12m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e8391be200874856b64b867e2867e83b : SUCCESS in 57m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/238e1e2e02244900ac6c0a2e3dc14fca : SUCCESS in 1h 36m 38s","accounts_in_message":[],"_revision_number":1},{"id":"791d33170347223858afaadc171c5d5806153ee6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 16:55:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a12c976292f13f1e0358d33614658d6d8fec9424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:56:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8049e558d8e43f9ce1b2145376f90b092baf9253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:41:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/923bd5b913634b5382ca6f3333eb2f69\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff466a8cfe4d4220bee558747eca5d68 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c27384533154345a0aa50391eee8b5e : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d16df01555ad4041a879011d858ad6fa : SUCCESS in 5m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b312c316534b45eb8467e42f7aece558 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c9434dfdd4b34c29b4955af4013302de : SUCCESS in 4m 51s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/32cea7c174cc494faa7bdc67cee931f7 : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/392dfe13beb74bf3a4f214dde89a8d17 : SUCCESS in 8m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/19842659ad4f4ebab3d66182351bc9f9 : SUCCESS in 13m 06s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0f6fcbb151a84e0eb07e569d9c974423 : SUCCESS in 12m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/50c6835f6d024640b3e66255eb701f6a : SUCCESS in 5m 35s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4597913e97dc48318506ab4efb94307b : SUCCESS in 6m 44s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/6808cc4a573146558909664dfb33e4fb : SUCCESS in 9m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6f43b645ff164c0483003fdb9c2f35b0 : SUCCESS in 9m 37s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/efe6c557922d4e2bb22cb729e969c47b : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2ca5f805aeb6454d945da5d27e925f20 : SUCCESS in 8m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2e7e83fa486a4574b471e6d649ba5ca7 : SUCCESS in 12m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/733aae0a618244eebc93f9b94b06621c : SUCCESS in 5m 37s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/162d6d2259a045a3ab8de69947fca82d : SUCCESS in 9m 45s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f31745d199414816bc82536c73da8b8b : SUCCESS in 6m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/d4a30e117889468fa1ef82ebaf50b7af : SUCCESS in 21m 56s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4f7c646094234a07a733d452162e6077 : SUCCESS in 10m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/0dd0ec52a9154d3aadce2876c2733dbc : SUCCESS in 6m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1c0bd2d595d54505a872855d14fd5eb8 : SUCCESS in 4m 42s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/0a09c73cd3e54b6a961c095935b4dfba : SUCCESS in 18m 44s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d89de488aa404113ab8a1a6966d33434 : SUCCESS in 6m 49s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/da89e613961b428ebac24923beb7842a : SUCCESS in 6m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/7e8870ce59a24e0395b522c9e97116b1 : SUCCESS in 7m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9b58ef888269433fa16a84573c7a1cbe : SUCCESS in 11m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cb7ef776472b4058990ab615a440a227 : SUCCESS in 41m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d95c7a21b9940f7a3859c1472903524 : SUCCESS in 1h 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"2e699fe17a3be8a6b1bd36f100ae7df7f7d03340","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:41:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ccda49cc17a491ad869f00e3d92c0e4f88576a33","revisions":{"ccda49cc17a491ad869f00e3d92c0e4f88576a33":{"kind":"REWORK","_number":1,"created":"2022-07-12 09:26:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/78/849478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/78/849478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/849478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/849478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/78/849478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/78/849478/1"}}},"commit":{"parents":[{"commit":"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013","subject":"Uncap warlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:26:05.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:26:05.000000000","tz":0},"subject":"update constraint for oslo.serialization to new release 5.0.0","message":"update constraint for oslo.serialization to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ieca97a0b177a3274eef71ca22631f08d240c172b\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I98282aa28c399cbd78ad7636fa20644b4d477961\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccda49cc17a491ad869f00e3d92c0e4f88576a33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccda49cc17a491ad869f00e3d92c0e4f88576a33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
