)]}'
{"id":"openstack%2Frequirements~849479","triplet_id":"openstack%2Frequirements~master~Ibb558999fd7978ab6b5a76993f422fe2d121e179","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ibb558999fd7978ab6b5a76993f422fe2d121e179","subject":"update constraint for oslo.middleware to new release 5.0.0","status":"MERGED","created":"2022-07-12 09:26:34.000000000","updated":"2022-07-12 19:41:03.000000000","submitted":"2022-07-12 19:41:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849479-new-release","meta_rev_id":"1556ec124c4900e6afd692b987b4f6d09acfd439","_number":849479,"virtual_id_number":849479,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 19:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-12 16:55:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-12 16:55:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 09:39:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 11:25:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:07:13.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-12 16:55:25.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a0ffdc802998741ed0d504b0052b364897056312","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-12 09:26:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"629d5873064773612d3ab332ca417ba97b77be23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:39:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b53345610844b68b0ca2813f2f2f09b\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/7b764e6e59f944789e94a4fa7812dde9 : FAILURE in 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"d6bd3b5d570f8dc776b924aef0de50af863d2ab6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 11:25:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/574f9987abb043c0813d64b0a5830457\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f03ed9dcdd64270a78283ccc1f62fba : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66da830021a3498d9966fce13468d0ab : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d1248d90e2ee4773bb48203e6ef77996 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd8599dc85d14a7d873c42c8526c4e56 : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/452aae0c16d14b39a445cd12eec246aa : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a70cfdbe26043a99ee17bdbe168f1cc : SUCCESS in 5m 26s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4730ff89f59d492c92c20bfbfd0213e2 : SUCCESS in 5m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7c078cc1f9634205b28a5e805ae95dd5 : SUCCESS in 11m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b432505f83a645069a169e633cb3b8a2 : FAILURE in 5m 56s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a8df527b91ec4d30bdbfccba361a81a5 : FAILURE in 5m 18s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/de741612a04b4f3b81013554fbd73512 : FAILURE in 3m 51s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/922836f107d84f1196b8cadbff4c24d5 : FAILURE in 3m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c7274d5f756e4761ae25f49ba04b4015 : SUCCESS in 14m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/56a00dada6b543dc9db9860407262689 : FAILURE in 3m 22s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5e49eb4988e14be98f22422d002411eb : FAILURE in 5m 35s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f725fa79456c4398b8c61ef3c89dc78d : FAILURE in 4m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/953d1d670ba44ad9bcd4c6470bc9756b : SUCCESS in 11m 03s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0c2f77e8f182418483d51ce6ccff7c68 : SUCCESS in 14m 11s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/59776db017cc488284e812b495d74250 : FAILURE in 5m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/3ad21c6f019644ab8203f87262a01963 : FAILURE in 5m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/271999213d914aa9b5f0477bfec58021 : FAILURE in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5cf704d190f640dd9b8483f7404a78d2 : SUCCESS in 14m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/2f3bb5f96c0b440482e73110f6e798cf : FAILURE in 4m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/44eb315d2bc641cab5b7f34855109364 : SUCCESS in 24m 33s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/1f45ee7f8e5a42199c21da19e6393259 : SUCCESS in 12m 44s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/8048e2053d064902b1ea018daa9e5879 : SUCCESS in 7m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/861b49da1af04ddea848bc969a6e25a5 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c11353b4610146f8aa78f012ecc030f2 : SUCCESS in 21m 46s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e5b129cbf5134e9c839b0e710f8fee44 : SUCCESS in 6m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0a32b58bb8674acc99c58f97baa9cf1a : SUCCESS in 4m 57s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/67e63e0160bf4404af47e4e92e7ede88 : SUCCESS in 7m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a95fcc4b2fcf4f739d850eeb7b901f4d : SUCCESS in 12m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8627fb6fbb674651853854bb6667dd71 : SUCCESS in 49m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b6bf49bc3ac464b83be6658510388d6 : SUCCESS in 1h 52m 36s","accounts_in_message":[],"_revision_number":1},{"id":"b18436b82ae85cc9fd5dd97f6e3fbba60d68a0d1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 14:07:13.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"daf98ba08f1fb623aa944ac76440fa97d7c69386","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:20:43.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e7b423e526c452fa078dfcc4904a9ac\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e05fc1040fad4c27917677642b4637fa : FAILURE in 11m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f920c3cd9c0b862499f60c1e7bc61f434b0283ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:33:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0233d9f5662443bbbd3d4132a613881b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/411d4fe925a74200994275c496748b3f : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc946f3464b14d96bd6c49f20752adfd : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e9764c51b7841a9853c809a6c8e3e02 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a0454bb5a184c90837509e734a9ba80 : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/def0f54c78584adf9f8fe7f33bd37d42 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/45ca2fc24d894cf3842aa0f7da015c09 : SUCCESS in 5m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fa4b809016964cbdb951099487081a0c : SUCCESS in 4m 37s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/461156f627a2474494c06326f57189cf : SUCCESS in 9m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/67f753be95d144beac94760e673d2310 : SUCCESS in 14m 35s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a23b06b5bc994c6ea1338ae9a3b996e1 : SUCCESS in 8m 37s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/f0ea37a3949243439904c2e14ef76010 : SUCCESS in 5m 57s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e933e3b9db974107bdb7e98795891522 : SUCCESS in 7m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3645d48070bf47938ad265cf8efb63ad : SUCCESS in 13m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8d8c392b29ea4b629878c0d7c4868d09 : SUCCESS in 5m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9d951eab8b2f40b39114011bcab548c6 : SUCCESS in 5m 52s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/9bfbb29a4cf54cb2a79be5ceae9e510a : SUCCESS in 7m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ffcf5fbd22c54cf38c5e805d8bde0fe7 : SUCCESS in 11m 11s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2fe5b9f1e65648d7ac232c3877568e6b : SUCCESS in 11m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/df89c92db9d44005833e6c28aaa3193f : SUCCESS in 10m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0704aea6f4144aa2a80a73031fa63438 : SUCCESS in 16m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/220d690ec29a46c3ac7d4f9a9af53b2c : SUCCESS in 4m 35s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4e01b6e126f3498d82e6487383738e3d : SUCCESS in 10m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/a2d241d42adf4f0cb6f8b30766f24446 : SUCCESS in 7m 29s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0705635e7eb84075a6853181862aa1bf : SUCCESS in 23m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/33225ca4bb6949fc91822a12b12dd002 : SUCCESS in 11m 15s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ec4589401f834419a2fb5541d4b15934 : SUCCESS in 4m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/298e0b42fae348fa9f84cd008d260f51 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/eef775e8959b45d994917015243b2abd : SUCCESS in 23m 39s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4687703a1fb444ce9a420c341ecab836 : SUCCESS in 7m 40s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a26752aa79b24adfba5264894800c0fa : SUCCESS in 6m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2121d44227b04f05a3279ac8d41ef93d : SUCCESS in 7m 36s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/51878594ca7f4ee9b39fb4168fe43566 : SUCCESS in 10m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e4265d1b90df4c7da8fde6ed623d5e5a : SUCCESS in 55m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d2c06a249bed426cb823976bb4ba619e : SUCCESS in 1h 41m 49s","accounts_in_message":[],"_revision_number":1},{"id":"779bb79ead005aa5d32e3d17bbf6656df763ccfc","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 16:55:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f9e601669a19fa129ee7e5c2f2d616214a07358b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:56:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b24f6d782f52ccfe3123c539819f5ddd18a97f18","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:41:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92d99029dcfa46c7a2cff6fa424d4bc4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6d58f3fd7f4add8dbb492bdd7eab2e : SUCCESS in 8m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9147cc28afa41fc925384338c8b4061 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9da05f7a49548f794931e509eb9bb81 : SUCCESS in 7m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70b503e1486242f68f78538338aa3be1 : SUCCESS in 5m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9d0cf67d0463402f900516bb47ca691b : SUCCESS in 6m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fc11658e31c845f292a21db7cedb150d : SUCCESS in 7m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e1784ba7afee4e3baa7e1ce6ce789f27 : SUCCESS in 11m 04s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/373c521b1a854c1591c1db58d205b13e : SUCCESS in 16m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6436634e63834bbab869cd0e4d7b67e7 : SUCCESS in 15m 05s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ac2f0a040d6c46778e83fc9276d99cf0 : SUCCESS in 8m 33s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f339692c439e457b80ee013b18619b14 : SUCCESS in 7m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f689f368366142debeb53048d28121f4 : SUCCESS in 12m 27s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e7198999188c469daf78ec751cadfdd3 : SUCCESS in 13m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3b7403be362d4342b750b49614159c57 : SUCCESS in 8m 47s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/115ae4d3aeab48db8278333cc3f3000c : SUCCESS in 12m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ae6a03464af243b79468217366808678 : SUCCESS in 16m 09s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7b79ead7825b4cfc955c7f21b97e64b3 : SUCCESS in 5m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/dbbbdc4bba6549f98e08c4d802f7ac4a : SUCCESS in 14m 14s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/52f4b4a373684c9d8f7a0f105ce9c66e : SUCCESS in 7m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/f48d38099e0a4bfa874b8f243a0c9ab8 : SUCCESS in 25m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/37dfebbc45d84beab4c81ce29b1d9878 : SUCCESS in 14m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/926cdb89e3af447eaa76c17b5f85919b : SUCCESS in 8m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cb52f2e48ca8441194ffc73feeb93f09 : SUCCESS in 5m 55s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8905f42fb7cb4e50812101aa667f6dec : SUCCESS in 19m 47s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ae2ed1884fe04c1eb46dbba98426cf7d : SUCCESS in 9m 42s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7c588e44362e4e4b9fe91d8566fb7c7f : SUCCESS in 8m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c00cb91831bf459ab2f81e31871dd452 : SUCCESS in 9m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fe1fedb19dba4c03b318062b70ab5fed : SUCCESS in 12m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9dfd087ad61749de8b17acebe5fe09fa : SUCCESS in 48m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/931be0158d64491fa37abb1f9c9b2e62 : SUCCESS in 1h 46m 03s","accounts_in_message":[],"_revision_number":1},{"id":"1556ec124c4900e6afd692b987b4f6d09acfd439","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:41:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"156908b3226c4218419d545955dfa80a741fe4e1","revisions":{"156908b3226c4218419d545955dfa80a741fe4e1":{"kind":"REWORK","_number":1,"created":"2022-07-12 09:26:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/79/849479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/79/849479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/849479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/849479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/79/849479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/79/849479/1"}}},"commit":{"parents":[{"commit":"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013","subject":"Uncap warlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:26:33.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:26:33.000000000","tz":0},"subject":"update constraint for oslo.middleware to new release 5.0.0","message":"update constraint for oslo.middleware to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I2d6c6842ddeec1ee645fac74aff06dd635e499e7\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ibb558999fd7978ab6b5a76993f422fe2d121e179\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/156908b3226c4218419d545955dfa80a741fe4e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/156908b3226c4218419d545955dfa80a741fe4e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
