)]}'
{"id":"openstack%2Frequirements~849481","triplet_id":"openstack%2Frequirements~master~Icc937791cb0f302ad7f520cb4cc98591deaa4bf6","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Icc937791cb0f302ad7f520cb4cc98591deaa4bf6","subject":"update constraint for oslo.upgradecheck to new release 2.0.0","status":"MERGED","created":"2022-07-12 09:27:42.000000000","updated":"2022-07-12 15:59:10.000000000","submitted":"2022-07-12 15:58:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849481-new-release","meta_rev_id":"dcb20d3bf34a6c7a9a9873d76ebb520c782aa111","_number":849481,"virtual_id_number":849481,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 15:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-12 14:06:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-12 14:06:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 09:49:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 11:01:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:06:58.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9f73a56151ea9225ad0b1a2408628212fc0aa597","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-12 09:27:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0aa3248c28d8a208b4645b23cdea04c98eb9b5d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:49:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe75e892489349c68f7d6a9c6271f32a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d24cf3a0db4e4c189561950f8768dc19 : FAILURE in 7m 54s","accounts_in_message":[],"_revision_number":1},{"id":"78acbd9f2adf4b42842710344cba6e3582cd25e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 11:01:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cfdcb6ec8bb54fc48948b843955c967b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/669effce909e4ac196559a9925da60ba : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd71a06e07cd4021af1b99313f372da2 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1d17cec8dd604f3eb71a50c2a347334e : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/056eac0a2d40462d881f24bca16afedf : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/960e6ad72838491fae3cf81280d7e17f : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f4ac1f3403e844e9a123ae24cbb7d044 : SUCCESS in 11m 09s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/8200135b6a8b4863b7082c2b43d950b7 : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c9eec8e9f92241589e6e125d77928050 : SUCCESS in 7m 33s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/845c120321ba4fb6becdcf5c8ff2bd0e : SUCCESS in 12m 24s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/aaaf7f27f2e24ad699403d6452dc0f57 : SUCCESS in 7m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b6adc286f5dc4a5abbd70f0272fb97dd : SUCCESS in 5m 44s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/f223dd5e514c43a6a0b71ef0aa516b0d : SUCCESS in 6m 22s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bc85d2170fdc4732a6078cdf27926392 : SUCCESS in 10m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fe1333808c434234ba63f33ba4b2e412 : SUCCESS in 5m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ee5e6d1cc5454963ac2671b95fe699bc : SUCCESS in 7m 38s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a54ae1c17f4b4cafa34ded20cb55417a : SUCCESS in 9m 01s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a9354aabb18f4437a57d71baa6c22252 : SUCCESS in 9m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/60f994509d8740ff982e12645287dba5 : SUCCESS in 8m 58s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/200af69c8d7141789e57dcb94fbee5dd : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4a73aa7f3430453ebd9028f8c69c0b69 : SUCCESS in 10m 30s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b7bb4cc034874367a6e6f677374d998b : SUCCESS in 4m 42s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/7f2d0f0a8ccf4d529f451a15ce1403e4 : SUCCESS in 8m 42s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/04a1fcb3e2f6448f8cf3bb2b9fd706c2 : SUCCESS in 5m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/744d83b1ae3e4973aa0287632c9ea71f : SUCCESS in 25m 19s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6228dc5ba43f4c48a793a8c7b8ebcff7 : SUCCESS in 13m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/12b855b0cada462a958aa020b98a35ef : SUCCESS in 8m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d8b151fc308f42c689c9ac5f8d3d2f7e : SUCCESS in 3m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/d0862ca4be6e4a8abb74fb7111e1d496 : SUCCESS in 22m 02s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/4813451ccc8f41e9bc2ddedd7588fb8b : SUCCESS in 6m 33s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a43b185401e2426fbce5a4eb400f5a2a : SUCCESS in 5m 39s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38b967e288f1485cb7b4593e8a914c6a : SUCCESS in 6m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f0685880b021477f8aba707223c12496 : SUCCESS in 12m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed0d7fd9079a41ee9d68faed17de720f : SUCCESS in 55m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/357f330c638944979695bdbebc17028c : SUCCESS in 1h 27m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9407e48cc1e8491731f897b82c0fc798557ffad1","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 14:06:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a6dfbafb3066fa4543ac63aac0bd879ed1ea57dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:08:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0eaf48adf00470e7e36b510c8c0bdef8e4b6b1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:58:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db0570a5309d4d308226520136bcd248\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c59315d1e004a4e945d0a63b15fb9f5 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5282382a7af2408ba1098a45fe4d5792 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2cc75a32c07b4f5897da2b7b753457e7 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c0c5935ee514543852ae86b2df8d8d5 : SUCCESS in 3m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca5380308be74a878163418b77ec5199 : SUCCESS in 3m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/827f064faa7349a1a8dbde3656db9858 : SUCCESS in 3m 12s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/5c2f3755b99d4a019954d1d1d59d380c : SUCCESS in 10m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/349feb9d0f6345b7941b081db948f416 : SUCCESS in 13m 44s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/4947c4e638f64186911a6a8c0a3ddd4a : SUCCESS in 11m 36s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b7cc8e9afeaa445383067e672b8e646a : SUCCESS in 4m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f97b8b03b2f94d01a3383457b8d5b2f8 : SUCCESS in 6m 25s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2b52800c735e4c3f9790a23f79469462 : SUCCESS in 9m 44s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/cb62b14ff3834c36a327f3179bf100c1 : SUCCESS in 11m 20s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/418c7d411e164a309ce4e46b1cfb7226 : SUCCESS in 11m 12s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/14fa92a4f11e4696b9d09f8472b72b32 : SUCCESS in 10m 51s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/35ac505b568445fc8612b0a29d870fc8 : SUCCESS in 12m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/39bde3f30b33422588e1dddb51403815 : SUCCESS in 4m 24s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/0ae57880a78e405eaefe893d9520dd1e : SUCCESS in 11m 31s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f7a6e2f10864ed48edb5136e248a579 : SUCCESS in 4m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2eeb33cef46c4c558f29f5af53295e2c : SUCCESS in 23m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/da56283c54a34a55ab0d31550580afc7 : SUCCESS in 14m 55s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6f39ae5921f94541ba52ed22455c3969 : SUCCESS in 6m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/931f2a1c808c4a42bd22b43c4f026f89 : SUCCESS in 3m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a92524c674774241b178a465bfd5fd04 : SUCCESS in 19m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1d7150749c97412793cc87ddff3b8a83 : SUCCESS in 8m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1592b5208bfb4a8a80b828c2fda2a3c9 : SUCCESS in 5m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/727395c4c21a4250afd2b4b691bda16a : SUCCESS in 9m 33s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/bc25a1267ba4459ab33a895f0addd28b : SUCCESS in 10m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d20ed8b2b705476c8f888a419ecf8ca2 : SUCCESS in 1h 05m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/13251bed6f8d405087b4abf5eaeabb25 : SUCCESS in 1h 37m 09s","accounts_in_message":[],"_revision_number":1},{"id":"033b62beb7d15cbc2377f87a6516aa88bd731eb1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:58:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dcb20d3bf34a6c7a9a9873d76ebb520c782aa111","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:59:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d55f19734fe4aea81d6bcf46fc6de99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ff09e1eedb24ec1b49169bb2c829270 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a8f43207f3a2a99d550d8ed25d124d1cad320545","revisions":{"a8f43207f3a2a99d550d8ed25d124d1cad320545":{"kind":"REWORK","_number":1,"created":"2022-07-12 09:27:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/81/849481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/81/849481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/849481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/849481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/81/849481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/81/849481/1"}}},"commit":{"parents":[{"commit":"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013","subject":"Uncap warlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:27:41.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:27:41.000000000","tz":0},"subject":"update constraint for oslo.upgradecheck to new release 2.0.0","message":"update constraint for oslo.upgradecheck to new release 2.0.0\n\nmeta: version: 2.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ie58e7114dde35568fb4fd3eb66f3c1d615352207\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Icc937791cb0f302ad7f520cb4cc98591deaa4bf6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8f43207f3a2a99d550d8ed25d124d1cad320545"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a8f43207f3a2a99d550d8ed25d124d1cad320545"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
