)]}'
{"id":"openstack%2Frequirements~849482","triplet_id":"openstack%2Frequirements~master~If0172f133f0fc504f4269096325ecdd25eb8535b","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If0172f133f0fc504f4269096325ecdd25eb8535b","subject":"update constraint for oslo.db to new release 12.0.0","status":"MERGED","created":"2022-07-12 09:31:20.000000000","updated":"2022-07-12 19:41:09.000000000","submitted":"2022-07-12 19:41:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849482-new-release","meta_rev_id":"969d666f610df700bdf50c73692f072c33d7df10","_number":849482,"virtual_id_number":849482,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 19:41:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-12 16:55:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-12 16:55:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 09:46:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 11:25:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:07:15.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-12 16:55:26.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e0d78e10e8d65493796ba5dbb2503b86b248b16e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-12 09:31:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdb86cc2c7005631addec263abfd842a034318fd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:46:07.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e78886a1b2f406e813a7af467300585\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/f98994e83d4343f7aebbcb265f23d6ab : FAILURE in 10m 21s","accounts_in_message":[],"_revision_number":1},{"id":"cf940d28d93d8bc866a8a6067520ade4f85b8bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 11:25:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ca6942b6d4cf4f9e907e417bafacebbc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435615c3be7945c29452c8b23cf0a62c : SUCCESS in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fcbbc8143e44c12b0509e8a2789ff6e : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6a490ea325f24839b553be9eabc46363 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4fa73bbd393459faea8413403d22394 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3eb1f46c1c9c49dea59347fcb3957ffe : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8265354edacb4ae9a2f583ecd1aa7341 : SUCCESS in 4m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fdd0bbd6844e45029dff262042146b98 : SUCCESS in 4m 00s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c52240a3122143eb96f6b0ea5eabcbf9 : SUCCESS in 10m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5739d6b2b7ef4f149c93f51c768d607b : SUCCESS in 14m 47s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/56dfd43f2c114ae2ad341ad14b7444c5 : SUCCESS in 9m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/698b464f3b4f46189e2b2bb602bf64c5 : SUCCESS in 6m 26s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/37e2731eee2d4665a87c46eb375a0113 : SUCCESS in 6m 57s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2c69c45edcd14f7eb2983981c7fe9bf5 : SUCCESS in 12m 33s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3622210cb3964da9bcd73edbd9ef5abb : SUCCESS in 5m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3a19605e9e6f493b967cffd7a45500a9 : SUCCESS in 6m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bc192d094d384f259e8d15f7cd352b85 : SUCCESS in 8m 41s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b9d73c0a453f47179e3f4a0f51c89669 : SUCCESS in 11m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1bbcad3aff954ceeb0d6634c5dd2ca29 : SUCCESS in 10m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e7b00de70484416b9335cbf9ba29405c : SUCCESS in 9m 54s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1ead83d762ca47e98e176beb4acf2914 : SUCCESS in 13m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1bda2694b359481f9fc8cf4fdc7f1a87 : SUCCESS in 5m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/bfe123a179ab4bce88f797f57f10ed67 : SUCCESS in 9m 40s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/540807f0f48541f4b255e7a0974f255d : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/90a910a3b97f4cbda637cd9eb5cccf6a : SUCCESS in 25m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f811e78b0215456d9d70f04df4584708 : SUCCESS in 12m 56s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/2c53e2a04b7740ddb4674b4f03c8127d : SUCCESS in 5m 58s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/59c86405eaaa4ac8a4b0ee2c1b85e866 : SUCCESS in 3m 14s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4d3394b0f78e4e899a4ebd1305b08b71 : SUCCESS in 21m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1f9d628875e94c6e98a0258a5c96caf4 : SUCCESS in 5m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/75a89015280842b8aa5596c99f87bc92 : SUCCESS in 5m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/de05f113a76c41f680ed1e7a945d3cdf : SUCCESS in 6m 40s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/cd4aaa84c4204a3db5d48b4ba73e94ec : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ca230acae0554434b28f328ed439f2f5 : SUCCESS in 55m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b03026403ba945c5ae3e0f06795b1847 : POST_FAILURE in 1h 42m 24s","accounts_in_message":[],"_revision_number":1},{"id":"675be169f4e26babe0d3627a1e41ecaa402c9171","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 14:07:15.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"27d521eef9ef8fa8e04bd760a993fc379889c7be","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:25:34.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5aff25df35a74ff3b4ea688d81f59b25\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/9dca14bcd24743aab16461a7f0156f0c : FAILURE in 8m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0de44c15c401cb255c99ea4429b72738ce2f026d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:26:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f6592f829134035b1b8ff208d0c4ae7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69bb74898dfb4e5ba39074c9505bb381 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dbf1929fe664a89a266b1548b31665c : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0e36e5359f5441e9a300c92d62b4ddab : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/04d93774722d4653a0511f21bfddd240 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29eec0eab4cb42768f97ba937e7d504f : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b531363796d947e89fe2618cd4817bb9 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/fb948a9fb5bc4890bdcc6b08b65a327e : SUCCESS in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/132381b27356405ab12bed5f07dbc919 : SUCCESS in 7m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ffda4f0796064eeaacd6e4f95f5f6f4e : SUCCESS in 13m 46s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d8a9de892c1c4304b8ecf8a89c12f158 : SUCCESS in 8m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/cbb0b80d460a483584721f4dc95cec45 : SUCCESS in 6m 13s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/9f1d0df579af4de194980c062a7b3f46 : SUCCESS in 7m 02s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c5aedc69dcc24a5da8571c021138e465 : SUCCESS in 11m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ce535d4c39784de8a0857a154721f81a : SUCCESS in 5m 30s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/507a7c5333b94bbfbf5d4d82322fb30c : SUCCESS in 6m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/591aaf5451e74d93aa19884158e458ea : SUCCESS in 7m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/8915467e99cf4ce5a19a3d31d67ab97e : SUCCESS in 10m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/fadcb6fa83504df980a8215044aed0b2 : SUCCESS in 8m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/62b0f899727048edbd36f4ec3a4c916b : SUCCESS in 7m 49s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f97981fcffc941548e5d79dbafbfc991 : SUCCESS in 11m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8655c410457c4782be06eab35caa6c66 : SUCCESS in 4m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/62bef28dd7ee4ab5b30eb024cca5efa4 : SUCCESS in 10m 23s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0fcd64b2750243eda1090954a2de080f : SUCCESS in 5m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9fda0e1cb83f48f6b711a613f9a27b5a : SUCCESS in 23m 38s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/56774dd51daf4d79b2016d86682d766e : SUCCESS in 12m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/36f203333be1474682d3babbd3edea85 : SUCCESS in 5m 38s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9266c5c8648c49bda06cf4eda1818d6c : SUCCESS in 5m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f212c78eb0c047f9a154b2df78306508 : SUCCESS in 20m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e070e217f1e64ce3953afbf45bb6d95d : SUCCESS in 6m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ff942abe8e10454787ea761660767035 : SUCCESS in 6m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/62acd517c0384175a0b88416d0e33bb6 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/292336c48df44bb1946a98a0264b2cb0 : SUCCESS in 8m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b42a2c2e585a4d3ca538a930591ea9f1 : SUCCESS in 44m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3d2b0c74d7fa485389ac7999c5ac9976 : SUCCESS in 1h 33m 57s","accounts_in_message":[],"_revision_number":1},{"id":"07c4d422d7857a4e74e0a1f1c2f342ab599da431","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 16:55:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9355621a5ec417df1277c1ed541ed01b4350f55f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 16:57:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99c2f28d835698abcced0eba2b972cc117834a00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:41:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ce895f50daa4be69e4960bc59640e5c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7cc952b093247ea98785b3f2464a261 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30e42f2d9500499da914400a2b232d29 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dac8219494754fcd9ee6ea79b611a765 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2ae400671e14c7c82ed3ad315115c0c : SUCCESS in 6m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b1ec598591142a6a0e3620be8e59b4a : SUCCESS in 5m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0feee99ab25943c4accbcb4c70f40b9e : SUCCESS in 5m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/556ab64ebcdf4e1fb5ba83251da57425 : SUCCESS in 10m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ae083e05652c433fb1bd715d6d5239b0 : SUCCESS in 16m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/230ab001e3f14edd84b9680b5842c10f : SUCCESS in 11m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7fd237e7c9984001aabe6744ade12c21 : SUCCESS in 5m 55s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/59f10cc2d9364f448ff8cb77a54cf365 : SUCCESS in 7m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c47b6863be2c4cdf84170df8e6f6a9dc : SUCCESS in 8m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/06347dcb7920401e9286e8f3308bed34 : SUCCESS in 11m 19s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d884561fc99e4ecaaff28ebc109c8145 : SUCCESS in 11m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/112cb287d6f84d40b7799573cfa85974 : SUCCESS in 9m 36s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0924ca012e294738a38db80297d1ac2d : SUCCESS in 14m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/03bd698bf9c2410996128a35a775f33b : SUCCESS in 5m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2ea7afc1a70a4ebda3db1843f139e7b8 : SUCCESS in 12m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/04a52cd1a39b417da8dc152ec261ab65 : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/654292f5835446b18c014dbdf743c516 : SUCCESS in 22m 23s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/91c052f761af400683dc56070f160d03 : SUCCESS in 14m 20s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/7f1b61c44b354f3cbdf7e76e4334c655 : SUCCESS in 7m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7f557ef103414437901a60a9e4fb1760 : SUCCESS in 5m 43s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/bb617ad1b49249728db3d932853514e9 : SUCCESS in 19m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6154e969c2f346668a6dbbcc0fe53f6a : SUCCESS in 8m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/09c41bb1dab54ed880e33d5461b73a7e : SUCCESS in 6m 54s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/452650582cb440da97c0ecffdd31bc54 : SUCCESS in 9m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3ebe721b25e045f5a94f4e8566ed3c09 : SUCCESS in 13m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/24b1ffe50c0245ecbd669de16fb42936 : SUCCESS in 53m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bdd8ca7cce5f435e9317d05be93aea5c : SUCCESS in 1h 39m 22s","accounts_in_message":[],"_revision_number":1},{"id":"969d666f610df700bdf50c73692f072c33d7df10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:41:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3bc1714bb9836cd6df408b47dacae5bd1ab3114b","revisions":{"3bc1714bb9836cd6df408b47dacae5bd1ab3114b":{"kind":"REWORK","_number":1,"created":"2022-07-12 09:31:20.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/82/849482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/82/849482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/849482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/849482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/82/849482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/82/849482/1"}}},"commit":{"parents":[{"commit":"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013","subject":"Uncap warlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:31:17.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:31:17.000000000","tz":0},"subject":"update constraint for oslo.db to new release 12.0.0","message":"update constraint for oslo.db to new release 12.0.0\n\nmeta: version: 12.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ifa18eab0f4fc2f2ac728efc5e22950b41834bc89\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If0172f133f0fc504f4269096325ecdd25eb8535b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bc1714bb9836cd6df408b47dacae5bd1ab3114b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3bc1714bb9836cd6df408b47dacae5bd1ab3114b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
