)]}'
{"id":"openstack%2Frequirements~849484","triplet_id":"openstack%2Frequirements~master~I2ac4f2d5ff479228fe7fa6fdf9f664e05623d3f2","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2ac4f2d5ff479228fe7fa6fdf9f664e05623d3f2","subject":"update constraint for castellan to new release 4.0.0","status":"MERGED","created":"2022-07-12 09:36:39.000000000","updated":"2022-07-12 15:59:40.000000000","submitted":"2022-07-12 15:59:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849484-new-release","meta_rev_id":"f6479e3e02213a55343c73dc70d496a12056c57f","_number":849484,"virtual_id_number":849484,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-12 15:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-12 14:06:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-12 14:06:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-12 09:53:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-12 12:51:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-12 14:06:59.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"d27539e32b9b44f0e21bf83deecf5dfab3f43a00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-12 09:36:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6149224a0564da788f34d280266d7aa8449151b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 09:53:09.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b241c40f6da84942b98250eb734f4c4d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/fc190853dbdb4349ab3f55f1b2c93fc9 : FAILURE in 7m 56s","accounts_in_message":[],"_revision_number":1},{"id":"04e1babf1126d39dd682a0410cd391105e9aad67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 12:51:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33a5e091bde14a72969d192e6b057f34\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8c0efb18cb2431f90b3320761a586e2 : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c45d88623941119f2d1cecd1faef66 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06443190ecfa4b7fa7b78be9b0464979 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e83a6f96017a4b2f8c7ad24be1c99f81 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a862dcae6ae64582979f893a514a9e0f : SUCCESS in 5m 43s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9bfe6336cc94823853cada110854875 : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/603e85e9a3d64d498b51c0b53ade456c : SUCCESS in 3m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/526f5584df3b48a5bfa54e27dec77830 : SUCCESS in 8m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0b7cade8281442b0afe1a4a4df0c2a1f : SUCCESS in 12m 52s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/70e0691b116245b390b1d6c0ea473c6f : SUCCESS in 7m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1c3cac5d119743e090ce06894591bea9 : SUCCESS in 5m 01s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/f75a0c7e04fe4003b23675f104f82bd9 : SUCCESS in 7m 36s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/22db6c0969964a7bb6505de322cb0586 : SUCCESS in 10m 43s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/cbd8643ea32f4999943cedc7d418dc19 : SUCCESS in 5m 25s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/02073834ab47459390d89a2e53ee2372 : SUCCESS in 5m 20s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/bee9d2f1da674fe0b3f98dd0a346b87b : SUCCESS in 8m 08s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0631f120451a423d95165bf7530e184c : SUCCESS in 11m 09s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/93776f9b45764b2a99137a69c51f59ac : SUCCESS in 8m 22s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/131982019cdb488d9220cf6977a5e998 : SUCCESS in 8m 46s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4f5f7033710c4058a2e58d26bb45e080 : SUCCESS in 17m 50s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b49189303a65487793aad2f5c73207ec : SUCCESS in 6m 36s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e2aa6a88092145ad8e36948167caa89c : SUCCESS in 13m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6222c6a03e594ef997289c1c3cb38cfa : SUCCESS in 6m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6f99ed12069e421ba3004322622d356a : SUCCESS in 25m 34s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f371ca8f1bb6492b99b9a94039d2f9c5 : SUCCESS in 15m 17s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/ccde9871373a4fd5bf585cf3cc5c2a00 : SUCCESS in 5m 56s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c56102ed83be49f5aff29b630d564542 : SUCCESS in 6m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ad3713b82e7742cb9dca16ec98e307b0 : SUCCESS in 19m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5c842e5de21844b781096e6fd119b1f3 : SUCCESS in 7m 04s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0232f3f79be1480b89cffa38c8eb7da3 : SUCCESS in 8m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8984033da15a458683f6369f906cdeac : SUCCESS in 10m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/302c4e3db04d4eaaa0b38f18bbdeb813 : SUCCESS in 12m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/839a1b7d5f444525b4145c99e2aeea55 : SUCCESS in 55m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/42b4eff9d34b4e068c1ca45495a942e9 : SUCCESS in 1h 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"9d357fe689d739f14ef9f3fead17bc2e15a95ee6","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-12 14:06:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea224b7b136bb4e273ff16fe5a4b7a7b355cf8e2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 14:08:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"092ca98ccea5d16afaaffbf624d323910dd7c4c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:59:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5f146ece43e468592bd9f042348b671\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2a7f781e1a948d5aded0bec0331965c : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b70598da1db84057a4b237ef34646744 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90e9923b38b34e66b8f078b56d584ac2 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bda01daee60492384dd6d3b4195892b : SUCCESS in 4m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f864b459d6aa460bb9e82c549359a640 : SUCCESS in 4m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e180e0e1ee7149fc822ef5a9e25ae676 : SUCCESS in 5m 15s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/26695b3958a24b8aa62564c16c169d7d : SUCCESS in 11m 24s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/efc6ad8e248040ea86fbbd8570845c7f : SUCCESS in 16m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/bef501f044ac42f8ab4ea47436f16fb3 : SUCCESS in 12m 20s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bec45162e33c47f489e381f6b20f30ec : SUCCESS in 5m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8a1105ef17464d969b7eaa75ad384ce0 : SUCCESS in 7m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4a5fbe89fef5410e8c104344d2f868a4 : SUCCESS in 8m 45s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/41d706a3b255414a91a822db7016bde0 : SUCCESS in 10m 30s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e73c892e4d844fab3c65c3df30843b2 : SUCCESS in 9m 52s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/78a265bcdde8493d8af196800201c134 : SUCCESS in 7m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/aff5be9e8e9c48bba4943bde0f235ade : SUCCESS in 10m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/7b5d185d4dfd4ac2bab9647d666e6f49 : SUCCESS in 3m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a52aba1de53e4503a68d132b423ae713 : SUCCESS in 12m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e2d6d562618b4b9da8b98b81973c79f0 : SUCCESS in 6m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3b5e86cbc40548029da1e592f1f4a2fd : SUCCESS in 21m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9a481bda94af43fca5413bf66a4f73db : SUCCESS in 13m 48s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e0f47d34f5b14cbb977820c110af7bf6 : SUCCESS in 7m 12s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/d9d21d2d469347b5a449d9337b5e3813 : SUCCESS in 4m 32s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ea4ad98673b44efeb8e80b846ad22117 : SUCCESS in 22m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/029f35991f2340faa4b2974f60820e6b : SUCCESS in 6m 26s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3afcf402fd97412888f337a9a3c24360 : SUCCESS in 6m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/a5651eb4a16b4b1b8aec238d104ff823 : SUCCESS in 8m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fdceb498bce84a8ab4e53e308da6f060 : SUCCESS in 17m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bc4f3c237af4e02bb6b18a5edf05603 : SUCCESS in 57m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/78d2a159a44a4ad9ae9a22bffe962e35 : SUCCESS in 1h 35m 11s","accounts_in_message":[],"_revision_number":1},{"id":"f6479e3e02213a55343c73dc70d496a12056c57f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 15:59:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"62c98fc8e609a42325cf88b829182174850630fd","revisions":{"62c98fc8e609a42325cf88b829182174850630fd":{"kind":"REWORK","_number":1,"created":"2022-07-12 09:36:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/84/849484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/84/849484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/849484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/849484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/84/849484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/84/849484/1"}}},"commit":{"parents":[{"commit":"041f85b9ef2f496f9fe2ee711ec9a8f51cd50013","subject":"Uncap warlock","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/041f85b9ef2f496f9fe2ee711ec9a8f51cd50013"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:36:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-12 09:36:36.000000000","tz":0},"subject":"update constraint for castellan to new release 4.0.0","message":"update constraint for castellan to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ic2d923471eab11beeb2aa4a4c79081f9dfe01c94\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I2ac4f2d5ff479228fe7fa6fdf9f664e05623d3f2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62c98fc8e609a42325cf88b829182174850630fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/62c98fc8e609a42325cf88b829182174850630fd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
