)]}'
{"id":"openstack%2Frequirements~849656","triplet_id":"openstack%2Frequirements~master~I7c14d0eed1ae047904fd4670c2c286a44fe47824","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I7c14d0eed1ae047904fd4670c2c286a44fe47824","subject":"update constraint for python-glanceclient to new release 4.0.1","status":"MERGED","created":"2022-07-13 10:10:38.000000000","updated":"2022-07-14 15:39:48.000000000","submitted":"2022-07-14 15:38:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849656-new-release","meta_rev_id":"525346950dfc989f315d0726399af4c848bc5dd4","_number":849656,"virtual_id_number":849656,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-14 15:38:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-13 19:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-13 10:46:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-13 19:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 10:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 10:46:21.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-13 11:54:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 19:15:38.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-07-14 09:03:09.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"}],"messages":[{"id":"5e1cdb4b5448c903804179945bb6978552fa5ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-13 10:10:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37d1a7bf058be7eb5162ba60ff9894565bcfc903","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 10:21:41.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f60ae65233934bc6b61c643fe2f050d5\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c8a8f53a021340bb9ed886d11aed0215 : FAILURE in 6m 20s","accounts_in_message":[],"_revision_number":1},{"id":"8a3a4f6ba785bfaf2996fb3d92c623b1ca0d92b7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-13 10:46:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9c1c702336ad389ca5724ecc06a4f8998295b447","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 11:54:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b17c7af104d44f47adc38122db91513f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2085cd832e754226814e82a070747b43 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41a24ce8f8cc42489e5c43536b8acdc1 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9a2cbded4654d5d84213107dab518b3 : SUCCESS in 5m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b194fff9af7549d5876ca6d092eded13 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9ee51bf1130743379a1fd46f464eb8e2 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f675e1cdf0e8417aab12000d4910434e : SUCCESS in 4m 27s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/816166356ca2403eae4a5ad6676dcaf6 : SUCCESS in 4m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c969eb804a454300bde12e2c72cfea09 : FAILURE in 3m 59s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/51debff9b70d45caa568372cf38a8b58 : SUCCESS in 13m 21s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/8a5ed5c7beac411cac5d054605890239 : SUCCESS in 10m 14s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/11a49bf5d0204bcf81fc86a091838efc : SUCCESS in 5m 56s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/445cfa5d1f7a45d2b048a9997bdb314f : TIMED_OUT in 41m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a93239af30f14a11b702eb0f4c19e2c9 : SUCCESS in 12m 56s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8bc3896f2b5d4a69bca42ce7c8059f1d : SUCCESS in 5m 15s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d8f0fd4281af44888cc925d771acd734 : SUCCESS in 6m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/514ddcf75b5845b399838b3e6f59b5cc : SUCCESS in 8m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c892cfeaa8e440338c03f034c441dde2 : FAILURE in 3m 29s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8b6282a98f144c5eada9f7b4d1b476e6 : SUCCESS in 10m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/11fe8f9c14594abfa458a96156bada9b : SUCCESS in 8m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a789590b13af4d8c9c0b5bfb55494afb : SUCCESS in 13m 42s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dbec1372a3b64298bebe8eddd1e8ebfe : SUCCESS in 4m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/490d390e24a349f18fe0854bc76a9825 : SUCCESS in 10m 38s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/acd873d22aa74118a6d7a1f7fc385b45 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2ce0ef897978462d866e27befd54dbcb : SUCCESS in 22m 55s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c71bf22bd96f41b39bfb6cc105ba2cf0 : SUCCESS in 13m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d6b8eeb2989b404b9754572f7b3ddf62 : SUCCESS in 6m 57s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0f87127a1884446a9c4c8ef3cebff245 : SUCCESS in 4m 34s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f43fbbe22f9b40d7abc1d3577916a41e : SUCCESS in 19m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c89499a0698c4be396fd13b93ac05a38 : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f3716a466fd0410eba9b18e3bacc5a0f : SUCCESS in 4m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/705db998e1d3497aac7e7e77efb6b789 : SUCCESS in 7m 38s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/2b32553cd9af48f9868e52a7482d1511 : SUCCESS in 12m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/432379eea17a4b7680d27c99ad29c601 : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0a9ad7bd0e964112be134c63b6540a4c : SUCCESS in 1h 41m 20s","accounts_in_message":[],"_revision_number":1},{"id":"5dede7644c58881b339b5d12622cf9647fc84de8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-13 19:15:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"eec17d11c6ccd00742d6ef8ace6dda0066b6f07a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 19:28:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5de75e7d127e4e8c98abf42a638d1a2e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5912a49fd04a4b1db8af1856af861c7a : FAILURE in 8m 14s","accounts_in_message":[],"_revision_number":1},{"id":"75bbaaa8ad5e9cb72b486d8a61e6010f5723c949","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 20:48:09.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ba7a774d91e48cda270676d3e66a9c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4639b68f920c4bf4b8ef96910e079f54 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee22540f675945aa84f92178c4d26df4 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab91512dce3c4978941cc0fc3031af89 : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f22cd428c02f47e5b64434f0aa15bf32 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29980ed7774d486fa5b9916440bbfbcb : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2c7798fe26a042e49d306294c9cb10e6 : SUCCESS in 4m 57s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cac9e0805a0f465495da5b7fe1cf8e62 : SUCCESS in 4m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2c3fb49b067a44dcad3c71fdb17b6a9c : SUCCESS in 10m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/954390f232aa4ed29cbd0471b3f85daf : SUCCESS in 14m 23s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/78455fa5804145b391571aa0b8f7992b : SUCCESS in 9m 10s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/dd523997bf134548a2b6886a202576f7 : SUCCESS in 5m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3b3cc6efc97b4408b0bd6812567317ac : SUCCESS in 6m 50s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/e224e933f6e14dfd88a5a7396c1f200d : SUCCESS in 11m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6dbe67ed0b2a42d5956a085b3fdc3e79 : SUCCESS in 5m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/10ac79bf14624d499706806045756c92 : SUCCESS in 6m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8252c0eca53d46b8b71a2378d37b7208 : SUCCESS in 8m 49s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/9ef96bdee343471ea4bd796182931b1b : SUCCESS in 9m 48s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/dc0a41dd6a3a46e1b06cce47d4dfaff2 : SUCCESS in 9m 10s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e76068105ed141fd9afa98a372872ecb : SUCCESS in 8m 57s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/359ad57f73b5449da505548d9e8e2e2f : SUCCESS in 19m 04s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4701876f64e44a07ba6480aa1be2dad0 : SUCCESS in 4m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c8ad15ae704a40b588d9f0b4be89a732 : FAILURE in 18m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cf134af1f9ed46de8ac166b35101ff14 : SUCCESS in 6m 33s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/5cf1226d5d1e40329b5703d2de3fbed4 : SUCCESS in 28m 39s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d7510a7c03524e4abed173b0ea4e3331 : SUCCESS in 18m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e1d06db264354ae899aa4f418ce84eb7 : SUCCESS in 6m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/3daf6b51f8e64ce7a55e78240cf6d4c0 : SUCCESS in 4m 21s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a7948568fe5421b8324a15cf4962d71 : SUCCESS in 19m 07s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/435144508c974f5e856e4dd188e616c7 : SUCCESS in 8m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7b6662fd5b1e40d1a949c9df914dde02 : SUCCESS in 5m 59s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fefd766f63d940e2bb9e15ff5748d78f : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/3813d34693e445a58d90f246b7a8f1be : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2f791923fda84e4b90792b0f349247fa : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/174e0f6f78104616b0b86064a22fc22b : SUCCESS in 1h 30m 09s","accounts_in_message":[],"_revision_number":1},{"id":"f9a4d3ed7f81161618e505bf93b9c6ca9f05348d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-14 02:38:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"45009f2dfcff5e4f50b16a412b4baa7a4e89bab6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 02:57:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53e1a016d62c4a18adac74820be3766f\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/224103dd31c1462eafca411b7cc8da4a : FAILURE in 11m 39s","accounts_in_message":[],"_revision_number":1},{"id":"82e244dc69f8371e961e9f73c7c159897e86ada8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 04:20:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeb62e77802e49999ab36e0e80727e3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d0f9ff10a734e0386ccf1522056cdf1 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c596baf8b43447fbe3a118f95e9ce87 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6c3d0dcfde0470ca2dd0594a8e645c3 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f5448ff4ad949e6ae2fe7cb6e8f2108 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c70d3ca9480246ce97a6deeb8c118be6 : SUCCESS in 4m 41s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd68dbb084c2408780b54c5135e5e650 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a3330ac51551434b9ee37c6418747b04 : SUCCESS in 4m 28s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/60d9e213bda04832ac6ca7d479837e8c : SUCCESS in 9m 56s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5284e218ee4b4ad093370b48057f0919 : SUCCESS in 13m 16s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/0428ac3eb2e14ea887e495f2f2c23920 : SUCCESS in 10m 50s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c460daae80d44cffbbee299a54b8c052 : SUCCESS in 6m 16s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2e5214f74f144042aac8c2e55f30bda5 : SUCCESS in 6m 53s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/77e18c91a3af44a49ffe03f66f1a14b6 : SUCCESS in 12m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/dbcbadef570641ca88e171302b000718 : SUCCESS in 5m 41s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b38e0f8527ad45af822d12924c5be37c : SUCCESS in 4m 40s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ba47d5dc657748d0a80200ac0bd61aff : SUCCESS in 12m 33s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bf5ccfd0f53b4cd3b683839e0a8d301f : SUCCESS in 12m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/cf59dabe9ef5419b9f34dcd8599bf56c : SUCCESS in 9m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/69eed78cd4504ac9938adf85831197f4 : SUCCESS in 10m 41s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/266ceb6dea014272be0482e5e70f6201 : SUCCESS in 14m 08s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/0ea669c44ef14daf94e149fe28c37ff4 : SUCCESS in 6m 56s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4c54415b2b34487aa3e5b74dadc4eed3 : SUCCESS in 13m 01s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/b9b8f11002cd43bb81250807fc3a75ea : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6e3dca591d244f928e0d4a7aa58a32ca : SUCCESS in 29m 47s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f9182deb9d464a54aafadb5516bd9718 : SUCCESS in 11m 45s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/82561acf88934f82b72e4fb4bb18f378 : SUCCESS in 6m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/2edf18e972074c06b274b4fdfea8225f : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/cbbe3aad2d104a3186f75a834bbb9a33 : SUCCESS in 20m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e1ba40cca02043af918e6699a2edbd61 : SUCCESS in 6m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/3eed269a3aa94b8ebaca5c4e45556c87 : SUCCESS in 6m 56s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/912ad668bbb34aa5942097f09d3cbfae : SUCCESS in 7m 32s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9cda6468945a4df3a3e977c1f09c3996 : SUCCESS in 11m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65a139b3b1d745a6b13525143c9a5a86 : SUCCESS in 48m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f304aef7209a4537870dbb34736fd6e7 : SUCCESS in 1h 29m 04s","accounts_in_message":[],"_revision_number":1},{"id":"ba7fb257b5919701dfae33c23ac6c16944dccd8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 04:21:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"592a850636f62f9734b27d6c8297096046eab6de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 05:48:29.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a3456f9b3e446f0a3ff308bb0dbd267\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95919db3a86640b386885d84f11d18c2 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9efd6eefaf5405e9452e2f0eb9f6fc2 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fdc0dff7a59c4697acc987bdccfc584f : SUCCESS in 5m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/139e451dc725475e8e47ee19dfff96a8 : SUCCESS in 4m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2623875868b4349bb972bebfa6a1679 : SUCCESS in 4m 19s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c908cfc9481e43f39e4c8dbd30fbadd3 : SUCCESS in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/e2588b15d1e24ed4931a533a9cedbc09 : SUCCESS in 9m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d714687c24004d4a8cdffa2182582fa6 : SUCCESS in 12m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b4f7556f73bb4c55a083494179a54787 : SUCCESS in 11m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/8bc98c934695454d82246873f16119bb : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ce4c0e78c82447cba3d426f5b0e805cb : SUCCESS in 5m 46s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/248efa48d7af457cae6d7eb30a5d22eb : SUCCESS in 8m 04s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/116d56c8f0ee4239b4153704027b3428 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/819c072679ca45209af75517a6e0e860 : SUCCESS in 8m 55s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/f3a7f4ed44e04f0480889116b50fe974 : SUCCESS in 8m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f560454b21b1444d8d79e958a1b0054b : SUCCESS in 10m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/08e2d93a9b734ed795ab75f8c770b38c : SUCCESS in 4m 34s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/5c0830fde6fd4d838bba26940c7ceabf : SUCCESS in 13m 00s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/79d65575d056461aafa90adb96b66e05 : SUCCESS in 5m 40s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/43174bdbdb0844c9b40660e9553e33e7 : SUCCESS in 22m 24s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b24fb880619a4700ad5ce92827e0d029 : SUCCESS in 11m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3926d7f88f534952a622942078372165 : SUCCESS in 6m 26s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/a1669cc94d1e4e09971eccb239846122 : SUCCESS in 4m 03s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/26cc1e02a5cf43d4b256c940959f8613 : SUCCESS in 18m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7bd5140efb1644248a72a584f33c6351 : SUCCESS in 6m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/40c91dba9c0e4f299a6847f7a344631e : SUCCESS in 5m 16s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ff654d48f176488dbe3dd79f3fc6d90c : SUCCESS in 7m 27s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b27f317ccaf54ce188dc4751aecbbff1 : SUCCESS in 9m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b050d5b348c49c99a966ba479a418c9 : SUCCESS in 52m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/374172f4c23f4f53a3cefc8fae25ac4f : FAILURE in 1h 25m 58s","accounts_in_message":[],"_revision_number":1},{"id":"5fc96b400300155acfe672c1a1e25f3f178826a6","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-14 09:03:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7b8668a965a24e41d33a77dbbfd63024060091b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 09:15:02.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf7b6bb214b64451a2a3e6251f8b5df3\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/2b6c15a0c8b748f2aa834cd58ecf36d6 : FAILURE in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"02d992197d3660a28b8446577cb608b007841b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 10:35:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f8dac35ecd349bd8d4088e2dd5ba17c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20d2ea2a185e4f89b67dfc6ec8937f8d : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3504bd4a9a3b40b98f501ac08c454983 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23b5f1afbcad4dfea8ec4cc01ab9cef8 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ce26a52136c4b71a25246dce9baddab : SUCCESS in 5m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/152d6e0c03d541ad8e16aaf3ee0644bb : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a69dd45ff3844eeeaca16ff09cd83d85 : SUCCESS in 4m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/aaad34cd4b8a46e8b2cb94e9687bbda3 : SUCCESS in 4m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/44e249ff9cb94f13943cbd38612cf738 : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7e47ae081ddc4818ab9be2f041e37b51 : SUCCESS in 14m 37s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e9ba490bc4f84b3888b05b5eb4bfcf94 : SUCCESS in 11m 07s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/db3978e398e7411ca0db071136e87dd3 : SUCCESS in 6m 10s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/46c8231c6d7649d9917ef9a25c123c26 : SUCCESS in 6m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9ddc8828760945958a6321d9fc66d261 : SUCCESS in 13m 09s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/833f10c891684d658ebf6333eb6b014b : SUCCESS in 5m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/18f5adcebf3a4ba29090faca4f7e311c : SUCCESS in 4m 53s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/aac45dd5f96447fdacbd2858803fd227 : SUCCESS in 9m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c953490695e840c59f9ebf9a07c3be72 : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/29203d377f3f4340ab8c978f5a038bcc : SUCCESS in 11m 36s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cf50e992737f4d328c7865537dc3a2b1 : SUCCESS in 7m 00s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7ab9945e44bd4f6193d18bc82bd075a5 : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9c998188e5a54c9b879e81c9a5e19ba1 : SUCCESS in 7m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a6022f95e1c84f1196464b642296899e : SUCCESS in 7m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e8e933dad9fb458ca08545418ac4220f : SUCCESS in 8m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7e32ca8b17e74bce9df01bf35eb34051 : SUCCESS in 20m 54s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/45cc881f54b64edd84bf70fa8f6bf529 : SUCCESS in 10m 13s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/361077d08d044d85a46dd48194b7b97d : SUCCESS in 5m 43s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9b3d537923164a1f8c772d796fa9a1b8 : SUCCESS in 4m 31s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e16b2a3415f64a73ab05f72c76bcbd69 : SUCCESS in 21m 26s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3a8302b7a6fc40a0aac8b4051c329021 : SUCCESS in 6m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/dd1512baaa21451d9eedc50344755c00 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/38bf38437ece4f8e9674eb128da18beb : SUCCESS in 10m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c2f48a73f55a4dc4bfae5ac1074b9453 : SUCCESS in 13m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a0945d7caea472d97dadb9a0f24c2ce : SUCCESS in 52m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f723745e47446bc96887884cc259cd0 : SUCCESS in 1h 30m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e7a4d4ed5f53dd35aae1abca02563921a8b2877c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 10:36:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7d7551bb001726f23fc868e0d702ed45d28cce1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 15:38:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2e6756763dd4190b008bd9a76c708a4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92c242fe4f1e4d14a7479a2e116a34fb : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfba4caa313a4faa8a018ed721d0ba67 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/606f1d8ec0c9481c99766f9ad11bf12a : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/42798561aace44dab0a754dad6d30daf : SUCCESS in 3m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2876ea5badce44a5881646061a453441 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/471957c8f36f4880ab09505654663f7b : SUCCESS in 3m 01s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/be68f9cad56d46e0aa103ba0cc90f9d9 : SUCCESS in 9m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4aac95770701445bb3c473e95d2e5601 : SUCCESS in 12m 27s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/8a92bd6f8a8a4dcc93f0b57402ebecef : SUCCESS in 12m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/35c4aff8a49a4e74a95261a4eea5d07e : SUCCESS in 3m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d06ae4f2a11b4d82b7d460e71779e824 : SUCCESS in 6m 18s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a1625b845ea74d0eacbc96a519565ab9 : SUCCESS in 7m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/701b78e0f11e479bac2009c2a4cae8d3 : SUCCESS in 8m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/22b9acebc1a649ebbe3d7d6282b71c82 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d1cea523ae594f9d96bc2116b5ce6338 : SUCCESS in 8m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ef36e5d707d44c6e962874b8323c848b : SUCCESS in 10m 09s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/226ef7d4250e47a6b97148aeaff895e2 : SUCCESS in 4m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3e3f752aaa6c4462b225d0c640e29644 : SUCCESS in 9m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/067b7bc950e5464c97c3a09f98b0fb72 : SUCCESS in 5m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/82cb7ad4470f4c5a8c74aba0b4658edf : SUCCESS in 26m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3ed6ece0dfcf429f9a0da08b9d6cbe5b : SUCCESS in 10m 47s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b912a0e65fc3436a92fc3460ead2d669 : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/202c246e5ef64d22b3d7f3f5abf57960 : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1b698dc6292e4dc88d817bc3ea59d6b7 : SUCCESS in 18m 38s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/74f19188a3f04736a6915b967c439326 : SUCCESS in 6m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0df85ff34dd14e12b2cb311bb3f41cf9 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ecc165a09e214b1fa75387b0f0344d7b : SUCCESS in 7m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/69f4efc765f542f78d38ba578aafd196 : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ffd36808313f49fd97ac36e4cb491f49 : SUCCESS in 41m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fed1d244cef40c7a03fd31fe42cf120 : SUCCESS in 1h 49m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fa3182d1005ccc249e97c7bd585734bb7a8d1f6d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 15:38:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"525346950dfc989f315d0726399af4c848bc5dd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-14 15:39:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8e9bcc199f848009eca3454f8335ca0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db145bc9ceb410aa4f744d31a95ad60 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fd2c85ec96df2cb961407576c9716e127caa444c","revisions":{"fd2c85ec96df2cb961407576c9716e127caa444c":{"kind":"REWORK","_number":1,"created":"2022-07-13 10:10:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/56/849656/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/56/849656/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/849656/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/849656/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/56/849656/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/56/849656/1"}}},"commit":{"parents":[{"commit":"c542de223ef6893f910017dbada9416d18b94366","subject":"Merge \"update constraint for oslo.config to new release 9.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c542de223ef6893f910017dbada9416d18b94366"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-13 10:10:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-13 10:10:37.000000000","tz":0},"subject":"update constraint for python-glanceclient to new release 4.0.1","message":"update constraint for python-glanceclient to new release 4.0.1\n\nmeta: version: 4.0.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Abhishek Kekane \u003cakekane@redhat.com\u003e\nmeta: release:Change-Id: I080f774c770e5d3a50c37e58dcc939ad16e1f495\nmeta: release:Code-Review+1: Pranali Deore \u003cpdeore@redhat.com\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I7c14d0eed1ae047904fd4670c2c286a44fe47824\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd2c85ec96df2cb961407576c9716e127caa444c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fd2c85ec96df2cb961407576c9716e127caa444c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
