)]}'
{"id":"openstack%2Frequirements~849664","triplet_id":"openstack%2Frequirements~master~I4a08fc0b651b5da8197415a942fab215bb6ba5a1","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I4a08fc0b651b5da8197415a942fab215bb6ba5a1","subject":"update constraint for python-octaviaclient to new release 3.0.0","status":"MERGED","created":"2022-07-13 11:01:11.000000000","updated":"2022-07-13 18:28:54.000000000","submitted":"2022-07-13 18:27:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849664-new-release","meta_rev_id":"c64c58cb7cab4336cc113949f762017b1d1276a8","_number":849664,"virtual_id_number":849664,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-13 18:27:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-13 16:46:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-13 11:51:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-13 16:46:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-13 11:13:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-13 11:51:04.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-13 12:42:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-13 16:46:28.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6b379fb143cc2b7fdcc5793e1f065aedf40d88f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-13 11:01:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52d4478234f17f239939a35ae23580f463c66603","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 11:13:27.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8490f545d4442218a7f68646b627857\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/c6d1351785cc4ee0bf78a61744b6dd02 : FAILURE in 10m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d5d583e4a006e1a15a9a63cc94de6604060d1a3b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-13 11:51:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8567d5c53a47b259dd2cc5f0705b97339eda7349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 12:42:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f4a3c7465ca43b7ab006988cbb6c9ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d55279fc3d04befadf2cef692c42bc9 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88bfa8cd033741699f7e0000c6462fc0 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ace79d9247054b94a23222e7b18fa108 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a9a10971c2e4621812b531f6c6ffd02 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3377e1a2978f498ebe181d6953b57804 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a82568f996f242cea21708ec4a73965d : SUCCESS in 4m 03s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/4449c4fa7e204d71be38c3083e02712a : SUCCESS in 3m 59s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8c60de5058a2433ab92c584bd17fed37 : SUCCESS in 10m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/16bac9b819174d0598ec6ee83d87c5c2 : SUCCESS in 14m 42s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/9e76783c05d345e4ae310b3d9f4cd33e : SUCCESS in 8m 38s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/8418fa2af4db4d4d81fccd12f4b02f28 : SUCCESS in 6m 00s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3fae352933444aad8056993c2d6c2950 : SUCCESS in 6m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1c2edbeeae6f46af8dd3ef8ee947c33c : SUCCESS in 13m 23s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/fecf07fb8d3e4764b532840f721e3844 : SUCCESS in 7m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ac5d9267a10e4295ae680034969bb0cb : SUCCESS in 6m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1581a4de12c84dc3b66df4fb7eb0090f : SUCCESS in 8m 37s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6040ee5d8e144c74a92b82fcd1787bf2 : SUCCESS in 11m 10s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e00fcf83d49a4e448a3f915104bb024d : SUCCESS in 9m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/813b9de2b69d4394a0f82b03cad283e6 : SUCCESS in 9m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/788eee0a1bd943769eafa7f0222383a6 : SUCCESS in 12m 20s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1a98b36f6d4e4a7fb1c72397063fb1cf : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3a328b04ace14e18afa56366d20133ac : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/03f43a64cc964ffbbe11521efbce43d2 : SUCCESS in 5m 31s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6f86a7c75ab846e8a2682bd8622395a1 : SUCCESS in 23m 57s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/030bb689e38c432da0aeef14ff4298d9 : SUCCESS in 13m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/90296f3860444782820443ea71cacb0b : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/97e98bcda6ba4dc49f862e6baf6f7917 : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/6fadd2ef20a24bfc881f14b639dd2dfd : SUCCESS in 20m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/c4549cec06584a6aa850c690f2d3dcce : SUCCESS in 6m 03s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7ef59af86625480384b5b0c61f9d6086 : SUCCESS in 5m 11s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/057a588f329b4f7c970f8c4d5f9763ed : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a9d86ccca15d486a9bedcd4bc4ef3d6a : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98a85aa311c14008ba1a9a0c341e5686 : SUCCESS in 53m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dab0c1ca14be417aa7663f18b8cc8a72 : SUCCESS in 1h 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"df6284c2b341556c7efbb74f437824e011919fb8","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-13 16:46:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3f2ea1339f903c54f3ab4fe48d7c739b2dd4b209","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 16:47:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc6f3d0746da6188e567906748f35e0934bcbda0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:27:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/363b481a208249ac95ac5a3850fa9168\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ac1e49dde7e42a2b44c9c990cc8e822 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/462ce34b53bd47e681bff4965eade881 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/83dbf154186e4c4eaa2e5c0ac2d64762 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67c60fc20b864f409878ae13e11ba045 : SUCCESS in 4m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/161cd162223043aa86b1456893428a54 : SUCCESS in 3m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5aa0f6a96919455588689a3119621fbe : SUCCESS in 4m 19s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/34118ce894a449bfa8fb44c304297323 : SUCCESS in 11m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/52b5a939813846caac80b8545ee9fbaa : SUCCESS in 12m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/aec6a17325884936981d98098d7a054f : SUCCESS in 12m 38s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ef1ecfed6ee847e099f3df30853d31cf : SUCCESS in 4m 24s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/9d76ff521b2b44d2a0949b3fe63b66d7 : SUCCESS in 7m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c3b49c036deb4ed1a99da72ffddd370a : SUCCESS in 9m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0f040efbd13e4ad88107999a976d3a9a : SUCCESS in 10m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1c37b2ffb98a4966a22417987f50274d : SUCCESS in 9m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca3b50751a344bf6976c4ff906f4beea : SUCCESS in 9m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6f4e9119ed894ff0a9ebcdb82fd090fd : SUCCESS in 11m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b00a656f98814afcb853247c486fe398 : SUCCESS in 6m 48s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/812d38b26ce24d5a80dc2718972a63b4 : SUCCESS in 9m 35s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/83ad3669d8914dc599e1eae4d65fb0b0 : SUCCESS in 4m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/e95f218f372c409dade769bf69b045b6 : SUCCESS in 24m 17s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/6b23c8fa438e4fdcb691d815f359099c : SUCCESS in 11m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e379dcdce11e4f12a21d06bdc5584464 : SUCCESS in 8m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/973f9010a6cd4bc7b6dbcc955c535021 : SUCCESS in 4m 28s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/9efa410581e1483eaeefa84e0026bf00 : SUCCESS in 19m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f687a8140a0548f9b8aee86d958badfe : SUCCESS in 8m 14s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d46fbf9c9e4b4bf9b0efd54cde67b310 : SUCCESS in 6m 19s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/bfb63c73649d41d4b2524dadb2fe7d93 : SUCCESS in 6m 46s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/34b6ca49c28f411dbd70e6b7b83522b4 : SUCCESS in 14m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2a470240a3249c091e10abeffc6355f : SUCCESS in 54m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8d3c59915ed4f1cae2de197960762fb : SUCCESS in 1h 27m 03s","accounts_in_message":[],"_revision_number":1},{"id":"75d86a5938a9c594a24152e808ffbbeb531a9f59","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:27:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c64c58cb7cab4336cc113949f762017b1d1276a8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:28:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4fbb949948b24eac915a69757eb41796\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ac3e0b3ab274ca3a29327cff3f9ff6e : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f97f4d83861320987e9d209ef919813b92a90976","revisions":{"f97f4d83861320987e9d209ef919813b92a90976":{"kind":"REWORK","_number":1,"created":"2022-07-13 11:01:11.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/64/849664/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/64/849664/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/849664/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/849664/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/64/849664/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/64/849664/1"}}},"commit":{"parents":[{"commit":"c542de223ef6893f910017dbada9416d18b94366","subject":"Merge \"update constraint for oslo.config to new release 9.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c542de223ef6893f910017dbada9416d18b94366"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-13 11:01:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-13 11:01:10.000000000","tz":0},"subject":"update constraint for python-octaviaclient to new release 3.0.0","message":"update constraint for python-octaviaclient to new release 3.0.0\n\nmeta: version: 3.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia859546c4e9b685ef3dd6ffbbab7e182deefe1d5\nmeta: release:Code-Review+1: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: I4a08fc0b651b5da8197415a942fab215bb6ba5a1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f97f4d83861320987e9d209ef919813b92a90976"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/f97f4d83861320987e9d209ef919813b92a90976"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
