)]}'
{"id":"openstack%2Frequirements~849952","triplet_id":"openstack%2Frequirements~master~If923c744369ae070bfbd24cb448c8f4fa8912845","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If923c744369ae070bfbd24cb448c8f4fa8912845","subject":"update constraint for python-designateclient to new release 5.0.0","status":"MERGED","created":"2022-07-15 08:15:51.000000000","updated":"2022-07-15 23:24:04.000000000","submitted":"2022-07-15 23:23:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849952-new-release","meta_rev_id":"a9ab352c193221b1cfc030430da763798c903a26","_number":849952,"virtual_id_number":849952,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 23:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 21:28:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 08:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 21:28:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 08:27:44.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 08:32:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 09:27:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 18:30:14.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-15 21:28:50.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"a4361ca31394faf0726ac9729fa99fc8567c479f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 08:15:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"08863e6866a37561a95c6508608b6e0f06eca895","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 08:27:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"830b6106f01a64f291f9a7665e15c01e608841bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:32:16.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/263fe5b4cacc401f9324d4d6c26c9c9e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/644d9f1c0059426db03d82ff60431578 : FAILURE in 14m 33s","accounts_in_message":[],"_revision_number":1},{"id":"d6aafb652f548dac6859b11210056a78b500ad7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:27:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/44bcfa6e34ed4189ab4a218afe421afe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8adca446913d45d5b0334a18c6a4558e : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d59cde1d1cbc4eb38b01dedb90a659a0 : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b77f918b32e44ef58ddf0cf7c5ef5420 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0728a0fb635546c2900cd285970910dd : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/49ceadf9a49e45c5b95dc863c584d76b : SUCCESS in 4m 38s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c942e9cdace24f009321036854fe1300 : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a32b8b6d124c4aeb9251889778ab10bb : SUCCESS in 4m 29s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bbf9e899c4a545a6a59c52d36aebfe06 : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/a5a174530175490b9d0bb04ce707bded : SUCCESS in 14m 02s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/51c96d4b24b5467ab7645d6885bffd81 : SUCCESS in 9m 35s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/353b3b8560844da98660b76ef22acde3 : SUCCESS in 6m 21s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/15e2c2886f2046b58cdfc0f671fe6b2b : TIMED_OUT in 41m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9e2994c85455477180fcfdd7f95267f0 : SUCCESS in 11m 45s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9d83b9e766a24d8ebf2051b86bd01ef4 : SUCCESS in 5m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/32e5a042da5d4df7ba3f7c4793ebc9d7 : SUCCESS in 6m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a38a9532d00341ca83d4012c0dc2a98d : SUCCESS in 8m 42s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3366d3e5fc494f2181921bb83669da7a : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/38d2a5b4446c4404821d17ecc09a7978 : SUCCESS in 10m 08s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4b5d56f9ae844e14a89fdfe01e2debdb : SUCCESS in 9m 11s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d4e67c231a0d40369ef92470f8f44f4e : SUCCESS in 14m 38s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/4c88a5b38d634d8e89313567529f6bbd : SUCCESS in 4m 53s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c349859de4b94ff2b1dfe9337814ce93 : SUCCESS in 12m 49s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0dbbd1ca77e74472860227281a5fd09d : SUCCESS in 6m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/89227641f05d4594b24e4c825d1d71ad : FAILURE in 4m 28s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/62f91582a8c94c0f98bdaa79d0c05654 : SUCCESS in 13m 29s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/cabd97dae7494d29bc659beaea864105 : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/7fb31db74d7f4eeb87f228852fb0d264 : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fc7a53954cf644f89abbcc4cf03185c1 : SUCCESS in 21m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a5b23c5088544b17a7e9869612f562dc : SUCCESS in 8m 39s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/edae29252a354645a35cb33812290c50 : SUCCESS in 5m 55s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/eccbfeb873a64481abdd2ff14bcd54b3 : SUCCESS in 8m 13s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/336bb75afa6749538566a1e5dc388b7d : SUCCESS in 14m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38188594befa42e29c276ce4f6eb80d1 : SUCCESS in 53m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d428597ac8454677a241f1e3eff50f6a : FAILURE in 1h 09m 57s","accounts_in_message":[],"_revision_number":1},{"id":"c2f93979130f862aa3e91b473e1b559ead8b885d","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:30:14.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c13898a094a560a774bc751ec2ffbdca8ef66070","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:50:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b4e414de140c4a4994b24f5afae89ee4\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6f6805c6d9ea4e0f9af1f48090f5dca9 : FAILURE in 8m 41s","accounts_in_message":[],"_revision_number":1},{"id":"444470e44e81a5662e5502cb1bc3564539d76594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:04:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87949a48666443b293a0308ec47df79a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c2bdff0ca74bd5bec0af7f1cd25f22 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8484aa8fa43b4e22adf1a76dd39afbdc : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/608d36e2fc44420c957de7cfc1b6d3e5 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecefe35f6d0e4b9b8a6d802fc80c6a53 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/56567ea9cd7f4859a669c81fb0cf76fc : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4ad440fc13254cf5a309dda33a66db88 : SUCCESS in 6m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c75a6562400a48db95b8f6824b4ae286 : SUCCESS in 4m 27s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/2e0f87976413407492c3571f3df94b83 : SUCCESS in 11m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/438572e363bb4181b84f86902dbb8bdd : SUCCESS in 14m 13s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/cb7a2ba10a164b0db16dd53d63571120 : SUCCESS in 9m 41s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/83a921e365af4ea29fb95d05338444a4 : SUCCESS in 6m 42s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/5bbfe797dbfc4c3e8b619162e6facc3a : SUCCESS in 7m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/3767c218fd364b89bbffcd977b297237 : SUCCESS in 11m 47s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/14d6b83eab994be88dff9f44c5f556d2 : SUCCESS in 5m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ae64bb031bf843afb84337e149a8622f : SUCCESS in 6m 05s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f6c5b023b9174688b81cd3d570ea01c4 : SUCCESS in 8m 36s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4248ef4b7e154ab99e91ddbfa5f83de7 : SUCCESS in 10m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3658cf07afbb42c49236f0b32674888f : SUCCESS in 9m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e39f4311e2f14c32a268a4f5a4034e43 : SUCCESS in 10m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/49be80e90c0b496e83ae66f8c801aad9 : SUCCESS in 15m 59s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b0659bf8f1904164bb18afd1d447f0b1 : SUCCESS in 8m 08s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/526ff30fe07743d1a7fd47304f26b27e : SUCCESS in 12m 39s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d8016bd542e942578a1b8634d0b37b47 : SUCCESS in 6m 35s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/0ca9dfbe6b0f44cbbb52fe9d73533d5d : SUCCESS in 24m 11s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/553dc80f63784068936832ff667a1608 : SUCCESS in 14m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e7161536111e472f9e2a06cd1fe6b06d : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/84e8613cd7b84a48881428cb39ebeb1d : SUCCESS in 5m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3dccfe5c962b47d3b6acb2eea41fb750 : SUCCESS in 21m 54s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0cd495f569624ed8b9888c6cfc214af9 : SUCCESS in 8m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5498a46f7fd14f2c81bcaceff0279330 : SUCCESS in 8m 12s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e1cb16802c84758a4e15336fea94437 : SUCCESS in 7m 47s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/12144c4d0b004de3894fda15aa29d731 : SUCCESS in 13m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c192528fe9e4476493a0344ba3d2bee7 : SUCCESS in 54m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b139da06e0fb469abb4fd971bb857ac7 : SUCCESS in 1h 29m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5c4c31dd3e7a8cea3f1f530d4b0158791b1e30bf","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 21:28:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c3342f54a038bdfc478b2e034d7aa1b9f64c1d5d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 21:30:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4ec27b75b48015f049b4fc5f2af61f485f7b1ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 21:44:39.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4beb0d3b3124132b90c0254a5e867d6\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e6521ad854f5473d88fa8689253f7b5d : FAILURE in 12m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1d2fb22b3c4898ca747a0c328a0cf7a9dcce3272","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:17:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12bed13731a64ddbb0beaff3f10ee40d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc44b176b30d4d8c99946ecf95a55446 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a94a369aee4030b78e9530afce651b : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4010d7a1ae904424bb29ad9f82cd5dbd : SUCCESS in 7m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb060ed41b374b97a85deb8dda9ab1bb : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f8a73e0f68b47a8893a4c934a98525f : SUCCESS in 7m 00s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2dc9652025804c708f935df06bd4b8b2 : SUCCESS in 4m 48s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/397495b4ba0241e5b7e2c741129bd49a : SUCCESS in 5m 42s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0e5e248b8c7b4dc6b8c89f86292d5c91 : SUCCESS in 7m 21s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e1a444b7f8be41e0b49150924131d48f : SUCCESS in 13m 55s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/2df6d60af8ec401c957c7a3b593ea806 : SUCCESS in 9m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d07035757ac7449ab7133572a9ab5d8a : SUCCESS in 5m 46s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/74b4dc7c6ff64b92b30968f7ef3b757f : SUCCESS in 9m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a458e7809fbd4f3bb82ec8a4d7c21063 : SUCCESS in 14m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/23fe92d57539484f9a8299fbb1d4b57b : SUCCESS in 8m 05s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7ba1b484f8dc425484722a46cf5a8106 : SUCCESS in 6m 29s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d1215ad1b5ef4d60b588f5b405ae4dda : SUCCESS in 11m 15s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/7ac6f8d4347f44d3b873ab99abdda456 : SUCCESS in 9m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6390905907e7439c85aa303de30a6264 : SUCCESS in 10m 49s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2d616b487496451190d9410e1806450c : SUCCESS in 9m 29s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/f303da96bf22406488a4842345c75ec2 : SUCCESS in 15m 15s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2c1d7d74b27148dc93740756b84e13b7 : SUCCESS in 3m 47s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8a690b61d4f34cde8f5bf65d4d9dab14 : SUCCESS in 8m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/0fb648117c5c4d8c9295c03538388600 : SUCCESS in 5m 54s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/12b7cfc45cfd4c09bc6f60607f35d5d3 : SUCCESS in 23m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9c2baa4d6c1344c1ad8dcf7323353a2d : SUCCESS in 12m 42s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/78a840e8d4574719b2592f7ac195aa5f : SUCCESS in 4m 55s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6ed16cdb53140d682ad4b367cf1b50e : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/dce33301eb2a4ffdac7e8d4391a23d61 : SUCCESS in 21m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/81d85d6367d24e95b57fc39fe315e060 : SUCCESS in 9m 37s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4bf405b9cab6479a9718cb883893ade5 : SUCCESS in 7m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/fe33571e63a44d389533fab1a33741bc : SUCCESS in 10m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/62bd01f672a64e498db256bd98cc18c9 : SUCCESS in 11m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5adb1f0e0734a56a540b0034e6ef386 : SUCCESS in 48m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5acd327b08454311a8004d660f20824d : SUCCESS in 1h 42m 01s","accounts_in_message":[],"_revision_number":1},{"id":"76ff3404770ef0395868dbd544dd3dee2d62f1d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:23:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c6a86cb71d849bfb42e356137b9fc8f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be65b01d177048bc8acf2baf89641e7d : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1424645fdfd4f51943f99524c53ec1d : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0503300913124fd09b4e8e08cfe9c7cb : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b716261cebd43bc961023d7eda4ab18 : SUCCESS in 6m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc38932999d247e6a340205f6d22590f : SUCCESS in 5m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/79c91b4b0fa14c159652ffea5f0c4610 : SUCCESS in 5m 13s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/67d28304376b4563b3b35d455cd2f24e : SUCCESS in 10m 10s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4225cc5fd0754a5fa6ea39b9124c6087 : SUCCESS in 13m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/26d266bdf6ba4b7292efec717540ed3e : SUCCESS in 10m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/9e54c3a5b1e54478b457cc0321c0ddb2 : SUCCESS in 6m 43s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/054ba9c1c43a465ab9a58e19a12ba6e1 : SUCCESS in 7m 21s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b5743003403a4fa3a51f416f1e82a601 : SUCCESS in 9m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3472cd4e1545447abbeb2b20d9b898b1 : SUCCESS in 9m 44s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/19a619f731de4b058f712bd40914fb17 : SUCCESS in 9m 48s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/53543519b3314532b712869b2ca8e372 : SUCCESS in 11m 17s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/1bb1d5aa3f204dc2a3b57f08a6a5f5f7 : SUCCESS in 12m 41s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/e85e720675ce40b18d22c4860b947dc9 : SUCCESS in 5m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3a10968ff97c4aedb410ef9ac17002e7 : SUCCESS in 10m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/641ae349d9b34878becd4ce8be5937e9 : SUCCESS in 6m 12s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/3aa9161669294c46871693d5ba921253 : SUCCESS in 25m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d17c140bbfba45098b8cb3be614bcc85 : SUCCESS in 12m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/67fd94768c17404594e1a74c7457a860 : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09e20b7ead0e4603b79a1dfbc2587c74 : SUCCESS in 5m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/52ea2852275c407a8d46cbb392c2fc90 : SUCCESS in 21m 00s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/03ed29a0930d4ea9bfb76f3e59407ed8 : SUCCESS in 6m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/632b7f536e4e449685201b394cc82152 : SUCCESS in 5m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/35b4efd7916440da9ca74873f9c92bff : SUCCESS in 8m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c3baf9dce16340b88e9b4c6f970ed9cb : SUCCESS in 10m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8114f627e99c43e4928f7b5902b9b250 : SUCCESS in 57m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dfe2f9a602049d181586de32a79f75b : SUCCESS in 1h 34m 53s","accounts_in_message":[],"_revision_number":1},{"id":"804c7f0b755d714dc42f538f6ccb68f4f4da937f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:23:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a9ab352c193221b1cfc030430da763798c903a26","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:24:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d0d69a0a70e437ba9f54afe1f9ce517\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/773cb3050b6541219378b6fea9b12a4e : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"42ad160a6e2943a0bc759197846f79c9e44baf37","revisions":{"42ad160a6e2943a0bc759197846f79c9e44baf37":{"kind":"REWORK","_number":1,"created":"2022-07-15 08:15:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/52/849952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/52/849952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/849952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/849952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/52/849952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/52/849952/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:15:48.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:15:48.000000000","tz":0},"subject":"update constraint for python-designateclient to new release 5.0.0","message":"update constraint for python-designateclient to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Change-Id: I13e5603f23d2a8db84c5a529a12a2fb670719123\nmeta: release:Code-Review+1: Michael Johnson \u003cjohnsomor@gmail.com\u003e\nmeta: release:Code-Review+1: Dr. Jens Harbott \u003cfrickler@offenerstapel.de\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: If923c744369ae070bfbd24cb448c8f4fa8912845\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42ad160a6e2943a0bc759197846f79c9e44baf37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/42ad160a6e2943a0bc759197846f79c9e44baf37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
