)]}'
{"id":"openstack%2Frequirements~849953","triplet_id":"openstack%2Frequirements~master~I5c1e24a0dc425058650ef7a4a6465bf07ddee51d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I5c1e24a0dc425058650ef7a4a6465bf07ddee51d","subject":"update constraint for python-cyborgclient to new release 2.0.0","status":"MERGED","created":"2022-07-15 08:17:42.000000000","updated":"2022-07-15 23:08:09.000000000","submitted":"2022-07-15 23:07:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849953-new-release","meta_rev_id":"78bef2970a4f238f9cb4d72071296526333d8039","_number":849953,"virtual_id_number":849953,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 23:07:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 18:22:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 08:28:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 18:22:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 08:28:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 08:31:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 09:34:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 18:22:46.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"e68838e5fe13740364c7580d8ac860be8b07c746","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 08:17:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d330952d94af03e24c4bcba12f18ac8c32d8bdd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 08:28:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2bd905f2e6d0078d369151c0198be0cd103a57d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:31:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/95a613b8f3314fd3b111908f60d22d0a\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/010f4eb09be242389ba0b6eeb8f5661b : FAILURE in 7m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b350b4851aa0e27d61807afd40bd035713175900","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:34:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc6c75ca63ef44af9ad2e5b65837ee10\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a92cdc751dc544228f7d19a97356198c : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8919ae98a10446a95a310fa920ce84e : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8d7c43ed96f400ca240981a8da57164 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad5e198ef7104e3e989ab04058329d3e : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/148c641898a444f39a14c694594b6252 : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d509229ccfee47718a6025c72e65bb97 : SUCCESS in 4m 43s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/b0d2b71b96f64bf0af809ee99210ef61 : SUCCESS in 4m 18s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/63e1dac795de484fab3ef84911ee150e : SUCCESS in 9m 57s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/0dc41df585cc460f9f019e6f00dc653f : SUCCESS in 13m 28s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/2a50b86bdd6b4b918c526e7eb59d283f : SUCCESS in 8m 03s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/37ae62b8512040e994790165646fee1b : SUCCESS in 8m 47s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/140d988ec7934ec99d5c8383ad0cf7bb : SUCCESS in 6m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/9ad84e412a8148d9a92e21ffde08185d : SUCCESS in 11m 28s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/526995f58f0140839557b985d85f3c87 : SUCCESS in 5m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e02ce12386124e06a986f1d82afbd5d1 : SUCCESS in 7m 47s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1d465347644c4494a27f8e119266ac82 : SUCCESS in 10m 06s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/091fb2d885e24753b9e02990d61c3002 : SUCCESS in 11m 02s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/af8dfa523fc4403d87c488a3aec41a8d : SUCCESS in 10m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2f70b4552d894845bb92490a63fa6bc4 : SUCCESS in 8m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/7735b399b1dc4798b10ea18a055a3738 : SUCCESS in 11m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6ac4b8737ac9435ba85c414454c8f5ef : SUCCESS in 4m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b4fc87f8571c408abd1e3f22e2038be3 : SUCCESS in 9m 50s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/54fc0b7736db46ec94e50e8fd0cc47d5 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/8c97584231494e14a53d5101d8d76007 : SUCCESS in 22m 12s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8e25221b44574c9e954b0da98740913e : SUCCESS in 13m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6982fb4ff4aa4092ae877a259bab7fb9 : SUCCESS in 6m 08s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/09824e249b3d4cd5a920a92297d38c95 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a3a7d1e66048409ca6cbc71ba131e24a : SUCCESS in 20m 45s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/27fe69e2c3e742a0a09e9ae09f76fa74 : SUCCESS in 8m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/edce5a8fbd2a4ff195f6d00382da3cd0 : SUCCESS in 5m 27s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/49c4cc5fdcf5456e98afa610f4720513 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/272c34327fdc40c6bc68a6b66111e326 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41f4bb267b8c42f9bc303c4a3cb7dd91 : SUCCESS in 53m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6c8534673cd741b2b85e427a86c82faf : SUCCESS in 1h 13m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7892a5b0489a7ab4e21b94fbb37883cecffb32fb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:22:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"651dddf796ef9eccdc8f4e6a115e3b9824a3403f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:27:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b9a5ea7f6ae44e7d24d33bc7e302e225ad6c7124","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:07:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e6bc3aa3b4a4c53b67553f3b416878d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04489b714aad4213b352e9362090ddc8 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04bd24c450b44f3ead4d884cf349913b : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6aeb46f4c0424c4da44f6c06762ca1bb : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa13133ef002416ebf8c3f94a2f6b4af : SUCCESS in 5m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/108f87a876ae4052aaa98b8b5d1009b5 : SUCCESS in 4m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/a651a5bf009d4300ae93f5b80006d4ac : SUCCESS in 3m 23s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bba359f74b72454cb6cbe144cdff74c1 : SUCCESS in 9m 55s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5c30b40d2bc949c184bfff8125bda926 : SUCCESS in 13m 15s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/96cbe59ebbd5429b9bf98ef1d44db2bb : SUCCESS in 13m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a7564d58bda24f6bb615d7371774920c : SUCCESS in 5m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/01ba9af5998b456c85e69d59b9ca4c8e : SUCCESS in 6m 35s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ae9e697be67e40c78cfb3957a8f55e9f : SUCCESS in 7m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c0131c494c584d338ff4a36f7c0ded07 : SUCCESS in 11m 58s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/78f5143c1cf54967be222ad5f09d95e2 : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/92550c355a0e4f79a14ee0239e8749c7 : SUCCESS in 9m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/26f15e7f05c44696a5fc22780b429a13 : SUCCESS in 11m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1f383c33089d49f5a3f1cbe71ccb2bf2 : SUCCESS in 4m 52s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/680aa3d31b1a4f76bded2dbc2fd8d5ea : SUCCESS in 11m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ae9ac619921346b4aef0796ab7fa7424 : SUCCESS in 4m 46s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/54d8f02a37e841acb26c22117686382c : SUCCESS in 20m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/62df3677db9f4ed18d9767a5bcee60f4 : SUCCESS in 14m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c38052aaaa3d4dd7bc708fab75719e85 : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c6cdb82d73ff42c99403b2d80b33170f : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/035f6e25e69e45cf90b66ba905a70e94 : SUCCESS in 19m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/81339919f0404c15b810cd351b89d1ff : SUCCESS in 5m 55s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/21bf45a8572c4804828593e87b42f7db : SUCCESS in 5m 14s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/05e55341bd0f4ab09a6f2b41ce464ba3 : SUCCESS in 8m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8c2f89a6b64542fd89266abf2715774c : SUCCESS in 11m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/028893b5288f4ec587cdcebc212c9938 : SUCCESS in 49m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2f5a1f66bbe442808b952313deaedb28 : SUCCESS in 1h 37m 36s","accounts_in_message":[],"_revision_number":1},{"id":"7f1a188150c22a4baf44333bc83070221ea3f462","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:07:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"78bef2970a4f238f9cb4d72071296526333d8039","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:08:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a1ffd3a77c443f5b1ffa848386439a5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf43be78213b48b5982931bb488be5fc : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b1a1aa557954c633a7b02b97558b2d9b8b745137","revisions":{"b1a1aa557954c633a7b02b97558b2d9b8b745137":{"kind":"REWORK","_number":1,"created":"2022-07-15 08:17:42.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/53/849953/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/53/849953/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/849953/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/849953/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/53/849953/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/53/849953/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:17:40.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:17:40.000000000","tz":0},"subject":"update constraint for python-cyborgclient to new release 2.0.0","message":"update constraint for python-cyborgclient to new release 2.0.0\n\nmeta: version: 2.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ibaa7d0a30f53427bc04a0156d056c193525209e7\nmeta: release:Code-Review+1: Brin Zhang \u003czhangbailin@inspur.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I5c1e24a0dc425058650ef7a4a6465bf07ddee51d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1a1aa557954c633a7b02b97558b2d9b8b745137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b1a1aa557954c633a7b02b97558b2d9b8b745137"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
