)]}'
{"id":"openstack%2Frequirements~849957","triplet_id":"openstack%2Frequirements~master~I2f35c02fb4b6d3a646cd0752db17ca925c5839ad","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I2f35c02fb4b6d3a646cd0752db17ca925c5839ad","subject":"update constraint for python-neutronclient to new release 8.0.0","status":"MERGED","created":"2022-07-15 08:19:34.000000000","updated":"2022-07-15 23:03:55.000000000","submitted":"2022-07-15 23:03:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849957-new-release","meta_rev_id":"6c1c201753277bbd44558d06b659962ef47ed076","_number":849957,"virtual_id_number":849957,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 23:03:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 18:22:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 08:29:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 18:22:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 08:29:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 08:32:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 10:00:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 18:22:44.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"66e94e24b7cb058186cc6aba9818000bfc29f43b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 08:19:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d4cad357795d09d141fe2b09b0475e8828db383","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 08:29:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6510b1c937648e14560fc46e96131b5fe687ef9b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:32:17.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7545fe006b7d40a8ac5713984e9adb74\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/a5930a4105264721a889640e4c877375 : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a92a3b1fb6a72ac1cba492a1c3726fd3ec9a7b58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:00:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fcee3acb2f5437ebd314f19609fb7d0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf19a6755c204b5d8a4f9bfb905437d6 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a16825ec9b24c88aa425c586a074551 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/acc4e70dbfe5417a82e9f9377817807a : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da7dd005a3c24702966d86036fe666b0 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/751d0abdaf7c49b1a5953ff0e9cc82d4 : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1ae2414b80f646efb83eb58be8b831cf : SUCCESS in 4m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bdabbfd4614d4fb0854285cc02386bc4 : SUCCESS in 3m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/204c78bcf4d5488d856ff9642143e98d : SUCCESS in 7m 52s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/330d48fe4b99429fa14b3aefc5ac3793 : SUCCESS in 12m 22s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/88917b8ede7d4d02a9c183c3b098b5de : SUCCESS in 9m 31s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ae2d19acc5b243ec8ba9048dedf57353 : SUCCESS in 4m 47s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/5a69d7e4704a4601b5cb5b7a354060be : SUCCESS in 7m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/7d002c3d19b5497a82ad9da5ca3e457e : SUCCESS in 10m 12s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3fdf8ecd68da4fc7a4f713ec644367cc : SUCCESS in 5m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4810296704c84dec83e4c0ef4a205f9e : SUCCESS in 4m 49s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a52dfaee318141868b5cf7d889320430 : SUCCESS in 8m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/ed30172b4e504836bcd561bf32ccc675 : SUCCESS in 9m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/816511b60d774cfa977c1659f34769d3 : SUCCESS in 9m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/bbece9747c534a9884c323039b3ac623 : SUCCESS in 7m 19s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b783a27569724af688ab94dfdf8d24ce : SUCCESS in 12m 37s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/92b42bf485cf46508ec57c31618306c5 : SUCCESS in 3m 57s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e8d095b37e874966a6c913316a83aadb : SUCCESS in 9m 55s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5c894d91413244ed8433395aaa47ad18 : SUCCESS in 4m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/cb2ea953e7974ec893dce07b19e8e567 : SUCCESS in 24m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/5e501fb369534fbc8da199c95d650c8e : SUCCESS in 10m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/1a0ee33b057e497580f672cd3b0eccf3 : SUCCESS in 5m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4c9c5ad6924641fe8a0f45a3cfd595d5 : SUCCESS in 3m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a0c2b988f6c141a4b7779eaf02b3c9ff : SUCCESS in 21m 13s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/87a798561530488483c16407e6f7b26a : SUCCESS in 6m 23s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/ac8c8d00633342118d586a4c4bae925b : SUCCESS in 4m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0e0a68e122a040d0836c419a993baf44 : SUCCESS in 7m 19s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e0d94a5e5b9b4be6ad583ba98e52d324 : SUCCESS in 11m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c25c2fd7e1764728b53afeef7259cbdc : SUCCESS in 47m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/49d198929ccd4932ab4c2958c4efa55d : SUCCESS in 1h 34m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ee91d943d46a2795a7beb8efbd086dfdd6e36601","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:22:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"17a49dd5c9f8cea568525702f7a7e4bc65284c9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:25:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f2529f068ba7ffaf792d1adb545ba8339650e0a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:03:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f7be9ee6ce74dd6932e3f3446e31d18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd43216795314a6eaf5f504d412d80ca : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb4438b30cdf42588cc2f22c1912781d : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/921e7a148e22407b9f925e3adeca38e4 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f90ae5e3f2d243e5b6b6c8001f9aab35 : SUCCESS in 4m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd70085ff59c4551aa5bec89908a7b62 : SUCCESS in 4m 36s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/69c3841625e14e64a192839279f6c90d : SUCCESS in 3m 25s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bb50ebb68b3a463bbcf123d7681d17cc : SUCCESS in 9m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cb282c637fc145e99426815c40658b92 : SUCCESS in 13m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6ecb5b3a3b6c42c8a21cc8385ea25be0 : SUCCESS in 14m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d8233a06d8c948b6b940452e258be07d : SUCCESS in 5m 38s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/88a9a4f64d8f4de2af9269a0d4156f86 : SUCCESS in 6m 00s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b4a70896fe56449a8c5ca5963bf3dce7 : SUCCESS in 9m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/107fcaa94ef24d33b31f424fb6d200ed : SUCCESS in 12m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a6c72b708cea46678140bfcaec07bb86 : SUCCESS in 9m 37s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/de86bad9b0d449068bcd7473c5e5851f : SUCCESS in 8m 09s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/51612d642b02438985b02e4b35b87d87 : SUCCESS in 13m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/89eebb40a43a43b0abfc6d29135a6fd8 : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/937c511765644471a509e6ea8e7aac9c : SUCCESS in 11m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/f6becd15a96a4d7f8887adb4d2271afb : SUCCESS in 5m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/68a6f2163a934f5ba3b92093464e47a2 : SUCCESS in 21m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b602e4a882ae44a1bd856e4c3ba04342 : SUCCESS in 13m 05s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/f4032e328df6421e94f915cf0f6898a3 : SUCCESS in 6m 49s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c2c7f694818c4999833a7f99b49178fb : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/115f85e2cee94380b683f51835612dd1 : SUCCESS in 19m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7fed650034444734ba94ec0febfbc0d1 : SUCCESS in 6m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6a1f9f6682f04c07a527b19f9c4a6218 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5a9385c5ea09465d9cd53748701be746 : SUCCESS in 7m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/1738609cdf824dec93e66a9f7ec58f6e : SUCCESS in 11m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65abc692a7f04c15b64bbafd54b7a533 : SUCCESS in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ae5941a9db584a30813cd523bb3a595f : SUCCESS in 1h 32m 21s","accounts_in_message":[],"_revision_number":1},{"id":"6c1c201753277bbd44558d06b659962ef47ed076","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:03:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"039dc1ac1f2e475b0de71a0590e95c6751f7f313","revisions":{"039dc1ac1f2e475b0de71a0590e95c6751f7f313":{"kind":"REWORK","_number":1,"created":"2022-07-15 08:19:34.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/57/849957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/57/849957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/849957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/849957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/57/849957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/57/849957/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:19:31.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:19:31.000000000","tz":0},"subject":"update constraint for python-neutronclient to new release 8.0.0","message":"update constraint for python-neutronclient to new release 8.0.0\n\nmeta: version: 8.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I903cd6f2c888bbf14e1fa8e5ce4ca4aa5cc5311b\nmeta: release:Code-Review+1: Lajos Katona \u003ckatonalala@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Workflow+1: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nChange-Id: I2f35c02fb4b6d3a646cd0752db17ca925c5839ad\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/039dc1ac1f2e475b0de71a0590e95c6751f7f313"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/039dc1ac1f2e475b0de71a0590e95c6751f7f313"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
