)]}'
{"id":"openstack%2Frequirements~849960","triplet_id":"openstack%2Frequirements~master~I15db1f7d011c1210892c4725fe34d11b71e1ff51","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I15db1f7d011c1210892c4725fe34d11b71e1ff51","subject":"update constraint for python-keystoneclient to new release 5.0.0","status":"MERGED","created":"2022-07-15 08:22:37.000000000","updated":"2022-07-15 22:57:33.000000000","submitted":"2022-07-15 22:56:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849960-new-release","meta_rev_id":"20b335ef32d115457d53025b1cf9d63ec05fec90","_number":849960,"virtual_id_number":849960,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 22:56:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 18:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 08:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 18:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 08:29:32.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 08:35:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 10:15:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 18:22:43.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8b579069cc402802a22406e66a92c170f1e271e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 08:22:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca483a031abc2cd0c4665f1298f121a29b5cdcad","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 08:29:32.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d3fa6da48f3612161f62fef9081b0b4f868448dc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:35:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ccedbf20744b4a0ca3438fd51ca4fc77\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/564a5dee07cc4f7d860c0683bba5bc35 : FAILURE in 10m 37s","accounts_in_message":[],"_revision_number":1},{"id":"688127019bd560791d938b3749a09f14dfd7b347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:15:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d43c028e8e3046ae8c18d1ff8009bed4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a89663e00ef4fc69472e28b75f6fd19 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37566b740650447c8310e8ce439cb681 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6da13a6e92784cab97da190b340cbba3 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e74bb34c4ab435db2a262e64938aaf8 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9448a512b383488d8250fe93f7abd9de : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1e26d4a394344f84912922a24daf8e58 : SUCCESS in 4m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6fb9b16c176249d594e8d9d635b8151e : SUCCESS in 4m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/dc3cd0a4db8748009695d6a57c57db42 : SUCCESS in 9m 42s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/f019bcb549d749129ab5165a0fd30ac7 : SUCCESS in 14m 21s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ba9eea6dea91493f80cd5f182511575b : SUCCESS in 10m 48s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/637bbec7a8874fd78c3cef75a1fb3680 : SUCCESS in 6m 21s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/ad894be7ded34431868ed3888a4a79b6 : SUCCESS in 7m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ac60f63f0cd64b15b02cd4977f818244 : SUCCESS in 11m 59s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b5e1659a3beb408b99be14f2b616387a : SUCCESS in 5m 47s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/dd131d69df3445c6ad8428a5842a57ad : SUCCESS in 6m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/793c3b9359df4856aa88267f1353e629 : SUCCESS in 8m 46s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/c5f4bef0613347dd9cf30588bb97b47a : SUCCESS in 10m 38s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/7ebdc517d4d14da69cfa2a64086b2695 : SUCCESS in 9m 20s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5fa083883c444aa9bbcd605caf671b11 : SUCCESS in 8m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e2d1f6543eec498d9442d82ad9c2463b : SUCCESS in 13m 46s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/322e0cacff1642908d714f755c93c682 : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2b1dc43ef0464b07ad6bdc13c93c1de8 : SUCCESS in 9m 28s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ae96424b52be41669300879acbc8e0c7 : SUCCESS in 4m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/709dffcd0a5f4ac491de54460dc40e1a : SUCCESS in 21m 00s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/edc05f3f12ff4ff9be7eec6fbb31da97 : SUCCESS in 12m 41s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/efc69796bca44a3a835131c34c504e57 : SUCCESS in 6m 25s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9cfbf1f9b2d142c3839f779600ade7db : SUCCESS in 4m 06s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3d958ac177124db0826f2baa6ed09676 : SUCCESS in 20m 23s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/90c4bd5cc26349aebca597fa7c90fa62 : SUCCESS in 6m 00s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/eec451a26fdf4371b851071fcb878762 : SUCCESS in 5m 24s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/8fd43ca440a048108465bc08c42f98df : SUCCESS in 8m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ec237ea1f1544b9a852a20226df81652 : SUCCESS in 12m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60377f2086464eafa0739d1bd8bb4ec6 : SUCCESS in 1h 04m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10b64e643dff49a591b8fbd08dcd8f79 : SUCCESS in 1h 45m 37s","accounts_in_message":[],"_revision_number":1},{"id":"ab1e55924f6d2c1b8dd3b1e9f8ce77ec8fe81818","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:22:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"27f0026f12eabf4c52a3871463ef46e931799ba1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:24:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6daa8eae43cfe6e009f6ffee82bd7714008e8424","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:56:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4187d78309e5444b95ee1142112b932f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82c86ee7755e437aa72b015b5a143e76 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d13999b3828d48c1930062d5449de7b3 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0a1b237e54b4a6bb52bc9cb0ae3941c : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f77d1efb1ac64e7193e3b25c0d0bc7a9 : SUCCESS in 7m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a69cf451e0094033b9c006999512c28d : SUCCESS in 5m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/55617f16cfde48c2a94124f158ea328d : SUCCESS in 6m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0090933b4d154613bec921af6de2c6d1 : SUCCESS in 8m 08s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/33fb088f28384f7cad9762e17859fcdc : SUCCESS in 13m 16s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/962e49268e084062bdeaac2d2f4f4c74 : SUCCESS in 11m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a7d48c09864841e79ce7fcc62b85ced3 : SUCCESS in 7m 31s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/1df6ecaab9a3415c87961ce0a865ff06 : SUCCESS in 6m 43s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0cd99919823949549dc1b26df9aa4140 : SUCCESS in 10m 22s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/44a217db76dd42b3abf155477f5b87cf : SUCCESS in 10m 53s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9553c1c314e645c68c88902c9b873903 : SUCCESS in 8m 51s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/35dfae9dfe044aa1a4ae9a3b3e1cfa56 : SUCCESS in 10m 27s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ac7e9eabff8a4a1a98258f62a9f28bf3 : SUCCESS in 12m 32s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/73de0715876f433ab7ae15f9259bbf43 : SUCCESS in 4m 59s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d1897d821b35493d86bf44f18962f317 : SUCCESS in 11m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/1094e38b2cb24807ac4a20b0565321ea : SUCCESS in 7m 50s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/fb17d4c898704b3abe9aa3795fde0df3 : SUCCESS in 24m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3e4438cc148e4c1e80ddfbf042769a1b : SUCCESS in 13m 08s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/6ad5d3bc1fee467aa48fda80eac8a149 : SUCCESS in 8m 20s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b992468e8290443daa63eb808bff62b7 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7cd2c640a7564a80b140f0cb6ec7a293 : SUCCESS in 21m 43s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/df33de5dad674931bc5c86275cca090f : SUCCESS in 9m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/96b11ad897d145f49a1f76f04f5ce547 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/be86042d6f294052a9269af0fd480a80 : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/b6e7b0f574564d42b4a3a2925cb5ae88 : SUCCESS in 11m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f029c0a4d468429ba324a33fac9c243a : SUCCESS in 1h 00m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/013007a7d84d4cf495756e2b32c5bc20 : SUCCESS in 1h 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"0cd4b7e907f39b0a833f55dbbb24bb80f6825597","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:56:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"20b335ef32d115457d53025b1cf9d63ec05fec90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:57:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65df9c41d9144ed5abd54c10a45b30db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c57eed31c424f6fbc69feb04c3deba6 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74621ae428d8f28c08221f906b68687edb5af395","revisions":{"74621ae428d8f28c08221f906b68687edb5af395":{"kind":"REWORK","_number":1,"created":"2022-07-15 08:22:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/60/849960/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/60/849960/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/849960/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/849960/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/60/849960/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/60/849960/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:22:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:22:36.000000000","tz":0},"subject":"update constraint for python-keystoneclient to new release 5.0.0","message":"update constraint for python-keystoneclient to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I95103dff2ac80cde5e775ca547c2764adf3b5ac0\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I15db1f7d011c1210892c4725fe34d11b71e1ff51\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74621ae428d8f28c08221f906b68687edb5af395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/74621ae428d8f28c08221f906b68687edb5af395"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
