)]}'
{"id":"openstack%2Frequirements~849962","triplet_id":"openstack%2Frequirements~master~I60129821cae721fc6de1b48a977ac1750670bddb","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"I60129821cae721fc6de1b48a977ac1750670bddb","subject":"update constraint for osc-lib to new release 2.6.1","status":"MERGED","created":"2022-07-15 08:32:55.000000000","updated":"2022-07-15 23:23:24.000000000","submitted":"2022-07-15 23:22:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849962-new-release","meta_rev_id":"319a2113849fa12148d197e0fc5576d3bc5876a5","_number":849962,"virtual_id_number":849962,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 23:22:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 21:28:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 08:45:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 21:28:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 08:45:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 08:47:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 10:22:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 18:30:12.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-07-15 21:28:49.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"6099e2a418bb2633673b1e02dbba44bb1d111f48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 08:32:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0eb456f7e8fff56c1fc896f57892efe8dd6f6170","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 08:45:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a4bf8169f430b8cc7f2b122257e7b2f758d90d08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 08:47:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ecdc7216fc844a7b60fa0f53af87312\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/e9b649d691f04a6c9fe5b31fdef0827c : FAILURE in 13m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9c0c56d05732b9169a12e92428730a01889d41e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:22:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6ef848ea9fc4e5da2beb69cafc69fa3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c612fc4e024580b8f62d0728ec8244 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0d7a519cd8b401991f6341e5d17bc68 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b366c4e1d96f40cc8605e03502fb5fe2 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f8a433ebb00415583cb1a7c32450349 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/074ea2ab3e7e40bea4b20b9bea1e25a1 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/edea8d67fdee49098440f3ca6547c0f2 : SUCCESS in 4m 29s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/14f0f59de0414f4494c0c29d7b3cd3ab : SUCCESS in 4m 41s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/84b3f0667c484d18975b7a0e6dfda53b : SUCCESS in 9m 12s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6ea0a9f26230438b8ddaf5d4027bcec8 : SUCCESS in 12m 29s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/e95444c6625b442c8825b1ce2ab70d19 : SUCCESS in 9m 24s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/1f1696b244294f988604cf7864dfb3cf : SUCCESS in 8m 22s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/dd00f93ff3d14ebbb734a7fcd34a4d0e : TIMED_OUT in 41m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/57b1754909c74e39a854a9c28bb64f1e : SUCCESS in 11m 37s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/557484aa22cb4aacbd954b169b190b5a : SUCCESS in 5m 52s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/4dbdd4ad15014c5a944a20aef32ab08a : SUCCESS in 6m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/cd53f3f3f2b14a0d9250313dfad4fbbf : SUCCESS in 9m 02s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/d5038b375a494524af22cf4cdc3f6d85 : FAILURE in 4m 00s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ea4179b73f8149cabb7842a06482c182 : SUCCESS in 7m 54s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/8075ba6e362f4a8192a30746892781b8 : SUCCESS in 10m 08s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/76f29d84633a41049713028b2bc83cd5 : SUCCESS in 11m 56s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/526ac681dd3f4173bb8b1dc4d5354253 : FAILURE in 3m 32s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/3670308d880b4df1ba72bf14364fe4e1 : SUCCESS in 12m 05s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/7c09602729ad4b898efd56a05527fea1 : SUCCESS in 5m 56s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/b984b85100d7448eb46d7e7f19967201 : SUCCESS in 23m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c4a34fcbc6e14af18c3a32c2f8b8ff9c : SUCCESS in 12m 30s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/63bdbbb9f6904f389024c4c0a29b45e8 : SUCCESS in 9m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bc7687bdf9aa4c118448d8b6c07c4bd3 : SUCCESS in 6m 33s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c0bc10eb8c874cbe8e6eab0b08b56721 : SUCCESS in 18m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1d429b3a92ab4dc4a0ca076bc85d1bad : SUCCESS in 11m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/6124c18f5e8c447e8d6a8a4f620f9593 : FAILURE in 3m 38s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/19470dbbaff04d9897364eed89125457 : SUCCESS in 10m 51s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/423724e3e71d4b1986b1062acf6a6f94 : SUCCESS in 13m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/25c8fad8e4234391bc1d29747e5a296e : SUCCESS in 1h 03m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/360f3c86adbd46148292ae5e62c65cd2 : SUCCESS in 1h 46m 58s","accounts_in_message":[],"_revision_number":1},{"id":"690940d33a70cecf394cec0bb2538ada23767c16","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:30:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"20fc7bae9d4dae007b7cb23e26a2a0f401052bbb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:45:15.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e81a3f1cce34702ababd887d40f11a9\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/97582877ae89415eb1e4df0454f824eb : FAILURE in 12m 26s","accounts_in_message":[],"_revision_number":1},{"id":"cefa9510165141b613487c41f46d483b9d0cec50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 20:17:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00bf52c89a9241c2a32b3fbdbba563ba\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2896ad0daa3e46a4bcb85b3d7b75e166 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f4df9744cff4ca4b9ae44661f38008a : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5cf65dd70f2a496291ed7fa4e87efddd : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d415d6293b19487aadfac91aba95f191 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/790c0b8f9e294cb3b3b85039fa75a13e : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad1588aa5d4043b7a49df7ab8bf0cf3f : SUCCESS in 5m 34s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/07246d27b018491d995c96cb6a6f2c9f : SUCCESS in 8m 50s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/664978cc832a474793c7faa393729f84 : SUCCESS in 10m 22s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/75aff8ea6e54403b8acd8665e371b59c : SUCCESS in 14m 41s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/981994a3082446cf9a036f6635a86a63 : SUCCESS in 9m 47s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/55303905a18d40fbb299ab4b483493b5 : SUCCESS in 6m 50s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/a019c16b6a5445d28c470381fb016286 : SUCCESS in 7m 52s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/a3a5cbd6c70f4ac581d0805d64cac08a : SUCCESS in 15m 51s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f90eceda8af5408ab83e7051c115c318 : SUCCESS in 6m 26s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/0a17d212e22a43869599e8fd343434e4 : SUCCESS in 7m 57s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e882e3afb13f4e0183e618441b175fe3 : SUCCESS in 8m 29s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/b143ca4c408d451f90322869310ce3e1 : SUCCESS in 10m 51s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/e994a62dec29427fad61740d163436cd : SUCCESS in 9m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/47ba207e33f04036bc83c7aebab4bd30 : SUCCESS in 9m 23s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/86f550c29bc44c3db5af6cc7347a51ca : SUCCESS in 16m 07s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/cd01a05de49a48e38ec4dde689d62c93 : SUCCESS in 4m 50s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/2c57a1e3ae694fcc834ad8e587cfac53 : SUCCESS in 11m 20s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/8c2e5b3069af480487309d6720735142 : SUCCESS in 9m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/a09a439eeb5f4e9a9beb0672d1e70da3 : SUCCESS in 24m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/66d5dfd04a514c7cb4e879614ca77b53 : SUCCESS in 12m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3288f3d368ae4f7bb5bda57d2a838f9b : SUCCESS in 5m 54s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c09cfe959d2b4f3abe3cd53e8aa2f07c : SUCCESS in 4m 48s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/85ca1184612d4443988f3f9f4911a309 : SUCCESS in 20m 16s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/af5d2da4b4b943b597651febb788d5ea : SUCCESS in 9m 38s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0ef35fc79c9045f58ede57d3343dc4da : SUCCESS in 5m 58s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d4223d1121114949b7d75abdb7e5357d : SUCCESS in 8m 45s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/dae043c20d4e4a878ddd792144bb881e : SUCCESS in 15m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7627a2fa5d5c4e2483e75761a4fbbc25 : SUCCESS in 54m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/79ee351488c541568ea92dbbe3b5b79e : SUCCESS in 1h 39m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e01d106007aab1d678f805050b1419a50586edfb","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 21:28:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"29466f263b525d3f32723f9f0fd0e0c5ed245814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 21:30:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"991f417f36f324eddd0cb7163f322af8c5e9300a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 21:44:37.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d1fa5617ba34b58afb0fca82bfbbbcb\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b08d84dac0ed4782bc6ce3d61f29d706 : FAILURE in 12m 55s","accounts_in_message":[],"_revision_number":1},{"id":"42da8d3d91e67eb6b7c2a254b32815a43c1d9ac6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:19:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7b888d24b542e88b7b408ca4e4f772\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f666446c01d4ef488f356313ecbdec8 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92324b6e3d254d11b4daf59dbfcc953c : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f31a08b21ecd4d2c83bff6d38b642fd2 : SUCCESS in 6m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01a2631af13744dfaa9c25263a10f697 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e858e458e691440aba57edfd26538a19 : SUCCESS in 5m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1104627dcd0e43e18ab393da061f12de : SUCCESS in 5m 13s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/92035b698af44a04969fd1cd70709cec : SUCCESS in 5m 32s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3553b71a4a0942f897391cb0432373f3 : SUCCESS in 9m 02s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/076a29ae60f84c2bace9faf14ef4e53f : SUCCESS in 15m 54s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/25eb97a036c7471eaf65b69a672baa1f : SUCCESS in 9m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/01a9114b858a4a0b945021f4b33a7327 : SUCCESS in 6m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d4de2a88ffa24c6f87a5b7092929a2b8 : SUCCESS in 7m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/99eedcfa46a64d3497587c668427ab6e : SUCCESS in 13m 26s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0dfdc665f91444529c9e9812c0ede326 : SUCCESS in 6m 32s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3bb7b87a706546af995048c97fcc675c : SUCCESS in 7m 59s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/2aa491d1ca2e4145a1a3ed84b9c7b2d7 : SUCCESS in 10m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6e27b6355f98428a9d76034860e7995a : SUCCESS in 11m 01s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ee1093045d344316ba9db83663d2bbeb : SUCCESS in 10m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e6d14933e8dc4c339e150589cb387bfc : SUCCESS in 11m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/45569931cb7343f0a400d0c25ffe9ef2 : SUCCESS in 13m 12s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/bcd5e1db548045659f276f9650abe6d5 : SUCCESS in 7m 17s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ee0bb461c1cb40f2811818cca5f84835 : SUCCESS in 16m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/23dd5b3d03c944e9ad3a0ce2b4289cc1 : SUCCESS in 6m 37s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/9f2a34b50bab44229a8c38b0db2b9a5e : SUCCESS in 30m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/ea11b028d6b147d39d42cc128d60a9d3 : SUCCESS in 13m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9c68d671d81a4963b64799b5f561c74f : SUCCESS in 8m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c64f6f67e0144ba190fb8ff339876441 : SUCCESS in 5m 24s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/a94b51dcf0104b62aaa8ac564e495510 : SUCCESS in 27m 15s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3bb25f1b5288445f9808f23d87f60b50 : SUCCESS in 7m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/22f789f6f0f844ffb04b22f65cb2a73e : SUCCESS in 5m 49s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d3d7a0046dc049509eabc8ba8bb7f39c : SUCCESS in 9m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/39b2fe04e1cd440aab88ae0b4053c357 : SUCCESS in 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/650766f84e9e432eb45afda835747095 : SUCCESS in 51m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dd9bdc6304e4d58ab762e4002c1d89c : SUCCESS in 1h 45m 57s","accounts_in_message":[],"_revision_number":1},{"id":"febd593b04638238f67ad0e7e6713f73e06fab1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:22:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91efe252b8124192a42e35797deaf0da\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ee1a7d39b449829ff83638e77d75c1 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/748c361fb6794765888fc584622859a3 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe65a55f8c5c4194a57b12cefcbf41b1 : SUCCESS in 7m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/463b14ff9b374d4eb1576719a8d1d3e2 : SUCCESS in 6m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa755a7d99c2408f986532fb266b412f : SUCCESS in 5m 33s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/2b2714234ea9436e81950508627ab1cb : SUCCESS in 5m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/895671af4ca24ea89f5f9b06282aaab0 : SUCCESS in 8m 41s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/d2019347bbdf40b886e034b53f503ea2 : SUCCESS in 15m 40s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d483b7df253240c3929cf8310fdb1a34 : SUCCESS in 11m 36s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/ed8f3fbddf4a4892a275d877001843ea : SUCCESS in 5m 57s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/58c7c2b191e14fa0906bd0d7efd38958 : SUCCESS in 7m 23s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/e7115dfb70cd41389f02eb2adf357f67 : SUCCESS in 8m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/292950188aa5454f9d16eaa3142853e2 : SUCCESS in 12m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/08d4aef7735d4548b058813b00d56fd7 : SUCCESS in 9m 34s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6c1c72d22b7144509008fa2ace5b751b : SUCCESS in 7m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/debc14e7acd8411eba88deee67a63091 : SUCCESS in 13m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/25ce5cb9ee5649cfae3d1e6606e73432 : SUCCESS in 6m 38s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c2dad247d29d4607ab29cf21a57cf989 : SUCCESS in 11m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/dca7420633024313ac526578238f5530 : SUCCESS in 7m 13s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/faf540c1d2b640ee833cc409b8c07bec : SUCCESS in 21m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/f25763ca75e54e56946a4c23447fb2c4 : SUCCESS in 11m 28s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/4f6032e61fc14bd7ade3c4c383344c94 : SUCCESS in 8m 02s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/5ef6e5625f9945a192fc914cc0fdd61b : SUCCESS in 5m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/3f195500fc344c16aa3ce2d8d86f8dfc : SUCCESS in 19m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fb493192778e4ae991fb460af6722170 : SUCCESS in 7m 20s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4b426cf7fe7344daa079b5cbe8169d7e : SUCCESS in 6m 42s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2fdb6e5797714993ad8c335af64d0fde : SUCCESS in 8m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a6d58548221844dcb53f85942b30011f : SUCCESS in 11m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ab4898242ec40a98c872ee13b8d327a : SUCCESS in 40m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b0146494608049208450b4795c87e953 : SUCCESS in 1h 44m 31s","accounts_in_message":[],"_revision_number":1},{"id":"bcf0a300ef0bfd7b356c24013eec28fb5896de62","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:22:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"319a2113849fa12148d197e0fc5576d3bc5876a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 23:23:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c62433a28f3b4fcd9a2816fbb9107190\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b21b74fc4954520b46a0a56bd28f942 : SUCCESS in 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9bc02b4d001a925af3264a9988556c6a3903a58","revisions":{"c9bc02b4d001a925af3264a9988556c6a3903a58":{"kind":"REWORK","_number":1,"created":"2022-07-15 08:32:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/62/849962/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/62/849962/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/849962/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/849962/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/62/849962/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/62/849962/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:32:54.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 08:32:54.000000000","tz":0},"subject":"update constraint for osc-lib to new release 2.6.1","message":"update constraint for osc-lib to new release 2.6.1\n\nmeta: version: 2.6.1\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ib61c64626b1110b196d5d548843765891164151f\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I60129821cae721fc6de1b48a977ac1750670bddb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9bc02b4d001a925af3264a9988556c6a3903a58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c9bc02b4d001a925af3264a9988556c6a3903a58"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
