)]}'
{"id":"openstack%2Frequirements~849965","triplet_id":"openstack%2Frequirements~master~If17c1b9c78e8dc47b8bade81593d7f3be39398ed","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If17c1b9c78e8dc47b8bade81593d7f3be39398ed","subject":"update constraint for python-magnumclient to new release 3.7.0","status":"MERGED","created":"2022-07-15 09:00:17.000000000","updated":"2022-07-15 22:55:22.000000000","submitted":"2022-07-15 22:54:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849965-new-release","meta_rev_id":"71c743d233282f635959877a418593095bee1e62","_number":849965,"virtual_id_number":849965,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 22:54:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 18:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 11:06:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 18:22:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 09:10:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 10:39:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 11:06:24.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 18:22:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"ac151a2bca887532907383ebc2533bd964fa0d49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 09:00:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b6df0c140b0758facdf1b051861d119e9241288","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:10:53.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5d4afd68b7cf4876a8d5f25aa299cb90\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d25446848fa24299abef742f67016d38 : FAILURE in 6m 27s","accounts_in_message":[],"_revision_number":1},{"id":"3b0098ba204ecae3c816faa9da89ff59fc6e0cd0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:39:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8730572d4c5b44918c21608ca9c85936\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd2af195ec8e4441a6baf66d9f9829ef : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0e5a60f735d4186af5b448b9847c6c9 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/990293e2c9e746c8a7fe5aca1ea04f4a : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd1952cd753646da81feceec1b936027 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/237cbbe0528143e783fbdd58260f615f : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/93b8160ff7c3492ea282ebe904b13b75 : SUCCESS in 3m 44s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0830d0539887416999d44873ffc2c0df : SUCCESS in 4m 14s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f7ac49b67f7d47e98dc1550566a759d9 : SUCCESS in 10m 03s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9872aa0399f54ef4b56385212e4551be : SUCCESS in 14m 38s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/13be118547df44be970a832c936fc1ff : SUCCESS in 8m 12s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/47df83ecd068469e9a3605f692e50346 : SUCCESS in 6m 00s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/f4cef9e35f2b490287b5e9b1f2f656f7 : SUCCESS in 6m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/2bb7f5b067534e219e93caf2e1d7c053 : SUCCESS in 12m 49s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b9da365683ac45df888217d6b57b11f9 : SUCCESS in 5m 12s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/89faeb9170db4cedbfc37feb46a6794a : SUCCESS in 5m 35s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1f6a8460b102431a8c828a03fa4b7278 : SUCCESS in 9m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3fac07e90f5b472485c5b9a7b5b1b7ca : SUCCESS in 9m 08s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/bb5681e60918492db557f2e9b2e6d69e : SUCCESS in 9m 17s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/408e659a0d984e95b6b8f13e3cdbda0b : SUCCESS in 8m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/b7dc125749a645cfb350514f1af1b6e1 : SUCCESS in 27m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/02564fae26e748679fe80c64238468af : SUCCESS in 4m 15s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/104d243ee0544e3c9ecd4b81864c78c0 : FAILURE in 18m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/85eafb14975c4ad5a5d6323ee7d63ebf : SUCCESS in 5m 23s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/bb0eb0a730e746288fb0c11cbd72fa18 : SUCCESS in 25m 18s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3eb47f59c0bd4c36b0e2048207ffbd58 : SUCCESS in 19m 07s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/a45f91cbce4f4061be74797280db0f0c : SUCCESS in 6m 32s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/b10d56725a144d92a2f6358855271db3 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8e8ec1a1e7384cbf9e906cde6b1d8591 : SUCCESS in 22m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/517aa010b65044e49d2bd1d7b4cd4d19 : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c067f499514740f19fa75def46b4bd09 : SUCCESS in 6m 05s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/444414db73094ca4a7aa1136da78fbe1 : SUCCESS in 7m 23s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/a145dbc2e21448d38f0bb835e9fce6a5 : SUCCESS in 12m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f56697003f2f415195f17cd66c5cb46d : SUCCESS in 58m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f05d65e216eb48ba9075db069cd4ab83 : SUCCESS in 1h 37m 42s","accounts_in_message":[],"_revision_number":1},{"id":"7822268505089a1695f28e9dfe89b22953ce300d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 11:06:24.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d1e42544b9d861c6a8c71c927c771e79a83eb35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 11:17:58.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c4fafdb8d944da9bb289cbe4d73cbaa\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d78aeaa46d524f6aacbbcd48ba2f069c : FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"a4b96f0162df16d99c90b6ac2b56adb57ef21f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 12:58:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f114abf4ba2b47ad9fca9a6f7bc6a035\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6f96e18e3d4c5c844923b37b742ff6 : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5a9e3b0a71d47afa8f22bee5e109f44 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73bb32ad72da41a2b673bfcddbaf6323 : SUCCESS in 4m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9088f58bc1274b69bf45d8577e9d8e29 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89e450eb23e148bea44cb438cb6550a0 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6c8f6d71101f4ec99dba4ffa9a9098f3 : SUCCESS in 3m 54s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/044b257449004bba908e031e903b0f6d : SUCCESS in 3m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/3aefdf4be4124d59b3883ce4fb0a5d90 : SUCCESS in 9m 54s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e55d5119aef04f209debe0d6826de92a : SUCCESS in 13m 41s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/12d66b6c89a449bda9f66356be9d84f2 : SUCCESS in 7m 49s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/416e7bbb2e1b4182be56f4daa0b2b860 : SUCCESS in 6m 01s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/807e1dea3fc84ac68213fcf7fd80ceea : SUCCESS in 7m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0f26f400d6164032baad3b22ed499dfb : SUCCESS in 11m 22s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/08ab5ca1759e4b5eabecb65939dff401 : SUCCESS in 4m 51s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ffef432ead9c46a0afa472c9dcecf951 : SUCCESS in 6m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ea92206a3f54477995a71d50099b70dd : SUCCESS in 8m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e215617a36af48769f14ba2a7a22a079 : SUCCESS in 11m 43s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/db17258a65c548b0b749a99425eb84f1 : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d9d853ad08334257893fadf5c4bc6f30 : SUCCESS in 8m 40s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/dbc5eb33cfd14504aa23694a302f7a04 : SUCCESS in 12m 27s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/fe0446263f1547bbad7905b8bd178133 : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/1d6075ee1d464f54a592d91167d5241a : SUCCESS in 9m 47s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/291cc2759a2049d3a2db9e364d22ad75 : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ae47bcc02073443d944de6fe2658d7ac : SUCCESS in 24m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/043821e555f1470da1a40c9b19d13d76 : SUCCESS in 12m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/11c5ddf5583a4f27a69daba2b6bd2de3 : SUCCESS in 5m 37s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4478be88fead484ca0533bd71f4813fa : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8380be481ebe419990d8d1dfc8caf90d : SUCCESS in 22m 08s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ae79b77052564ae1912e27aab174d34a : SUCCESS in 6m 11s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/81ba590c16db4fe697445b2a966dce71 : SUCCESS in 5m 31s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba625b5013a4476d991f44273412357a : SUCCESS in 7m 02s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4341f8bfb9bd499c976236260ebfae46 : SUCCESS in 11m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6fa78bdb02ce488a819405436b0eef81 : SUCCESS in 57m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15aff9059a1a4080bc63039773793b3e : SUCCESS in 1h 44m 21s","accounts_in_message":[],"_revision_number":1},{"id":"7ee8d2288c192c6c42436fa7a2d2271300249a62","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:22:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a2b0088dc9c5c4c080c7bebd990783ea49a65c6c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:24:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"61ebab6e936cca9aee0297e71a787ffbcddb8a4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd5f69c402eb4834be26c6f5c5c11a3b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61b50a8936fe42d2a6daf4ae8a8e1225 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecaf7f44c78f4f08971de01a8b26cc7d : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c503da651b246bd80ae56c23ee3fc5c : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37f6411429074afbad0e28276f5f14b0 : SUCCESS in 4m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4594157d26044cca89e265003b095808 : SUCCESS in 3m 58s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/36a57bba7a1748fbb7527a1c782bb37b : SUCCESS in 4m 06s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/0b1c6fdd6ac148fabe8c681d2cca7269 : SUCCESS in 9m 37s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/aec9fc3d20394c038576fe68c070591d : SUCCESS in 13m 30s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/56915e4709f14dd5a7980b6f682e0449 : SUCCESS in 13m 04s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3736eb3f5c8a4d458c3a6209c4cd7d89 : SUCCESS in 5m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d7c4fa1d6ea74ddf89232cf28eca7e0b : SUCCESS in 5m 31s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/1d1e7c422f02413782cfaab6cf89183f : SUCCESS in 9m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/45a8cca98f68453081516d5632c64d5e : SUCCESS in 10m 52s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a5955f8e990442b9b5d28d6ea7f9d66b : SUCCESS in 8m 39s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/40069d179df24795806f592b1e445997 : SUCCESS in 8m 45s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/adf386b6a30f4bb9a049cd83e7058eed : SUCCESS in 12m 53s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/6bf33ff4e40e4b16aac52a4593eb9e5d : SUCCESS in 3m 28s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8475335777d74705be3ccfd027d52c9c : SUCCESS in 9m 26s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/3050b02c1f4f4e729e713ac12b6f36dd : SUCCESS in 6m 03s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/742087419b9346d3b86ee6c1e4076190 : SUCCESS in 27m 37s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/d2b3157b7f204da6a5c3ffd4b8548ca0 : SUCCESS in 12m 51s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/08296ad258694da8b0e7e7dadb0dbb89 : SUCCESS in 5m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/054858dc8b414dacb1e5ef605426f11b : SUCCESS in 4m 35s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5dc288864ed74e0b98fdbdc2a84a8010 : SUCCESS in 21m 11s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b9ac64394cca4c4bb83823c0d90e79f8 : SUCCESS in 7m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/17dd3e8ad762453ca06c5b7e96b5eee5 : SUCCESS in 5m 44s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/74b38571492d43a09afceed1a9b1754a : SUCCESS in 7m 14s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/f2bf094939a04bd5ad1a379e8e02f8e0 : SUCCESS in 12m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a433dbaa38e84808b13a56e0f7797cd3 : SUCCESS in 1h 20m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/77462abe7a664c6182a685f91fe93e48 : SUCCESS in 1h 38m 34s","accounts_in_message":[],"_revision_number":1},{"id":"fcc5c8995ef645a71fe2b720efffa1ed1ca30f3e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"71c743d233282f635959877a418593095bee1e62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:55:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/483348ccb1194e449d38911b52b9056b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bb0b67e54cf47a2a46e4a7456d36b7f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3160283612765de9d8b26650d5458e660d038d75","revisions":{"3160283612765de9d8b26650d5458e660d038d75":{"kind":"REWORK","_number":1,"created":"2022-07-15 09:00:17.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/65/849965/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/65/849965/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/849965/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/849965/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/65/849965/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/65/849965/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 09:00:16.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 09:00:16.000000000","tz":0},"subject":"update constraint for python-magnumclient to new release 3.7.0","message":"update constraint for python-magnumclient to new release 3.7.0\n\nmeta: version: 3.7.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7c1619919889df23bdfd235ed47f23fddb63dd59\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: If17c1b9c78e8dc47b8bade81593d7f3be39398ed\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3160283612765de9d8b26650d5458e660d038d75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/3160283612765de9d8b26650d5458e660d038d75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
