)]}'
{"id":"openstack%2Frequirements~849966","triplet_id":"openstack%2Frequirements~master~Ia184e5acc6df1e1677cffbb66e862cd181603435","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Ia184e5acc6df1e1677cffbb66e862cd181603435","subject":"update constraint for python-freezerclient to new release 5.0.0","status":"MERGED","created":"2022-07-15 09:02:31.000000000","updated":"2022-07-15 22:54:25.000000000","submitted":"2022-07-15 22:54:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849966-new-release","meta_rev_id":"912d3b0a5665e7f3dec27e0ab3e239f697419629","_number":849966,"virtual_id_number":849966,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 22:54:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 18:22:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 11:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 18:22:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 09:14:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 10:38:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 11:06:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 18:22:41.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"8c0a39f26efc85d965dbd486a4700666414ae9e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 09:02:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28a9e430349ce72456cd3dd5245521556f709678","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:14:51.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9338ac5223094f3a9bb193040cc825ca\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/219b486e730745ebbe2c6dd33a163f82 : FAILURE in 7m 21s","accounts_in_message":[],"_revision_number":1},{"id":"e664b99c9fa65184b29f2ab718bc8b6774f8f97d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:38:48.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd43ffca038a4b91b549bf4fec41a4db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5e1180032f04edfbc340d078c103733 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d549b88dad14c238d9bb9d776ab3751 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf688fe8e73b4d06a4c15dc4a0ae1285 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc2b2c6bad234a60bcc4d02edae5cfc4 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34a285ffcbe64d918d033fddbce162d6 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e039f80b333b4ebbb19b20833878a38f : SUCCESS in 4m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/888ac3b3151e449886df7ee7affef8a7 : SUCCESS in 4m 39s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bd43c31329174690a2a34b98896d42bb : SUCCESS in 9m 46s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/51e6c463ae15437fbbb48fedb8befbc0 : SUCCESS in 12m 43s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/354a3768ada1459b815e7e27b3df42f7 : SUCCESS in 7m 59s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c6158f34825043ce9ef004b88cfaf3d4 : SUCCESS in 6m 00s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/51612cb83f2e4c95a84fc86100d8db37 : SUCCESS in 6m 41s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/752698b1b4414d91bc86a30580b430ce : SUCCESS in 11m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/e89ad8b2ec1c4bf183d6d15259fec6ca : SUCCESS in 5m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/332d3130b59b40b08ef8d53f1b5a5d7d : SUCCESS in 6m 37s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ed76329978244052a1cab5cb562d799f : SUCCESS in 7m 59s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0ebb77d0a0c04c8aab7dfc293793b288 : SUCCESS in 9m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0f65e0e8ac6645ff8de695a5e5a0286b : SUCCESS in 9m 14s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/adf6f5e11a43455ab588282a3a284552 : SUCCESS in 9m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/90c199dd71a74c38a0c8693528723b38 : SUCCESS in 13m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2ff014b48b654a8da556dbb3c66f119c : SUCCESS in 4m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ad5e36362e314c4baae2e7f3b6465413 : FAILURE in 18m 37s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/04a33ace049c4334b84907d405f29e22 : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6d214db1763b47489510249c7fb070f1 : SUCCESS in 21m 53s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/262957e9308c4596981d310ace094015 : SUCCESS in 12m 43s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/65cba7f01947416bb08199990c437c96 : SUCCESS in 7m 03s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4e7dbe809885491b81be172c0c5026fa : SUCCESS in 4m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/007761976bac4cfa92bca6d45d399f5c : SUCCESS in 20m 06s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5721370ca5cf41baa012d00ef7cca3a9 : SUCCESS in 6m 59s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c353c995d8884951b44158d5cfb68a80 : SUCCESS in 5m 48s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b67ead4715a24ad29e9ebfe3974b48e4 : SUCCESS in 7m 34s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/ce7de78b8296416180e33d351ba3028f : SUCCESS in 21m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fffd83edef914f6b975641790efdcb9a : SUCCESS in 58m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/904061ef620f4b3784b2ca48c55bd0d9 : SUCCESS in 1h 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"9a691a37de2f37cdf878d2eabaf60abdb871136b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 11:06:40.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14bd44520a490b10678e0b42bb9d0a8814401d34","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 11:18:11.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5931cfabb304e4290f39d2ee43cd40d\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/6c08c481e12c44c6ae6f31d1c1a3d26e : FAILURE in 7m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6a138f54388873ffbdb9a868bc80099e6762b451","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 12:43:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0538bfcb018a403584b58e0fb368706e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dc4075b926d4dd49d27a9b9f00eb489 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2d48e719804c86bdb33a287cab0940 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0236f92d4f8f404e82d441bc7706c38b : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9fc9ef94dd04a0da74cbb41f868a621 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e506d8c7cc7445cb184b014fa5c283a : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/fcbd10b7c8e046f58004d413cd735d2a : SUCCESS in 3m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/df795295fa1249bdb9e44f93903fc783 : SUCCESS in 3m 31s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/b9ab5fa91c23495ca8ca57130ecda7fb : SUCCESS in 9m 25s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b97f1e6fc05a4aa980407f61ee9ebadb : SUCCESS in 14m 08s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/5933693beaf94839a597cdc0cb4219eb : SUCCESS in 8m 04s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e3316d164bfc4127a3e29e3e1a26a219 : SUCCESS in 5m 51s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/03e4eb9fbbbd4e9a833b1be20211ae29 : SUCCESS in 5m 47s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/793615632f0f422c8fc73c72bea1bfe7 : SUCCESS in 11m 51s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c2c1e468afdd447db8dc23542cd4fcf4 : SUCCESS in 5m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/5496d8a4957e44038240292353a2c3c2 : SUCCESS in 5m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a4582a72f9c04755b007633a0ab0bcef : SUCCESS in 8m 34s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/09ad0cda35ed4130905ba5fa304c892d : SUCCESS in 11m 17s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/c4134d388b73437db8ad22b022d10203 : SUCCESS in 7m 56s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/b078c26069fe455e8f84f2e4ad75cf95 : SUCCESS in 7m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e99d5b3dbf7741e6888db254d84c07c2 : SUCCESS in 14m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/085130fe627d42f2ba9bc036dfc71aa2 : SUCCESS in 4m 00s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/60c66fda82af4968ad4937786488250b : SUCCESS in 10m 09s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d7ac8c12346c4b848b014e4781c0c35e : SUCCESS in 5m 38s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c3b6d50b35d34ad2abed59568bff0598 : SUCCESS in 25m 43s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c79793afb32d4e03bce1ee20c983ac74 : SUCCESS in 12m 10s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c507f2026d684920be5986170bc4bdeb : SUCCESS in 5m 18s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1b30772555244fd89fa6ad55fa913abc : SUCCESS in 2m 59s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/990c61be9da04922a7781a5369757f8d : SUCCESS in 21m 27s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/1f7cdd480885431297aace02dc5b9152 : SUCCESS in 6m 09s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/1047c2434a0042238a805c3a1da5d7ef : SUCCESS in 4m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/d93fdcb3a7314722aad78ff32f8e3387 : SUCCESS in 7m 30s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/d0b64fc307a44fb89588ae03b921652f : SUCCESS in 13m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e2e0f485a99a4fd79d8262f9f3e7e09a : SUCCESS in 53m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05a767f175bc409f86cc7d6304cd69ac : SUCCESS in 1h 31m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6547e0ec14c2691426f332e1fbcefe52be721271","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:22:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"35922921e3714debdaf373a966a9422069aa2688","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:24:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"68dbacbfa9bfb6522c3dd249ea7bceb139c96457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/359726a045e24672b7651eed822b12e6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea1c6d27cd284bd087a7fe794c178c08 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e1661b42e4f4e49b423ac3da5595afb : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f6c20ba8f064a448dff384a15b5fa87 : SUCCESS in 5m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab4aca47188847a1a3d9d63b1b13da25 : SUCCESS in 4m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/057dec35b36942c698e81a164a5ec3e4 : SUCCESS in 4m 35s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/cfb72882c5d042ba89ef4bdb4d2ae2b0 : SUCCESS in 4m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/81cab947d6fd4dbfaac28b88b77fe0a5 : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9b90fb1396954a758c9e04dfe9628961 : SUCCESS in 13m 13s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1d9f616cc35b477ca8c57e64e3c38103 : SUCCESS in 12m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/bd437b86379e43d99c21bae63028b1ab : SUCCESS in 5m 44s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/c4c6618ca6ff41cd8198109239e6f810 : SUCCESS in 5m 55s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/b549ed5d57734a2486e1759933677f04 : SUCCESS in 8m 13s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/069c4e10bbcb44e59af9dbb877733d9d : SUCCESS in 10m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20c851b4a0e0472a9094bd97d41cca24 : SUCCESS in 9m 41s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/303dc90c7ff74a98b333a701efe91fd2 : SUCCESS in 8m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e2888fbb91ed4121b636d3daa408943d : SUCCESS in 12m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b1d71049335a4a118217a36d4ca9538a : SUCCESS in 4m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/98784fac9c5f471cbfd64246f3690fa3 : SUCCESS in 12m 08s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/88bde54768e843ab8bc6f6f63d7898ff : SUCCESS in 6m 27s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6494fbc139fe499a8de958c5f1ae8cea : SUCCESS in 21m 59s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e1685fe63a4d43ddac8d7fbd0e024a42 : SUCCESS in 13m 02s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/dda5550d283e45a8a5b1b4a25a063040 : SUCCESS in 6m 29s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9a797dd1f8b64b128363666327496c97 : SUCCESS in 4m 27s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ef5b666aa55c4e76aaa11393e1ba6d12 : SUCCESS in 20m 52s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/b7565d501d114339971966d4b1ac43f5 : SUCCESS in 7m 17s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/a980e310d2ca45f3b6bf253e7978ea28 : SUCCESS in 5m 53s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/f006c584ff174c2fa7f6d6fc989b6474 : SUCCESS in 7m 37s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c6a5d34e31f7435b82827ffe3a83309e : SUCCESS in 15m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d205fd2b30c411d811fb4288a7d0758 : SUCCESS in 53m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/10d1ca55b19a473bb2d73720503673de : SUCCESS in 1h 32m 37s","accounts_in_message":[],"_revision_number":1},{"id":"912d3b0a5665e7f3dec27e0ab3e239f697419629","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d4240abc0b7cb09345ee2c868665bd7b2b6d3c83","revisions":{"d4240abc0b7cb09345ee2c868665bd7b2b6d3c83":{"kind":"REWORK","_number":1,"created":"2022-07-15 09:02:31.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/66/849966/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/66/849966/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/849966/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/849966/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/66/849966/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/66/849966/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 09:02:30.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 09:02:30.000000000","tz":0},"subject":"update constraint for python-freezerclient to new release 5.0.0","message":"update constraint for python-freezerclient to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ia2cab2e904d51655892e94bc813c82cf02106bd9\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+1: gecong \u003cge.cong@zte.com.cn\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Ia184e5acc6df1e1677cffbb66e862cd181603435\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4240abc0b7cb09345ee2c868665bd7b2b6d3c83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/d4240abc0b7cb09345ee2c868665bd7b2b6d3c83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
