)]}'
{"id":"openstack%2Frequirements~849969","triplet_id":"openstack%2Frequirements~master~Iee0a065375fafa94518a1acfd0d6dfbc6208bb0d","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Iee0a065375fafa94518a1acfd0d6dfbc6208bb0d","subject":"update constraint for python-ironicclient to new release 5.0.0","status":"MERGED","created":"2022-07-15 09:34:02.000000000","updated":"2022-07-15 22:56:10.000000000","submitted":"2022-07-15 22:54:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849969-new-release","meta_rev_id":"fd5d439273ed02d0f3de36ab355d0828caa8a2fe","_number":849969,"virtual_id_number":849969,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 22:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 18:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 09:41:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 18:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 09:41:43.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 09:47:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 11:09:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 18:22:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"39a9401105f67477f6c0b5df82a3e5e935e90e93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 09:34:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47d44568918e68f346b976e4a4292bbbb8ff30c2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 09:41:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d5847a7c695a3a57dece9aebc476fe0f6b20f5e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 09:47:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f59307647fe245b2b4b50836b1c1374e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/182bf3d498cb48df81897814e512c188 : FAILURE in 11m 44s","accounts_in_message":[],"_revision_number":1},{"id":"2639b26261b0dc8136cfaeb7bceba32624da7a8f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 11:09:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f83eba8942413e95349aed30a57360\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df97d45e147c4fdaa902572e1912845f : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05025511756a46099adff6f28e66f0e1 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1afb3378f84244a7965d25b79d12d3b3 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6b0003fb9a794762a658c1b90850c8b0 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d80ac3f44f3c4807aa5f11245bd8f05e : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb8da50ef9594d4590ed441114a0a48a : SUCCESS in 5m 07s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/66da05ade2f64b9490cde39fb6527c45 : SUCCESS in 4m 34s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1347f8ec855244b3825d9c637061854b : SUCCESS in 9m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e7070835ebbc41f48ea847125d7b64a7 : SUCCESS in 14m 41s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d875473ab2d94195895ef0864e548ceb : SUCCESS in 9m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/77b461c12906410780b035b8a2bc72a9 : SUCCESS in 8m 36s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/4b5c4be0981446d692eef95005bdfee2 : SUCCESS in 6m 25s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/66a22cd6d13143c68d92e86f17625b52 : SUCCESS in 13m 32s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/b9c3538651574e7eb516ae529b9b2b02 : SUCCESS in 5m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/be32a732655349a89090191553e7a076 : SUCCESS in 8m 46s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/0b2a0b42e487453bb5fc178866e1e3e3 : SUCCESS in 8m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/57843cd9eba7473b82ba8740d0915b5f : SUCCESS in 9m 39s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/2f67e4ff1d8e4163bce0e84a506278e8 : SUCCESS in 10m 29s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/49084a0b4e3d4758a2b6a3343f82a0a3 : SUCCESS in 10m 43s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/62b476aa2ae743288ad6eb0113544277 : SUCCESS in 11m 34s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/83c7f7630a834141a4e0fcead2ec74a9 : SUCCESS in 4m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d5b5f08a320a48a0a0207ad20f9fa252 : SUCCESS in 10m 56s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ce36760475824cd8aae4a9db3bdc562f : SUCCESS in 5m 55s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/df80c6b326a54bc1943705aabaf43218 : SUCCESS in 21m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a30235e0738a40c1a7d230cb217b6eb4 : SUCCESS in 12m 34s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/34bc7fad05d34622b227440a0a0ea83b : SUCCESS in 9m 09s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/83747aebeed0458ebf3477b935a8da81 : SUCCESS in 4m 17s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/e3c9aed09ca7444e880c15d044060646 : SUCCESS in 18m 53s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/11362f5f42d649a8be4af8da4978879f : SUCCESS in 6m 16s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/fe566727a7164915ba63b2de6a9a7ae9 : SUCCESS in 8m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/582bd55927a2427f89c93eb81bc16eb3 : SUCCESS in 7m 58s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/04119d7d6f774e31b3f4c24dfb2ad290 : SUCCESS in 10m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64a234b3d55940239f47ab0566d437dc : SUCCESS in 57m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d190ed5dd70a4a12a18cd61f2154b749 : SUCCESS in 1h 34m 15s","accounts_in_message":[],"_revision_number":1},{"id":"d49e603fead4e88d81416b6f054dce6d2cbd52be","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:22:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0fbea6205e55f87e8543ad8e8f8cfa8d8597c02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:24:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fc312d2d89a74ee893276c907c472115ff2a392a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26be8bf89f454e70a5003bc9c4577faf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dc89c8970424a4fa6fccb8486a5aa6f : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d91b451fd44cf1ae1b27d03b66f5b3 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c22cefa29725457abaa132539f996be3 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b3e05de8f02b4a74a7f100973cab4562 : SUCCESS in 4m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/69307e9ebd914ff8b5366b1dbeaad4d9 : SUCCESS in 6m 23s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/0b3c99964b624399aa9897cfe9154c42 : SUCCESS in 3m 52s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/1ee94e3b8cc54ff48a39367f135be39e : SUCCESS in 10m 14s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/377e78a4cd1d46eeaeb2a44ca3d768b1 : SUCCESS in 11m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/344eca44151d49aea40f38aff5279b9e : SUCCESS in 13m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d07e64d945fb473e96cade9d10bacdb3 : SUCCESS in 7m 37s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/96815df8dc9749be88043189e84262e2 : SUCCESS in 5m 56s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/be15712b991142c8995db70616e873fe : SUCCESS in 8m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/e91546de31ad439f949e23a772d738f6 : SUCCESS in 11m 32s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/a1b7f5065f534706a599ac81ff628ebf : SUCCESS in 11m 38s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9c61d68d87bb4e698ec67142cd910c64 : SUCCESS in 8m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e507bd7209e8493990824aca44c9f3b3 : SUCCESS in 12m 18s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/79bac3f736ee43d79af817936cb71226 : SUCCESS in 6m 46s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/eda3f9287dab48feafee47227f189439 : SUCCESS in 9m 43s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ed9ca2bfb15144daa350eda59ed91a32 : SUCCESS in 5m 32s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/6f65c5f3213e485eb92a24196cd75a4d : SUCCESS in 25m 08s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e9658b30282a4419992001b05e48a435 : SUCCESS in 12m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/33e428ef88c14050addcb2d14ac9f33c : SUCCESS in 6m 01s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/8d0542ed046c47329c8d35696960c395 : SUCCESS in 7m 09s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c79e6361cc8a4bdfbcca03c7d1ff661d : SUCCESS in 22m 05s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/6d2d1c75c02b42b18c41ab1143b059de : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/4ab782dace964387bc022dd00f1753f8 : SUCCESS in 7m 47s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/46840fd5c5054c73be3591e8de65842f : SUCCESS in 6m 42s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4a8582ef6b9c4511a8c2f77a0996727f : SUCCESS in 10m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/33daf3520abd4a5dbe491a441ce1f17f : SUCCESS in 1h 01m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1b96add68d744839a97c0a9d240ce957 : SUCCESS in 1h 18m 24s","accounts_in_message":[],"_revision_number":1},{"id":"6390e6ff10c89693a17e47935ae41d78da046978","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fd5d439273ed02d0f3de36ab355d0828caa8a2fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:56:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb63ffa5f23045ba91c97f0ead7cd647\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693cf3f569ba4eb8822ccc1c866d2bad : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5c296bd8dcfddf00d7e72cbeee103a1873048d3","revisions":{"b5c296bd8dcfddf00d7e72cbeee103a1873048d3":{"kind":"REWORK","_number":1,"created":"2022-07-15 09:34:02.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/69/849969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/69/849969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/849969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/849969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/69/849969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/69/849969/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 09:34:01.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 09:34:01.000000000","tz":0},"subject":"update constraint for python-ironicclient to new release 5.0.0","message":"update constraint for python-ironicclient to new release 5.0.0\n\nmeta: version: 5.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: I7938cd1c4d9680e9678b6429d7e1e0197f96e06c\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Iee0a065375fafa94518a1acfd0d6dfbc6208bb0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5c296bd8dcfddf00d7e72cbeee103a1873048d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/b5c296bd8dcfddf00d7e72cbeee103a1873048d3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
