)]}'
{"id":"openstack%2Frequirements~849980","triplet_id":"openstack%2Frequirements~master~Id768d933519e3680428d37454a79c5b433558fc8","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"Id768d933519e3680428d37454a79c5b433558fc8","subject":"update constraint for monasca-common to new release 3.6.0","status":"MERGED","created":"2022-07-15 10:08:52.000000000","updated":"2022-07-15 22:54:28.000000000","submitted":"2022-07-15 22:54:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"849980-new-release","meta_rev_id":"505ed79d3c2aa3079d85cef487c460760d0125c8","_number":849980,"virtual_id_number":849980,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-15 22:54:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-15 18:22:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-15 11:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-15 18:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 10:20:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-15 11:05:57.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-15 11:39:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-15 18:22:42.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"22f7b230632c39d88318c16ccbcd2f8074f307c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-15 10:08:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c65733f58058c495954fb4c52a07ed3ce243b24","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 10:20:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/505a88c963d64b9d8bc34bbe3d4f2072\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/1eecee727e9e4e1ea1bb79ebb7f44bf4 : FAILURE in 6m 31s","accounts_in_message":[],"_revision_number":1},{"id":"2a2638f0cc4f3243be7cb232182814bcc9f52892","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-15 11:05:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cd4b2e83436785360e0cc9b50ba22f69ca633b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 11:39:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed0fd5ad91be4a258010b2b73f4d3b86\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1439ec3430a84f8080d22598e8637edb : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b7828559a06490685242f3c02f2f982 : SUCCESS in 3m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4212d4c08c844479d76949b1d79907e : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c851bb31ce714508b791389e28eb6d1c : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4107c6db968949618b5a799bc7f3e38d : SUCCESS in 4m 19s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3a8ec0a7e3aa4843bbf3fa5f323bf546 : SUCCESS in 4m 12s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f021647e798431ba326c85fc24d1c33 : SUCCESS in 4m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/c0fe8c5f0d1148c9a840d75a5840d0fc : SUCCESS in 8m 29s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4b4bff73ac4548cc833cf21065ac387b : SUCCESS in 14m 19s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/f5d48e5e1a6a41a7bc88f8233dd81bdf : SUCCESS in 7m 23s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/578f2eeb239748d1bca91e896dda8b7c : SUCCESS in 6m 11s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/e372c7f15d2c432e95b529ba3557035a : SUCCESS in 7m 23s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/d5bfcaad099b4b7ab34b20e616264fb1 : SUCCESS in 11m 03s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/89d43827c265473e813b787550faec52 : SUCCESS in 4m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/f63b7af7a7d94d14beeaecd6a43c2e0c : SUCCESS in 6m 27s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/3bc4585a876444df9144c1d0a13eff0e : SUCCESS in 8m 24s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3dc9df360f17403e886a865a9df2846f : SUCCESS in 11m 55s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d6bd0c49f6a84173a6aab6492e9fd6eb : SUCCESS in 8m 01s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/e8571be8c44e4d5c9e860cfdeeddae23 : SUCCESS in 9m 01s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4e94707fea8846189a61c6681bc8ea15 : SUCCESS in 12m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9ecf8cce458d447bbc5c971248c2c23f : SUCCESS in 3m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/4cbe88d834e749d7989739dfd24203f3 : SUCCESS in 12m 16s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/eaeede152a00423181c811e773e59efe : SUCCESS in 5m 17s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/c18e1461ee5642bbb86b4530cdfeec02 : SUCCESS in 21m 48s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/951b286417d44384bb6598d9a043cab6 : SUCCESS in 12m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d546a1353a2c44558139b617332b081d : SUCCESS in 6m 27s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/c9dd021456e34508ac1e46da00571c45 : SUCCESS in 3m 11s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/c64df21083a24639a8a93dd73db76555 : SUCCESS in 19m 48s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0b4f80966f6243488e3f0f89d33f1767 : SUCCESS in 7m 02s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/9fceb73f1bb14e9996454ae97d5bf070 : SUCCESS in 5m 17s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/6899bcb635cb4650a01ba048b8457f19 : SUCCESS in 7m 57s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/01d6d58bb66d40baaafeff89975fb739 : SUCCESS in 11m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b8b5d435f7b1460d941b6be954e43aaf : SUCCESS in 46m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1f1e9aebe99f44579ea7cae41934ddb2 : SUCCESS in 1h 27m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5ed00c4bfdbe25b0030341cc2f7250e1a1163b11","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-15 18:22:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c486adc50568a4979922c6f754d6864b8f441cb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 18:24:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1c4e2395b159e33f4f3d189e4bf6733e8fc2d61c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/183ae1e950cb4e199d749cd3d8d14f41\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b3bee84c03480b83ca3f68024d1864 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6289f43b94bc4a46843dfdc93bd91ad2 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e52712ed5a484a258fc36182c2d4bd65 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db56500a3b744b20b275efea37044437 : SUCCESS in 5m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9131d60ddfc74b74a5e2fcb632f591d6 : SUCCESS in 5m 02s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/5f4ca3439cd6436ebbf5395008e89203 : SUCCESS in 4m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7dc25cd15f504cb0bd0973c3809ee652 : SUCCESS in 9m 44s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/1163fe141c0049fd98a00fcf7784ecce : SUCCESS in 14m 08s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/35cf0ff2d4424352b93f4b8b1b7f6e18 : SUCCESS in 12m 52s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/011906da862e48e6ae8a0294a12e803a : SUCCESS in 5m 21s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/b17a22866b55407fb23417288a9b027a : SUCCESS in 6m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/ef411e11d326411bbce89800ee9ca704 : SUCCESS in 9m 03s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/65cbc705c2c944568a29f65a9547f31b : SUCCESS in 10m 45s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/48ff7932efca4a8d91db400b3e124854 : SUCCESS in 9m 42s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/697efb4bfe6b44d791979a92963f0e83 : SUCCESS in 9m 38s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/5f53ce04e2d4495299fa1258fbbeeb0f : SUCCESS in 15m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/dfd26f1431b347f6b6758362a2f8a132 : SUCCESS in 4m 27s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/ccd71e6a90944a969492e151fb434bad : SUCCESS in 14m 25s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/d372d83f68764b1abd044cdef985a436 : SUCCESS in 6m 21s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/be5a4b5fca7247a290c05817de578e27 : SUCCESS in 24m 42s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/0aa1c76562124f0181dbe501dee290ca : SUCCESS in 14m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/35d127ee8e414287b9377bf6c1df26ef : SUCCESS in 6m 21s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/cb88bc5e4d7f46fb8cd32c61da0802b0 : SUCCESS in 4m 10s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ba11b6d7feba40c59560feab51f1b98e : SUCCESS in 22m 29s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/5966591d4dcd466b9e1145e43e19f09d : SUCCESS in 7m 08s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/83663cb6140745c982929dd23e1ff7de : SUCCESS in 5m 32s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/c66827d4dc6342868538e091e8447fc2 : SUCCESS in 7m 52s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/295a7eb21788476f9d69374770decd63 : SUCCESS in 13m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60b7293d14334c0da46ee39781b82a35 : SUCCESS in 47m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cdcea9dd62741d8a77c7d403d231227 : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":1},{"id":"505ed79d3c2aa3079d85cef487c460760d0125c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 22:54:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b711bd1aaf2a0215868b7669484b96abd1e2ceb","revisions":{"8b711bd1aaf2a0215868b7669484b96abd1e2ceb":{"kind":"REWORK","_number":1,"created":"2022-07-15 10:08:52.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/849980/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/80/849980/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/849980/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/849980/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/80/849980/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/80/849980/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 10:08:51.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-15 10:08:51.000000000","tz":0},"subject":"update constraint for monasca-common to new release 3.6.0","message":"update constraint for monasca-common to new release 3.6.0\n\nmeta: version: 3.6.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: no\nmeta: first: yes\nmeta: release:Author: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Commit: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Change-Id: Ibc29d950e7cf477a3f01ce56e7d2db33fde9c932\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: Id768d933519e3680428d37454a79c5b433558fc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b711bd1aaf2a0215868b7669484b96abd1e2ceb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8b711bd1aaf2a0215868b7669484b96abd1e2ceb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
