)]}'
{"id":"openstack%2Frequirements~850020","triplet_id":"openstack%2Frequirements~master~I85501b4bff97d1c1e1873c3329ef998a8e501134","project":"openstack/requirements","branch":"master","topic":"uncap-importlib-metadata","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-08-02 22:41:41.000000000","reason":"\u003cGERRIT_ACCOUNT_12898\u003e replied on the change","reason_account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2022-08-03 01:50:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I85501b4bff97d1c1e1873c3329ef998a8e501134","subject":"Allow backports to have arbitrary python_version markers","status":"MERGED","created":"2022-07-15 15:06:17.000000000","updated":"2022-10-04 11:24:30.000000000","submitted":"2022-08-03 01:50:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850020-uncap-importlib-metadata","meta_rev_id":"ff3664860cd76a4b142a08224adec8941459d781","_number":850020,"virtual_id_number":850020,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-03 01:50:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-02 14:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-08-02 22:41:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-02 14:29:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-15 17:02:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-28 11:09:15.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2022-07-29 10:46:12.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-02 05:27:31.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"CC"},{"updated":"2022-08-02 14:29:09.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"2930db8001ea2bed135f6fdf1bd6c52542ad1929","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-15 15:06:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"425ffcb2489e7c480f1dcc8073a17df3e2ed958c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-15 17:02:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07a9d65736ee495c8c4c258094be4550\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148248ca81764e5490a4787f28e292f0 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/417531feac544d9798ff3a7664b4b7f7 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1010ef919b3047e09535e91b29bda95c : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2671d6c8d14d489d8d91019d172cf732 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/48971e6c8cd04a59871ad2607ba54ecf : SUCCESS in 5m 25s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad87093291414441bf424659128c48ba : SUCCESS in 4m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c617b4bc833745438fe03485eeda6853 : SUCCESS in 55m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6975b65e57b849f7aa5fc32e167515b3 : SUCCESS in 1h 54m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9ea415cc81a5aa838e529ab0a11e12ad9f8d6979","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-07-28 11:09:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"565c077e19905d4bd60347c5caf8676ae7614a15","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-28 14:30:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9579b07d7e3ae3c2e99351a6f6847b7999ddf768","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-07-29 04:40:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"086135a86af00e73f8efd6d328a74950b182cbf3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-29 09:20:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"22c2278fbfe54c23beb970158c49844719f35e7c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-07-29 09:20:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44c242a77e318afd505cfcfaa046544edc9000d6","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-07-29 10:46:12.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e45274553169fa2353ba4d05b9c0afa4e1de4946","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 11:10:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a82ea087fe0b44ec9de3b9cd6d2ba34b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb729bad3048450a9438cf2de27d825a : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17b3196da2c14d6795aeea71364098ae : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26267c7f5f0b4b31b97274ac836b6686 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6679b3421a584a87b145addae3bd8b1c : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/319597988a8b4250b752adface7e87a8 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e164ed2f87c848c7a781639b46366b72 : SUCCESS in 3m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4dd324893e0a4961ba3071f1ed787215 : SUCCESS in 58m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b0ce3c02ea647c6a981f172ca8549c2 : SUCCESS in 1h 41m 58s","accounts_in_message":[],"_revision_number":2},{"id":"78fba8e098c30930435aaa320f5e9e61b0e19d65","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-02 05:27:31.000000000","message":"Patch Set 2:\n\nI\u0027d like the blockers list to live outside of code where it\u0027s more visible (and hopefully not forgotten).  Tests would be nice as well of course (as tony says) but looks good generally.","accounts_in_message":[],"_revision_number":2},{"id":"6e7ee03bb655e5f8065aab0d74f31b3a41c7954f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 09:19:56.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"30a327d7f052f646745ccc61b388f9f59ba5d6b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 10:49:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72ab2cd08ac9467a85dc4c48ac9205ec\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae9fc2279d6d488aa9bd6097a93d64e2 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32f692bab9ad4f54845919cfe94314e0 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07aadea387624457a4b56aba887cd6be : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ecb401884e8d4f0f8b57a430ccb7c7e7 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67483ca17a1c401494757e54263aa581 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7ed37704784440598917bc302e340864 : SUCCESS in 4m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/97c06becc441493e96998280d0d9914c : SUCCESS in 50m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/845e2ca613b341139f53db5aba1c8df8 : SUCCESS in 1h 27m 51s","accounts_in_message":[],"_revision_number":3},{"id":"b6ccedb12926c0ead0b17cd77573f5c26c7cba4b","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-02 14:29:09.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dea9c5528bd056cc5d75317790eab75d993b8506","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 14:29:55.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"01156f6905073c42ef7a69e936deb07e60b45a35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 15:55:13.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6d815f05e374b569aa6902aa5466918\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fc389f7ad394a5f8ec1591c5253096d : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27058343bd3243be80f3349a61206c66 : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19754858a2044b91b6739291d55aadfb : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7339dd1405244d3e97ed9852203ced31 : SUCCESS in 5m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6f0b67ebf08b4229bf48a5153da94c51 : SUCCESS in 4m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/64d7baf7b882485a9ec42927def1024c : SUCCESS in 47m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18094803898c4812b6c52c25370f2970 : FAILURE in 1h 20m 52s","accounts_in_message":[],"_revision_number":3},{"id":"fc3f8652984d86ea5d6e809f47ffa0351bf94849","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-02 22:41:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cecea507ea6f3f8910f3d5315d80b43f785f6bf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 00:13:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e5ab5f9d6a34bfd8c928d77d1b9a282\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c58bf5cfea46480cbee758dd9839d83a : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08e7c21db0284ecf9157b26fbfddb03f : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b78e03aebf5473fbc2f9df290e5cbfd : SUCCESS in 6m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b69568a0a3f459994db5ef951edd310 : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbfa7db629f54752bb8e31fb658551de : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa259c255d6347edbf8c16f53046cce9 : SUCCESS in 5m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/acea2e9608f640c5adb07b9435d30fef : SUCCESS in 51m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f19f702c2b7c4a5f9fedcff0efbbc8d1 : SUCCESS in 1h 30m 23s","accounts_in_message":[],"_revision_number":3},{"id":"36a7973b4969d2b273dd547c3583a8913ed8ed13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 00:14:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"489cfabea10506551546d091925bcc440109af5b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 01:50:46.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f90d839d56e481a911afe4105cc5c9a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b4cb934d4424819987ee2b7b7b85d64 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64d96597af9c4e7cb8cb3e343f084912 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/20e4c0772f20477bb770f5403cffaeec : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071583acb36b4f5ab4ce8a1e99a4881a : SUCCESS in 4m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/67bf9fe2cbd64f43ad1a5be51dd94526 : SUCCESS in 6m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c90fef9cd1a7459bb5a7ccf3259b37ae : SUCCESS in 38m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f05f1996ddcf49999432cec8604cedf8 : SUCCESS in 1h 31m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e06f5d1d67134dde7bce53db315f21b9b0e2476d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 01:50:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"f935c82ff9b56e714422090c2f545ee85c207800","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 01:51:40.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f354833d14df4cd7848ed862b5e87107\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ee417ff928415b90c00ec16fa2e072 : SUCCESS in 35s","accounts_in_message":[],"_revision_number":3},{"id":"ff3664860cd76a4b142a08224adec8941459d781","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2022-10-04 11:24:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a968919ebe3cd8d25df293d5d5f1eefadd79d542","revisions":{"baf47231f58d5fec117162ec0ac6b41b7a6b9b32":{"kind":"REWORK","_number":1,"created":"2022-07-15 15:06:17.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/850020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/850020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/850020/1"}}},"commit":{"parents":[{"commit":"dd5375f5291e0353c7fa05e48c50116dba9c6e9e","subject":"Merge \"Bump pyroute2 to 0.7.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dd5375f5291e0353c7fa05e48c50116dba9c6e9e"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-15 14:42:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-15 15:06:08.000000000","tz":60},"subject":"Allow backports to have arbitrary python_version markers","message":"Allow backports to have arbitrary python_version markers\n\nThere are a number of backport packages in place for packages in stdlib\nthat are still evolving. One such package ins \u0027importlib-metadata\u0027 which\nis a backport of \u0027importlib.metadata\u0027. The version of\n\u0027importlib.metadata\u0027 found in Python 3.10 provides a function,\n\u0027packages_distributions\u0027, that is not found in older Python versions.\nAnyone that wants to use this function on e.g. Python 3.9 needs to use\nthe backport library (version 4.4, specifically). However, anyone that\ndoes not need this additional functionality can avoid using the backport\nexcept on very old Python versions. It should be possible for libraries\nto determine exactly when they stop requiring the backport library. Make\nthat possible by allowing the \u0027python_version\u0027 marker to differ for\nthese libraries.\n\nChange-Id: I85501b4bff97d1c1e1873c3329ef998a8e501134\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/baf47231f58d5fec117162ec0ac6b41b7a6b9b32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/baf47231f58d5fec117162ec0ac6b41b7a6b9b32"}]},"branch":"refs/heads/master"},"41adecc457bb788063417b993b5e92031ece8436":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2022-07-29 09:20:25.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/20/850020/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/850020/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/850020/2"}}},"commit":{"parents":[{"commit":"ce3bd6830a51382c8d53989ab174572b177a903d","subject":"Block os-vif 3.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ce3bd6830a51382c8d53989ab174572b177a903d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-15 14:42:19.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-29 09:08:19.000000000","tz":60},"subject":"Allow backports to have arbitrary python_version markers","message":"Allow backports to have arbitrary python_version markers\n\nThere are a number of backport packages in place for packages in stdlib\nthat are still evolving. One such package is \u0027importlib-metadata\u0027 which\nis a backport of \u0027importlib.metadata\u0027. The README [1] for this package\nshows a mapping of versions of the \u0027importlib-metadata\u0027 third party\nlibrary to versions of stdlib \u0027importlib.metadata\u0027:\n\n  importlib_metadata  stdlib\n  4.8                 3.11\n  4.4                 3.10\n  1.4                 3.8\n\nA project may want to use a shiny new feature only found in the Python\n3.11 stdlib version while continuing to support Python 3.8, 3.9 and\n3.10. If so, they could specify something like so:\n\n  importlib-metadata\u003e\u003d4.8;python_version\u003c3.11\n\nMeanwhile, other packages might require features found in the Python\n3.10 stdlib version. They would want to specify something like so:\n\n  importlib-metadata\u003e\u003d4.4;python_version\u003c3.10\n\nCurrently this is not possible as it is not possible to specify\narbitrary \u0027python_version\u0027 markers. Since \u0027importlib-metadata\u0027 does not\nhave a \u0027python_version\u0027 marker in \u0027global-requirements.txt\u0027 (not since\nchange I5febaed02e95ff27accd946abc32f3bcbb1a5ead anyway), both projects\nwould have to specify\n\n  importlib-metadata\u003e\u003d{required_version}\n\nWhich means they\u0027ll use importlib-metadata even where they don\u0027t need\nit.\n\nFix this issue by allowing a list of particular modules to specify a\nproject-specific \u0027python_version\u0027 marker. Currently this list only\nincludes \u0027importlib-metadata\u0027, but in the future we may wish to enhance\nit to include other rolling backport libraries (e.g. mock).\n\n[1] https://pypi.org/project/importlib-metadata/\n\nChange-Id: I85501b4bff97d1c1e1873c3329ef998a8e501134\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41adecc457bb788063417b993b5e92031ece8436"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/41adecc457bb788063417b993b5e92031ece8436"}]},"branch":"refs/heads/master"},"a968919ebe3cd8d25df293d5d5f1eefadd79d542":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2022-08-02 09:19:56.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/20/850020/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/20/850020/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/20/850020/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/20/850020/3"}}},"commit":{"parents":[{"commit":"bed46f4b7ab58badcddd45e0666481eb89aedcaf","subject":"Exclude Werkzeug 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bed46f4b7ab58badcddd45e0666481eb89aedcaf"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2022-07-15 14:42:19.000000000","tz":60},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2022-08-02 09:07:10.000000000","tz":600},"subject":"Allow backports to have arbitrary python_version markers","message":"Allow backports to have arbitrary python_version markers\n\nThere are a number of backport packages in place for packages in stdlib\nthat are still evolving. One such package is \u0027importlib-metadata\u0027 which\nis a backport of \u0027importlib.metadata\u0027. The README [1] for this package\nshows a mapping of versions of the \u0027importlib-metadata\u0027 third party\nlibrary to versions of stdlib \u0027importlib.metadata\u0027:\n\n  importlib_metadata  stdlib\n  4.8                 3.11\n  4.4                 3.10\n  1.4                 3.8\n\nA project may want to use a shiny new feature only found in the Python\n3.11 stdlib version while continuing to support Python 3.8, 3.9 and\n3.10. If so, they could specify something like so:\n\n  importlib-metadata\u003e\u003d4.8;python_version\u003c3.11\n\nMeanwhile, other packages might require features found in the Python\n3.10 stdlib version. They would want to specify something like so:\n\n  importlib-metadata\u003e\u003d4.4;python_version\u003c3.10\n\nCurrently this is not possible as it is not possible to specify\narbitrary \u0027python_version\u0027 markers. Since \u0027importlib-metadata\u0027 does not\nhave a \u0027python_version\u0027 marker in \u0027global-requirements.txt\u0027 (not since\nchange I5febaed02e95ff27accd946abc32f3bcbb1a5ead anyway), both projects\nwould have to specify\n\n  importlib-metadata\u003e\u003d{required_version}\n\nWhich means they\u0027ll use importlib-metadata even where they don\u0027t need\nit.\n\nFix this issue by allowing a list of particular modules to specify a\nproject-specific \u0027python_version\u0027 marker. Currently this list only\nincludes \u0027importlib-metadata\u0027, but in the future we may wish to enhance\nit to include other rolling backport libraries (e.g. mock).\n\n[1] https://pypi.org/project/importlib-metadata/\n\nChange-Id: I85501b4bff97d1c1e1873c3329ef998a8e501134\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a968919ebe3cd8d25df293d5d5f1eefadd79d542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a968919ebe3cd8d25df293d5d5f1eefadd79d542"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
