)]}'
{"id":"openstack%2Frequirements~850034","triplet_id":"openstack%2Frequirements~master~I054a38e3715dac3f014c729fce52c9d9c6bbd2fa","project":"openstack/requirements","branch":"master","topic":"pyroute2_0.7.1","attention_set":{},"removed_from_attention_set":{"14288":{"account":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"last_update":"2022-07-18 14:01:23.000000000","reason":"removed on reply"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2022-07-18 13:57:01.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2022-07-18 16:08:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I054a38e3715dac3f014c729fce52c9d9c6bbd2fa","subject":"Revert \"Bump pyroute2 to 0.7.1\"","status":"MERGED","created":"2022-07-18 07:50:21.000000000","updated":"2022-07-19 05:05:29.000000000","submitted":"2022-07-18 16:08:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"revert_of":849790,"submission_id":"850034-pyroute2_0.7.1","meta_rev_id":"f944acd3eae229c64a4c5f2f764a18f14c898dae","_number":850034,"virtual_id_number":850034,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-18 16:08:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2022-07-18 13:57:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2022-07-18 14:01:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":1,"date":"2022-07-18 07:58:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2022-07-18 14:01:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-18 07:50:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"},{"updated":"2022-07-18 07:50:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2022-07-18 07:50:21.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-18 07:50:41.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-18 07:58:35.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2022-07-19 05:05:29.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"CC"}],"messages":[{"id":"8ba1e7fe78644daeadc3cb19a29f97eb0e090690","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-18 07:50:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5443cfe0a7e42fd91936bfd0924c0c1902046413","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 07:50:38.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 850034,1","accounts_in_message":[],"_revision_number":1},{"id":"ab977a8f038e3d7c0bf4a7fd348cf769992fe63f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 07:50:38.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Error merging gerrit/openstack/requirements for 850034,1","accounts_in_message":[],"_revision_number":1},{"id":"d6ee52e136b684a1e15780d6c411b8b2f10363d9","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2022-07-18 07:50:41.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"278f3ff73c50173f469844ae5b9b30adc6f9273b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-07-18 07:53:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b75fc2ae7e5621e12a09e3c90bfd850e3911c1e4","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2022-07-18 07:58:35.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a8c91d1a695ad59ed3a2dadbac08f91a48810085","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 08:07:09.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb60c5519e94239a9d357f062a02606\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/b9d2901ed72944dd9931c9668e00ba2a : FAILURE in 11m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5f7144c1c940684e8f9ee5e8fe83774e939f8385","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 09:38:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d8312a5430e4ebf9a6348aafd26ef9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52a90db8884f4789bb24adf41e536997 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1f0c4164bdb43a592224612703c442b : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7a93b5d5724473187ea5b434001771c : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/caf806df3c4d44888e2da192c987ed93 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ad5f826fc7f4ab58b0779ff1d575d22 : SUCCESS in 4m 39s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8c67847a9f84439d849f3d8251e018c1 : SUCCESS in 3m 47s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/72fdc7b2a07c434db0c7bd60c13e9e87 : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35cc77210f0a456781ce4002ed00cd25 : SUCCESS in 8m 48s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/824e34b341284fa79baf14ee82ec9961 : SUCCESS in 12m 07s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/adfd9b7ce2554b55aff416f4bab2d7c1 : SUCCESS in 7m 29s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/9a284407073e48d48388c43a010756c7 : SUCCESS in 5m 44s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/13c9408dd8964177b5617ad68043c629 : SUCCESS in 5m 28s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/841ebbe10bc242cd9eb2dd4a5cc26566 : SUCCESS in 11m 34s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/1df59800a6a44c03aece3c608dd74d69 : SUCCESS in 4m 53s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/12cf208db45542bc9c9cf5842c41835f : SUCCESS in 6m 02s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/7339501e87b043c89517c7e4e1cce8b4 : SUCCESS in 7m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/611cc709daa74bffab46ecc71be13b89 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/adbc276bdac84e5e8e79f1536fe1a09d : SUCCESS in 8m 59s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/ca33edebfcac468885252ec65677c241 : SUCCESS in 8m 47s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/937509c5a4324c0c9f9914093f1ab5c4 : SUCCESS in 11m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/92747065793f49a1bb6987278bf6ba0d : SUCCESS in 4m 22s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/a5d8516abe8f410ba2b2199307f6dcd8 : SUCCESS in 14m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9c8c874252b3447ba08f1554dd218b5d : SUCCESS in 6m 01s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/abf0b19dc3a24808a474b500d42d8c10 : SUCCESS in 21m 50s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c8ed354f87f942639c2c312abb38d1dd : SUCCESS in 10m 25s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/c94e2363556049a49b1882cce2eaaaf0 : SUCCESS in 5m 53s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/6bb1a5aa09154751b1b272d3a9f66692 : SUCCESS in 4m 01s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fa78e71b6c6d4648bf2a13631bf2aa12 : SUCCESS in 19m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/a55de48b9fc84d6ca70d22f42a7bb122 : SUCCESS in 9m 10s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/df7cd06fbf5a4c53a73fea12f0587daf : SUCCESS in 5m 08s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/13e87588bd3f4960867b9e53ceb3cc1a : SUCCESS in 7m 39s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/39903b0da624429c90a4e246ffcf789a : SUCCESS in 12m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02f15de61078495ebc47ccc464a49a0a : SUCCESS in 43m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/64f70622410b4515bed9a88e393f6266 : SUCCESS in 1h 42m 42s","accounts_in_message":[],"_revision_number":2},{"id":"5e35ace8f968d81cda192c08cee3f738e7add175","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-18 13:57:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60ed220aa6b173e32c6c4f53c228aefbf0907e50","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-18 14:01:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b5d6369de5fcf5aa454070afdf66563f896e27a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 14:02:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b8a846f040b104407c2b257879feed18273e764","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2022-07-18 14:19:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3384a2c760de147332adb4fc8d5533c134bd98eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:08:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/931bc0d102e1475ab0992c49303db556\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f77fbd01ab644416ab34328813d359b6 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba167aff2827435abab6ebbcc164944f : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/316ccfb54f8c453dbd668e5e770ab07a : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/abb37ef685af438cb12d176d0e63aa58 : SUCCESS in 4m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/95fda15fef2840f6a6f8918336cd1a97 : SUCCESS in 4m 31s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ce2ed9f2ebc04459ba76e7143edef27e : SUCCESS in 4m 26s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/9fe34c5b221f4bbc918b578ae8ed1ce5 : SUCCESS in 7m 36s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/6321c1f6ac414c16b2de828a069029c4 : SUCCESS in 13m 26s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/b394cba56ea34749a08c98a8e3b07cc5 : SUCCESS in 11m 55s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/aefc49ac08354876831bf0893a272332 : SUCCESS in 4m 45s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cb987df635eb4457a1972b7f5fe8cdb0 : SUCCESS in 6m 26s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/659e333ae351425cac1cfaa8716a44b0 : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/eb413ee075dc4648b3e77b4d45f374cd : SUCCESS in 11m 14s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/45f7beec586049abacf0c1f67475524c : SUCCESS in 9m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/2d9cb8b00aa64dd4b6a634a2fd145582 : SUCCESS in 9m 31s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/798ea3afca8a496894ad70f01a060ea4 : SUCCESS in 12m 43s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/b80f031bb6cb45b5962f3758e4da7866 : SUCCESS in 4m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/99632e3cfb2e4cc49af5dd1bfbdd1cdb : SUCCESS in 10m 18s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/4f00244334c5421599aa2a70c882b598 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/23402139add9427ea2a8e21d60ce7674 : SUCCESS in 21m 15s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dc9f86d3eb514901b78c5a7669d897fa : SUCCESS in 12m 37s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/88abe2b047c441168683ae2deaca3516 : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/20a619cc05d1471b842cc47ee3d7bc1c : SUCCESS in 4m 38s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2d0d291c2ca146a1bdfc55c46f6592f1 : SUCCESS in 17m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/ea01d19e6c54486baf19eaa315e267bb : SUCCESS in 6m 15s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/7b96083b67a04849b2540eb749e044f9 : SUCCESS in 5m 43s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ddb1a923bd2342cdbbf4775bcd5d0fed : SUCCESS in 7m 29s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/05e63b8d739e4510afac528472ce72cd : SUCCESS in 9m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0cd5cb848034560bd01ce75eae76506 : SUCCESS in 56m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a360050bb2646ff9dc4ca0081e18c4a : SUCCESS in 1h 30m 42s","accounts_in_message":[],"_revision_number":2},{"id":"9d826b81c6b05fa6b7e89454e008b4a357e503c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:08:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f792830032fc5d6189ab1b002ed523bd2577f486","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-18 16:10:45.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/239aedfac764407ea374004f202fb526\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7224957c58479485309deaac0b9699 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2},{"id":"f944acd3eae229c64a4c5f2f764a18f14c898dae","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2022-07-19 05:05:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a7ee57dd51388aeb74a96f09b3f00be4e8b02c2a","revisions":{"47281f102659d100462a6e2e8bf7df915318b10e":{"kind":"REWORK","_number":1,"created":"2022-07-18 07:50:21.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/850034/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/850034/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/850034/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/850034/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/850034/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/850034/1"}}},"commit":{"parents":[{"commit":"fadd51c9b88cb518dafd6990f217e2c829178b25","subject":"Bump pyroute2 to 0.7.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/fadd51c9b88cb518dafd6990f217e2c829178b25"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-07-18 07:50:21.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-07-18 07:50:21.000000000","tz":0},"subject":"Revert \"Bump pyroute2 to 0.7.1\"","message":"Revert \"Bump pyroute2 to 0.7.1\"\n\nThis reverts commit fadd51c9b88cb518dafd6990f217e2c829178b25.\n\nReason for revert: https://bugs.launchpad.net/neutron/+bug/1981963\n\nCan be re updated once the issue is cleared.\nRelated-Bug: #1981963\n\nChange-Id: I054a38e3715dac3f014c729fce52c9d9c6bbd2fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47281f102659d100462a6e2e8bf7df915318b10e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/47281f102659d100462a6e2e8bf7df915318b10e"}]},"branch":"refs/heads/master"},"a7ee57dd51388aeb74a96f09b3f00be4e8b02c2a":{"kind":"REWORK","_number":2,"created":"2022-07-18 07:53:36.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/34/850034/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/34/850034/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/850034/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/850034/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/34/850034/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/34/850034/2"}}},"commit":{"parents":[{"commit":"54be253b51ef9fc4d9ada40d35b4cb158b73d0e8","subject":"update constraint for keystonemiddleware to new release 10.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/54be253b51ef9fc4d9ada40d35b4cb158b73d0e8"}]}],"author":{"name":"yatin","email":"ykarel@redhat.com","date":"2022-07-18 07:50:21.000000000","tz":0},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2022-07-18 07:53:14.000000000","tz":330},"subject":"Revert \"Bump pyroute2 to 0.7.1\"","message":"Revert \"Bump pyroute2 to 0.7.1\"\n\nThis reverts commit fadd51c9b88cb518dafd6990f217e2c829178b25.\n\nReason for revert: https://bugs.launchpad.net/neutron/+bug/1981963\n\nCan be re updated once the issue is cleared.\n\nRelated-Bug: #1981963\nChange-Id: I054a38e3715dac3f014c729fce52c9d9c6bbd2fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7ee57dd51388aeb74a96f09b3f00be4e8b02c2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7ee57dd51388aeb74a96f09b3f00be4e8b02c2a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
