)]}'
{"id":"openstack%2Frequirements~850545","triplet_id":"openstack%2Frequirements~stable%2Fyoga~I65b7709380095e3891beaf62219303f00382d8e1","project":"openstack/requirements","branch":"stable/yoga","topic":"new-release","hashtags":[],"change_id":"I65b7709380095e3891beaf62219303f00382d8e1","subject":"update constraint for os-brick to new release 5.2.2","status":"MERGED","created":"2022-07-20 16:46:28.000000000","updated":"2022-07-21 16:43:24.000000000","submitted":"2022-07-21 16:43:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850545-new-release","meta_rev_id":"1ec5c1ad8874437a6ab8e5468c5dafe6b7732e6b","_number":850545,"virtual_id_number":850545,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-21 16:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-21 15:19:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-21 09:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-21 15:19:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-20 17:04:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-20 18:31:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-21 09:41:38.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-21 15:19:53.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"aaf04189529639632d861c871d43e2be3944d38e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-20 16:46:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9a3d81968a0c03cdf347db71262f99eb65281e1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 17:04:42.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37fa7f85ef5249b3bb1a8c16e995e397\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/308e54e8fcd24e13890098857fd4825f : FAILURE in 16m 23s","accounts_in_message":[],"_revision_number":1},{"id":"a53939c00332a9157a277315c47862a6ff1f28a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-20 18:31:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b8ac107338445f86724ec46df03a7a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c09bbb362524718892c70c0654976d3 : SUCCESS in 4m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/536156054acd4bb1a2f6c79216003f88 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b11c915fd3df446da56f4ba7136cd1b8 : SUCCESS in 4m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6ff42ca9db9e4359bad2160c1a6e47f3 : SUCCESS in 4m 04s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/e56041037c6e482aa396eaedc71d63e5 : SUCCESS in 4m 31s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/10a743c406ba48e9844ead9ae87c4a1b : SUCCESS in 9m 44s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/35790196a7c74990a4504d07f5cad560 : SUCCESS in 10m 01s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/bd4df26da05e475892bff991c3ddf658 : SUCCESS in 5m 49s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/1b2cc7a8bc0f4aa6835ad307091261f2 : SUCCESS in 11m 30s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/0aea5f4b6ff041b59903d64e1e914766 : SUCCESS in 5m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/8670e16ced5645b0b9a039059e708e04 : SUCCESS in 5m 18s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2f397c9def22460fa90b73873147814b : SUCCESS in 10m 18s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/17e54c3dcbc9490bab09ee462d45b5db : SUCCESS in 8m 21s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/593ff34cfef24a5797932e7865ce96f5 : SUCCESS in 8m 56s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/34fdf1a5ec48425688ba13813a76999d : SUCCESS in 11m 22s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/59763b0ffe2e4df19f40e54a100ab9ad : SUCCESS in 4m 43s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/cdd8d28c15334da28d09b24c976aea4d : SUCCESS in 9m 59s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/cace7aabf9044607a44559b7bc6518b9 : SUCCESS in 5m 34s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/502e39a7a3684d4888e48ce2c9e95992 : SUCCESS in 24m 06s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/b18216ab3f0848389c547d362c45d7e8 : SUCCESS in 13m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/78f194c72c03405694c3b705bee32d90 : SUCCESS in 4m 52s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/af631abb4d524752a2ff17b744f35204 : SUCCESS in 3m 49s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/4b0dbab165a64539b61cfe90d39a23dc : SUCCESS in 20m 41s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/300458dc9a1c495384302a2c5333cf8d : SUCCESS in 6m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c4e67c03ad2141eeb38d8e2c1a59f782 : SUCCESS in 4m 45s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/dae251ac1ce5439b91d5729682e82762 : SUCCESS in 7m 25s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/05a244a99f854751a2837b4b5defe5d5 : SUCCESS in 12m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/87104e4ccaa3411b89a58918eca4137f : SUCCESS in 58m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/da0a0fcffd87488db4651e628b1aa04d : SUCCESS in 1h 38m 41s","accounts_in_message":[],"_revision_number":1},{"id":"3b581e244d6b81be251b5a278eead0356e1f7fbd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-21 09:41:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aaafee6e3e10efb838cb78aff2249f4a5601ac5a","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-21 15:19:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b9974850c45105d185d1299afd3e202167da6bc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:20:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8250d543c81c27603212d974511d2927c2e53ff8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 16:43:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9c5ee52c414bd3a209234d9975eb67\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70817b6fdf44460b53cbbab74bba60d : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d41167bdb6c4061905257ac277395e5 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b48f9393f394027ba915e52ab95aa57 : SUCCESS in 6m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c1425138f62b45c7b6ee97cafe97d175 : SUCCESS in 5m 38s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/af32ce9b40464dafbbb9749e841d3c37 : SUCCESS in 5m 18s\n- requirements-tox-py36-check-uc https://zuul.opendev.org/t/openstack/build/bf5677a8aae2437db0d3f9e6fc27bd7a : SUCCESS in 10m 36s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8645925988344e339575146f4ed1f6b7 : SUCCESS in 10m 07s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/5761d50104ec43489359f517d9fda0f6 : SUCCESS in 11m 19s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/c46da3c5bdb34c698f0b655d32eea6bf : SUCCESS in 6m 11s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/364d00f56d054c59bf810e22d562cf09 : SUCCESS in 9m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/2140db4d002f4dc9ae16562682f22801 : SUCCESS in 8m 54s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/6d25f6f2aa78404aa804648ef87f4312 : SUCCESS in 13m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/3da592348b9f4e1693adc00305860a23 : SUCCESS in 11m 06s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d38623dc911f4433a1274e6634cd9ccb : SUCCESS in 11m 19s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d71b90a97a154818a8e0e79b1a6aabaf : SUCCESS in 5m 33s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8f5bc2fd045745ccad431a20c4a4407a : SUCCESS in 12m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/ee695088f35b4fa7a0eae4c2c872099f : SUCCESS in 6m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7b327a82db18425ab7a5b9e7ca3ec45c : SUCCESS in 24m 16s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3ca7a52af32646f6a9fe00ed5c126050 : SUCCESS in 13m 33s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/49e84b4629704a9ca5c59fc8c9809c94 : SUCCESS in 9m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/44a038951e304ae58f4e0c13a7958aec : SUCCESS in 4m 23s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/31d8190334664b548b928f6e06173bda : SUCCESS in 19m 58s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/0783adf8a9b44f93a1a25ce5ca8656f3 : SUCCESS in 7m 12s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/e4bd5ac0b7e04b9295ca14ab25f3722a : SUCCESS in 7m 35s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5f0329bf9b9f4090b16f92022c84c607 : SUCCESS in 8m 00s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e427792efec44c81bee8c5ffb738aeaf : SUCCESS in 14m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/91df622d7eee4e51adc01ae08e85adae : SUCCESS in 53m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ff3b579be01b485b9df7b3635ff90eb9 : SUCCESS in 1h 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"1ec5c1ad8874437a6ab8e5468c5dafe6b7732e6b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 16:43:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec73045294da8baa881626604475f558f0052190","revisions":{"ec73045294da8baa881626604475f558f0052190":{"kind":"REWORK","_number":1,"created":"2022-07-20 16:46:28.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/45/850545/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/45/850545/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/850545/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/850545/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/45/850545/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/45/850545/1"}}},"commit":{"parents":[{"commit":"c70f9032bb885e861656605375324b8d016f23d0","subject":"update constraint for os-brick to new release 5.2.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c70f9032bb885e861656605375324b8d016f23d0"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-20 16:46:27.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-20 16:46:27.000000000","tz":0},"subject":"update constraint for os-brick to new release 5.2.2","message":"update constraint for os-brick to new release 5.2.2\n\nmeta: version: 5.2.2\nmeta: diff-start: -\nmeta: series: yoga\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Commit: Rajat Dhasmana \u003crajatdhasmana@gmail.com\u003e\nmeta: release:Change-Id: Idc41bf3d35dda3b68de097a8e21682ae6dd51364\nmeta: release:Code-Review+2: Hervé Beraud \u003cherveberaud.pro@gmail.com\u003e\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Workflow+1: Elod Illes \u003celod.illes@est.tech\u003e\nChange-Id: I65b7709380095e3891beaf62219303f00382d8e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec73045294da8baa881626604475f558f0052190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ec73045294da8baa881626604475f558f0052190"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
