)]}'
{"id":"openstack%2Frequirements~850615","triplet_id":"openstack%2Frequirements~master~If2a4b34928b464055311570189491bf7671fc6f0","project":"openstack/requirements","branch":"master","topic":"new-release","hashtags":[],"change_id":"If2a4b34928b464055311570189491bf7671fc6f0","subject":"update constraint for python-manilaclient to new release 4.0.0","status":"MERGED","created":"2022-07-21 09:01:48.000000000","updated":"2022-07-23 05:34:48.000000000","submitted":"2022-07-23 05:33:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"850615-new-release","meta_rev_id":"7c32a625ec3c4eaec836f362da800a0b0805b989","_number":850615,"virtual_id_number":850615,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-07-23 05:33:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-07-23 02:29:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":2,"date":"2022-07-21 09:41:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-07-23 02:29:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-21 09:17:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-21 09:41:23.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2022-07-21 10:51:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-07-23 02:29:54.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"9bb8930e62563ab7a043607e8e65829b55196313","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-07-21 09:01:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a8fe944d86f9510c1b4cd4380afc9789776d94b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 09:17:35.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fb3c8d4040549f18b3b47143ecd62a6\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/0d7285a20c1e4b10858d12452329334b : FAILURE in 14m 08s","accounts_in_message":[],"_revision_number":1},{"id":"e33c9d57ae12c8c64d44ad0d4e361044877ef6fd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2022-07-21 09:41:23.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c2a2fac7bb43b5db49bf4b06e5909af8350837cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 10:51:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb19b2448f0d432495ba81d32201f843\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4ca0554b220423f845944ece9c68b50 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/018aa59ef2454731acdd19c04a75b1a4 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4660e03cb4d54d7b9b818c8cc611b940 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c867e8ab6e8b4e458b7f3e16c98c7172 : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/289683cf75cc451e9a97555f1c060054 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f868917587204464925e2ecabf2c40d0 : SUCCESS in 4m 28s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6f62b6cff28a4cb284b0a3450e63aafc : SUCCESS in 6m 07s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d12f7ad4854e4702bf764029b8121bfd : SUCCESS in 9m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/cfac34aab16343b58781aa6de7650c72 : SUCCESS in 15m 59s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/d68c5f8d8ce9495f8a007749215d9793 : SUCCESS in 7m 19s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/ffb78b8e71e546ffb872972d384c8023 : SUCCESS in 7m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3084f602fca3450cbeb06dd71219f0d8 : TIMED_OUT in 40m 55s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/adb10e16e19c4c1c893e186bc6e704a2 : SUCCESS in 11m 39s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/7964ccdf68434f01a49d0b3bcd182bbc : SUCCESS in 4m 56s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/33c09897fdfe49858758f5955a1500e3 : SUCCESS in 6m 06s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c681c261e1854ed2a75eb4c014c34c2e : SUCCESS in 8m 50s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/0e30e3c8acb04290bb5a581f9312bcf6 : SUCCESS in 10m 31s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/d40f0fa8694d49d5ae2c4df1cbc0928c : SUCCESS in 8m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d5197bf777234092a25409a8a9a6835a : SUCCESS in 9m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/6ae2cbbbd9b84c9ca416fbd984f05033 : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2421e5b77e6a44bf9792a9b4db6880bf : SUCCESS in 4m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/735d915d19bc4296be17c18e024de4f1 : SUCCESS in 9m 44s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/958c21d4daa5489d815f60b66d55ab55 : SUCCESS in 5m 04s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/252b7e074b53457e878ff8b147500bca : SUCCESS in 27m 51s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/a22b9a95a6d6401c899116ccb6bc5ed8 : SUCCESS in 14m 32s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/e630435007a646a38c8af4136d6eecb7 : SUCCESS in 6m 42s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/31a80af9a8474ae7bccb266f7736ba7d : SUCCESS in 3m 44s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/7a089115316745d28ee42dfc1b69036a : SUCCESS in 24m 37s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/7f8bddf76e2c4615a5961a2c6f8cdfb6 : SUCCESS in 5m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5d687e26773648c086c02d9f4d5d239d : SUCCESS in 5m 34s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2f5c5bb7ab7b4314adcbfd4a69d7e562 : SUCCESS in 7m 03s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/8f7f2d77933d4c8693f3ba0c2c445beb : SUCCESS in 10m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00cb6109b6584c9bb49807e74857e658 : SUCCESS in 53m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/36eb7f20e0544e34b3be4b96e86a1033 : SUCCESS in 1h 44m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6c649d2049c8533d4871e8cbcba146106fc31af5","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-07-23 02:29:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fa5ce63114cbda05a46c2fdaeee1bbdf36520549","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 02:52:45.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1240f497e75f4874b53153c49688c319\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/019bd05a845344b59832209af9c0134b : FAILURE in 13m 33s","accounts_in_message":[],"_revision_number":1},{"id":"8da5b48a4e1a5a8a29b645766e2bbda80e39a35b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 04:19:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e97a1055b58456ab677618a87642ae0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fde3835156b444ab15f53562c4160aa : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc4e9f7d35f448c82c81edf1ad09f67 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ded8c23a1b5498bb39f1c7f8d5b78c2 : SUCCESS in 6m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c65c4fc5c889432ab30c054e1ebe0471 : SUCCESS in 5m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7827f3052f649d1afdea487e1bb6de9 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e3c5f77edca54cc891cb8c66a0ab16dd : SUCCESS in 4m 20s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/dc8ea953cb9f4b23bac780f892a879d3 : SUCCESS in 4m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/bf083f81b60046c4a5f628cc91f24a6a : SUCCESS in 10m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/4d899f43b7034754ac3f0865fe1481a0 : SUCCESS in 13m 52s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/6cb8350f5cd14f9bb0042c34c2202a1d : SUCCESS in 8m 09s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/c750e9da2c61450385400b6981803331 : SUCCESS in 6m 11s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/3e456911798c48a08f999bc118480ef4 : SUCCESS in 7m 19s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/20e39a6784cd4468b420337be0682c62 : SUCCESS in 13m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/4b43b214f3f845f6bb3d6970a2f3f4be : SUCCESS in 6m 01s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/90b5e22524014d009d175de25ce214f8 : SUCCESS in 6m 51s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/8f598b6be0f34a6a850d7ed24e8d582e : SUCCESS in 8m 54s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/a1bac13bf7d7450181d5ded444ba68a9 : SUCCESS in 11m 23s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/18f6b1b5f17b47f5ae7c20d041afd29d : SUCCESS in 8m 35s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/d48e19145e2c48e8b49b365ebe601bd5 : SUCCESS in 9m 18s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/cf5e718cfedb48b7bf5cfff09a974c70 : SUCCESS in 10m 00s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/9d4e3b54fad54699b1f8752d46de9d7a : SUCCESS in 4m 31s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/27156e27c9104dfa914739fc30ceb505 : SUCCESS in 9m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6fef6109232e47c5b0215adcd23eabb5 : SUCCESS in 5m 42s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/34ae7e0af71743b8828e306f25cb873b : SUCCESS in 23m 41s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/4e6fefd909f142369b0d441ff17c2f01 : SUCCESS in 11m 23s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/bebb8e6e99dc474cb0fa55cb64df6348 : SUCCESS in 7m 13s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/4859f74f6a954a389c141bb45e8d972d : SUCCESS in 4m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a85fb3555354e23ae7a1201de16507b : SUCCESS in 21m 12s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/93da7b81d38a4c9386681d10249af78f : SUCCESS in 6m 24s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/d51274f6229445988ac9e68b0638d237 : SUCCESS in 6m 36s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2219885c74144ca395cf6325dde27ce0 : SUCCESS in 6m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/670e355a40e4417faef7d48b0cf7f9ec : SUCCESS in 10m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/470129da06114ef88442e8b45027dc03 : SUCCESS in 50m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5c42ce38b2314ec4ac88ca30d9411cc9 : SUCCESS in 1h 34m 23s","accounts_in_message":[],"_revision_number":1},{"id":"cf180460a03f2ed85292cb86a61020383e6a35d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 04:19:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cea5efc25ccaad041c5b88e16390bfffe6b6fddc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 05:33:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5718523751e3407ba8ad0387026bef9b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb21921def642e0961eb2d406b3e2a7 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194decf8565b4f9184d182579dcafeb7 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b804f1e90d84264925981dbf5f3efcd : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5eafe62d0db4c1182e269ee5e912979 : SUCCESS in 4m 28s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dbba5a367f35428bbc67d7c0641c4019 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/44a5324e6ccd41a1b7447b84b0738576 : SUCCESS in 6m 10s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/ebd28b32205842e7a225cf1bcc214314 : SUCCESS in 7m 43s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/dfdbd06674464894ba307ccb8dbadd3f : SUCCESS in 16m 34s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/ef947720f5184689acd95b8615fb12d1 : SUCCESS in 10m 43s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/d1f55e9d093f4366aef5542f37669582 : SUCCESS in 7m 28s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/78f8277075aa4fa1ad688e810192dcf5 : SUCCESS in 5m 17s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c75f16900daf492483adc162bc618682 : SUCCESS in 9m 43s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/49b3e00ad9c64d35b53226a131a89e95 : SUCCESS in 8m 24s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/20fe4b0788494a7d81c7ae4b11d5d309 : SUCCESS in 11m 25s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/376837c7bf734cfe87e822cad02cf45c : SUCCESS in 7m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/40b0b8eb10cb4a0db1bf17760fb1af12 : SUCCESS in 13m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/65700f46e793462fa87c3635c3dbf662 : SUCCESS in 4m 06s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e1349a1bb2f245e0862496871ac01b17 : SUCCESS in 11m 34s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6fb2dbc0677540cf91432cf7917c075f : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/da218d10274e41e7bdc27dc98b514508 : SUCCESS in 24m 52s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/e96b12ccba1846ffa492b46a3750a4eb : SUCCESS in 10m 53s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/72f61285466945faa2a77e7670b717e9 : SUCCESS in 8m 22s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/ca2dfc719561408f869aa1d1416cce93 : SUCCESS in 3m 46s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1a6ee404021a44c9b168a9ae71709b9c : SUCCESS in 19m 28s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/913a31791ba34df7a29d1178488a2c2a : SUCCESS in 5m 58s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/38d0acda061f4fbbb4b4b927128a6ca3 : SUCCESS in 7m 26s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/0d6d2a2195584c64b3bcd0cba9357455 : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5bff1f376fec4fbf96b054a5185b2b4d : SUCCESS in 13m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7f52d451194e401681d42b897471f1bf : SUCCESS in 39m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f9f3303e6e6844b09cc4e76621eaadf5 : SUCCESS in 1h 13m 04s","accounts_in_message":[],"_revision_number":1},{"id":"4cfedb56f1f2b9a98f1035cc73b2fad00dd266fe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 05:33:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c32a625ec3c4eaec836f362da800a0b0805b989","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 05:34:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a213c0be6b184fef9970228f7304d78a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aab1b56156ab4238acd56a2dfd9bd96b : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96a5eeac6ca7ed7a6038fec33fad2bf68c5f7702","revisions":{"96a5eeac6ca7ed7a6038fec33fad2bf68c5f7702":{"kind":"REWORK","_number":1,"created":"2022-07-21 09:01:48.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/15/850615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/15/850615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/850615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/850615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/15/850615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/15/850615/1"}}},"commit":{"parents":[{"commit":"460323e81cf16f93b755a6a25e2c431c0f77e33b","subject":"update constraint for os-brick to new release 6.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/460323e81cf16f93b755a6a25e2c431c0f77e33b"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-21 09:01:47.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-07-21 09:01:47.000000000","tz":0},"subject":"update constraint for python-manilaclient to new release 4.0.0","message":"update constraint for python-manilaclient to new release 4.0.0\n\nmeta: version: 4.0.0\nmeta: diff-start: -\nmeta: series: zed\nmeta: release-type: release\nmeta: pypi: yes\nmeta: first: no\nmeta: release:Author: silvacarloss \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Commit: Carlos Eduardo \u003cces.eduardo98@gmail.com\u003e\nmeta: release:Change-Id: I2f8b04bfd262419d6030baa3cde1cebd8ad14d97\nmeta: release:Code-Review+2: Elod Illes \u003celod.illes@est.tech\u003e\nmeta: release:Code-Review+1: Goutham Pacha Ravi \u003cgouthampravi@gmail.com\u003e\nmeta: release:Code-Review+2: Thierry Carrez \u003cthierry@openstack.org\u003e\nmeta: release:Workflow+1: Thierry Carrez \u003cthierry@openstack.org\u003e\nChange-Id: If2a4b34928b464055311570189491bf7671fc6f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96a5eeac6ca7ed7a6038fec33fad2bf68c5f7702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/96a5eeac6ca7ed7a6038fec33fad2bf68c5f7702"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
