)]}'
{"id":"openstack%2Frequirements~851619","triplet_id":"openstack%2Frequirements~master~Ibf4b005a401c149fe11911471c1e6916639356a3","project":"openstack/requirements","branch":"master","topic":"bug-1869244","hashtags":[],"change_id":"Ibf4b005a401c149fe11911471c1e6916639356a3","subject":"[DNM] Testing Neutron patch","status":"ABANDONED","created":"2022-07-29 23:32:03.000000000","updated":"2024-02-18 09:06:47.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7ace432f7ddb53d1fc4a693c1d7e786511a6cac5","_number":851619,"virtual_id_number":851619,"owner":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-07-30 00:40:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"all":[{"value":-1,"date":"2022-07-29 23:32:35.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-29 23:46:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-07-30 00:40:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f4ba62f00b696c79a02387860f25eb9c34065703","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-07-29 23:32:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5cba6eb30954e0b98c1cca2e90a0559dc55bca41","author":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"date":"2022-07-29 23:32:35.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"2a4d9b4489c830d44e08b61bca9d2d109d5c10f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 23:46:19.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175124416a2940aa8bbb00be3700d55e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/57c2bd92a7e1430da128fcffbd14b6b8 : FAILURE in 12m 29s","accounts_in_message":[],"_revision_number":1},{"id":"cf0e90fea76e97caaaf881d9ed166db97b09fe5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 00:40:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15c6e90ece53417c814305309422ee27\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/444911d6becc44bc9d82aa977ab2af86 : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c00897487b045bdb08047821c7397cd : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9d6eeb957837473eb38e49908e76a418 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/98919d530e2f4a86bf509e0a0dcb8fa0 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6ddcff9b18354e1082db3734e5cd9f69 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0cca441e6fb643168b0ec253553a33b8 : SUCCESS in 4m 00s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/bca6ed1aea654cc89a5ee037a2249c0d : SUCCESS in 4m 09s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/19c52c470cf342eba229fa119c5f06a6 : SUCCESS in 8m 28s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/5564f394943b4fa896559e57f11c665d : SUCCESS in 11m 42s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/be9cd67eec334c8f9b554b9945a873a7 : SUCCESS in 12m 05s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/214e250d677d48dc87cdf60c665d3500 : SUCCESS in 5m 02s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/4cf6230d995a4debb2f16a130d6492fb : SUCCESS in 5m 37s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/6b2aeea744f14bc99be9cd46f3c511db : SUCCESS in 11m 40s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/79686dacae61484e82d8dce0d1b99c43 : SUCCESS in 5m 20s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/77aed5cfdcb84ecf982d303be2abba1a : SUCCESS in 5m 36s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/65e3eedfbc4c43cf83433fc35741d542 : SUCCESS in 10m 19s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/04e36d1a933344af8b1fc2543536bf50 : SUCCESS in 11m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/0e7c738e95074076b28d110713283d76 : SUCCESS in 10m 15s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/41161b65a93041b1b9377e849fae1974 : SUCCESS in 11m 26s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/4cf25ea1f7d7437d928ef8cf75e1bc3a : SUCCESS in 17m 35s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/f2c40b1d7d9e4c5e835522288dbe173a : SUCCESS in 3m 14s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/fe9acf6469db4a28ba3b03b32ebeadeb : SUCCESS in 16m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/5fab83c559a64dda81a9312c1bcc9286 : SUCCESS in 5m 10s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4c5c1b247ff2473d94eb77e0b70ed53f : SUCCESS in 27m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/779052e393ef42379d4579ae89dec309 : SUCCESS in 15m 11s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/03b84e7a0d364bc3b33ee91209acb397 : SUCCESS in 6m 07s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/1652a92bbb874497b76dfca27522fae1 : SUCCESS in 3m 41s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f1bbe1bd0c3c4e6bb492c703f001221d : SUCCESS in 20m 34s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/61a3bd1c39b44295b5069c2b586e75e7 : SUCCESS in 5m 19s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/5f20c2b3a0f1448da66fabd245bee661 : SUCCESS in 5m 37s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/15ef13f5f951469a9b0b64442bfd1a50 : SUCCESS in 9m 41s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4ca2f44d20db4a22a86c8bd2aa8e674f : SUCCESS in 8m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1abe563728834e5e85fe1224bf170b92 : SUCCESS in 51m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7882b1d77d564b33897b550327bc536f : SUCCESS in 1h 07m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7ace432f7ddb53d1fc4a693c1d7e786511a6cac5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-18 09:06:47.000000000","message":"Abandoned\n\nseems this is obsolete","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87dd5522cff9605a11bf14d4a4da3b5abd3e69b3","revisions":{"87dd5522cff9605a11bf14d4a4da3b5abd3e69b3":{"kind":"REWORK","_number":1,"created":"2022-07-29 23:32:03.000000000","uploader":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"ref":"refs/changes/19/851619/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/19/851619/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/851619/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/851619/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/19/851619/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/19/851619/1"}}},"commit":{"parents":[{"commit":"bed46f4b7ab58badcddd45e0666481eb89aedcaf","subject":"Exclude Werkzeug 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bed46f4b7ab58badcddd45e0666481eb89aedcaf"}]}],"author":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-07-29 23:30:51.000000000","tz":-300},"committer":{"name":"Miguel Lavalle","email":"mlavalle@redhat.com","date":"2022-07-29 23:32:02.000000000","tz":-300},"subject":"[DNM] Testing Neutron patch","message":"[DNM] Testing Neutron patch\n\nTesting Neutron patch:\n\nhttps: //review.opendev.org/c/openstack/neutron/+/837780\nChange-Id: Ibf4b005a401c149fe11911471c1e6916639356a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87dd5522cff9605a11bf14d4a4da3b5abd3e69b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/87dd5522cff9605a11bf14d4a4da3b5abd3e69b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{}}}]}
