)]}'
{"id":"openstack%2Frequirements~851632","triplet_id":"openstack%2Frequirements~master~I3677978ce3b7d7c709e65b8e2e353d199c6c7518","project":"openstack/requirements","branch":"master","attention_set":{},"removed_from_attention_set":{"12898":{"account":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"last_update":"2022-12-21 01:56:49.000000000","reason":"Change was abandoned"},"35058":{"account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"last_update":"2022-08-30 06:27:14.000000000","reason":"\u003cGERRIT_ACCOUNT_35058\u003e replied on the change","reason_account":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-12-21 01:56:49.000000000","reason":"Change was abandoned"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-12-21 01:56:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3677978ce3b7d7c709e65b8e2e353d199c6c7518","subject":"remove unicode literal from code","status":"ABANDONED","created":"2022-07-30 07:54:05.000000000","updated":"2022-12-21 01:56:49.000000000","total_comment_count":15,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"0e2800747c2a6af349dd7fa711619442f945005f","_number":851632,"virtual_id_number":851632,"owner":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-08-30 05:25:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":-1,"date":"2022-08-30 04:56:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-30 09:08:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-01 04:45:25.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"CC"},{"updated":"2022-08-18 15:26:34.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2022-08-30 02:59:30.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"CC"},{"updated":"2022-08-30 04:56:28.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"7c8c8fed97956b35328aa1d6ac15e94ec58a75ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-07-30 07:54:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5668cdbd57b7f3aa029aecaa3b01f31d09814a9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 09:08:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c9dabdc5a4840beadcfb4b97442656b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28664efef08e43698648173bdfdae3f6 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54a4eae3a57b4cdabcca68b5195cbdf7 : SUCCESS in 3m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0b0fd5d09d344b859e0e0e22a5d94ea2 : SUCCESS in 3m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7c8d090bce44a8fb6106a3cdf46a9d8 : SUCCESS in 1h 13m 25s","accounts_in_message":[],"_revision_number":1},{"id":"8d2c10067b11f7c4051279370f6462f6be5b8350","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-01 04:45:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bf50fdce4463b7dae706d6e70b9044c0cf5c68b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-18 07:07:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0325d0bd02aa34c75ae0bced2e2fb07ecc46748e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-18 08:39:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b90e38158ae411ba36c9d382a9370df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9957c63a314ca89dafc791660814b6 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8198c018e84f4ce6a78409a5a2451837 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a8840c6fcce442ea6ead2e5f7a6fd8d : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc50e2e72bcf4a3186aea71160a97c8a : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d18a8029f9a74501b8b465ca36773140 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d383e9996bdb4977ab7123bad5ef22f3 : SUCCESS in 4m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fa75810825841bf80747d7c8ce115d6 : SUCCESS in 51m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cb1be82f1044b5ca983bab132d73dca : SUCCESS in 1h 30m 34s","accounts_in_message":[],"_revision_number":2},{"id":"97987771844704ca111119a5734e7eb5f6f37940","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-18 15:26:34.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"252ad88baffe222a9065fed366c0fbb25a1384db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-21 07:02:42.000000000","message":"Uploaded patch set 3: New patch set was added with same tree, parent, and commit message as Patch Set 2.","accounts_in_message":[],"_revision_number":3},{"id":"6e2caf016291e975ad2bf94d2e3be8387ed6a19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-21 08:14:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73befc64807742fc87d11b6b5bab8fd2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/006bd726f828470b878c15dc3e450476 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/025bf730069c463b93ad82d4db285732 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bc42d1b9b3154959a2db9d650e5321bd : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7068a9cc434543b580ef4ddf4d349bbd : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35fc38cd9bbe4603bf972962d102d3bc : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1f8b0a3eb20547918c4fdcd04af5440d : SUCCESS in 3m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/92e3f9409dc348d49bf88a010b656066 : SUCCESS in 37m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9171370a4d4340a3945c784a30b81e6e : SUCCESS in 1h 10m 45s","accounts_in_message":[],"_revision_number":3},{"id":"c1f584a367b8cfd48aa11ff15d5a19308a628b11","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-26 02:27:16.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e250ed3717b4b35487aa7d1cf04a69c510885740","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-27 06:12:39.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"7e2d765882069e55d4ef748308ba9d0ddd6505d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-27 06:16:06.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"18e66cf7c1a2895560be3fabb26adbd55ebdecc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-27 07:35:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d37b8232e2d40b8aac7d39d54f0d9d3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05c491d1dc89427689b0389837bd3687 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2a12542f4f4dc0a406cb47f60fc56a : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18438e40d0db471daa01a48a1b2f0263 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd0505a7fb374e8a9daf035e8b6e143b : SUCCESS in 4m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c2aef63454e46bdad6693203cdb690a : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/017ed45bbda649dc8a18bc871e8cef1e : SUCCESS in 3m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fcdca9d3a15d4a639553d28dc06dd61a : SUCCESS in 38m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/67fbcea94bf540199c76acecbc489446 : SUCCESS in 1h 18m 05s","accounts_in_message":[],"_revision_number":5},{"id":"2c3247f9f5588b0b17b8d37aea5efa6a1fdbd4ae","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2022-08-29 18:39:41.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"edff3565f09006876fbbd52637df52ba00f9bdf6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 02:59:30.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"066a29e8e2ea322f904914ebe03decc68a3590cc","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 03:16:35.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f2e4d8cc58ac637419d26da393aa806bc2bcc0da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 03:40:39.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d6eee14eec5ef2d6fee2e4eb5bd79c143144fffc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 04:12:28.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"8e402b8581388037fed595ee72fd146a96180c74","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 04:56:28.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"3f56bc20706aa15378dbae3168d33f271992427c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-30 05:25:53.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca56d346b96a446c88d7df89c250f17a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51207a435b374b9c9747c8e60b89a442 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04044c691fc24900a9cbefeeb16d0572 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d73f902e499044fcaf80946dbc462609 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be25c6d7810e4fa9b6ef5795c1c02a43 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/694d1285f5ff4b15a9f4f7c2129a46b7 : SUCCESS in 3m 53s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6fd089ad7ae84da497396a53e5d5b8fc : SUCCESS in 3m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fa448ff3237429fb8e61624ea88fafb : SUCCESS in 40m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0a1adc355a74aa5b9070cc276693571 : SUCCESS in 1h 11m 26s","accounts_in_message":[],"_revision_number":7},{"id":"3eef9e441826c3d5ea26c14995eac72b27d28150","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 05:58:02.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"bf521ed1328bdfc7f8a07dd8854f55863a5fad59","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-08-30 06:15:13.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"526c6e0a12abd4efc761cb6a5e91d02ecdf95376","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-30 06:22:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"ef1640c8c1cf50fed019974bf63ba517434555d7","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-08-30 06:27:14.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0e2800747c2a6af349dd7fa711619442f945005f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"date":"2022-12-21 01:56:49.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"dc7a81b3bf1f0bf1f8a133be3b4190e344ac63f7","revisions":{"2d4c76f8cef914413bf080dfa262153f487a6bf1":{"kind":"REWORK","_number":1,"created":"2022-07-30 07:54:05.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/32/851632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/851632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/851632/1"}}},"commit":{"parents":[{"commit":"38f5ef60740e474d1bbe24f0ffffb46f4052ebac","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38f5ef60740e474d1bbe24f0ffffb46f4052ebac"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:28.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:34.000000000","tz":480},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nChange-Id: I3677978ce3b7d7c709e65b8e2e353d199c6c7518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d4c76f8cef914413bf080dfa262153f487a6bf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/2d4c76f8cef914413bf080dfa262153f487a6bf1"}]},"branch":"refs/heads/master"},"9713d29d100c35dcce2d8d33194baae410011f51":{"kind":"REWORK","_number":2,"created":"2022-08-18 07:07:50.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/32/851632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/851632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/851632/2"}}},"commit":{"parents":[{"commit":"38f5ef60740e474d1bbe24f0ffffb46f4052ebac","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38f5ef60740e474d1bbe24f0ffffb46f4052ebac"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:28.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-18 07:07:09.000000000","tz":480},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nChange-Id: I3677978ce3b7d7c709e65b8e2e353d199c6c7518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9713d29d100c35dcce2d8d33194baae410011f51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9713d29d100c35dcce2d8d33194baae410011f51"}]},"branch":"refs/heads/master"},"13751fec005f01e26f734c041d5d34472f4af437":{"kind":"NO_CHANGE","_number":3,"created":"2022-08-21 07:02:42.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/32/851632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/851632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/851632/3"}}},"commit":{"parents":[{"commit":"38f5ef60740e474d1bbe24f0ffffb46f4052ebac","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38f5ef60740e474d1bbe24f0ffffb46f4052ebac"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:28.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-21 07:00:41.000000000","tz":480},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nChange-Id: I3677978ce3b7d7c709e65b8e2e353d199c6c7518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13751fec005f01e26f734c041d5d34472f4af437"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/13751fec005f01e26f734c041d5d34472f4af437"}]},"branch":"refs/heads/master"},"55e976f40f33284c3d4b7790d0abef93d0b33fd5":{"kind":"REWORK","_number":4,"created":"2022-08-27 06:12:39.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/32/851632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/851632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/851632/4"}}},"commit":{"parents":[{"commit":"38f5ef60740e474d1bbe24f0ffffb46f4052ebac","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38f5ef60740e474d1bbe24f0ffffb46f4052ebac"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:28.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-27 06:11:49.000000000","tz":480},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nChange-Id: I3677978ce3b7d7c709e65b8e2e353d199c6c7518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e976f40f33284c3d4b7790d0abef93d0b33fd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/55e976f40f33284c3d4b7790d0abef93d0b33fd5"}]},"branch":"refs/heads/master"},"a7c27fba03e10b3ebef6156d79eb2850c4e2efed":{"kind":"REWORK","_number":5,"created":"2022-08-27 06:16:06.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/32/851632/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/851632/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/851632/5"}}},"commit":{"parents":[{"commit":"38f5ef60740e474d1bbe24f0ffffb46f4052ebac","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38f5ef60740e474d1bbe24f0ffffb46f4052ebac"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:28.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-27 06:15:45.000000000","tz":480},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nChange-Id: I3677978ce3b7d7c709e65b8e2e353d199c6c7518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7c27fba03e10b3ebef6156d79eb2850c4e2efed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/a7c27fba03e10b3ebef6156d79eb2850c4e2efed"}]},"branch":"refs/heads/master"},"ccccc303fec3ff71d5273223499c38582ca355c3":{"kind":"REWORK","_number":6,"created":"2022-08-30 03:40:39.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/32/851632/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/851632/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/851632/6"}}},"commit":{"parents":[{"commit":"38f5ef60740e474d1bbe24f0ffffb46f4052ebac","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38f5ef60740e474d1bbe24f0ffffb46f4052ebac"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:28.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-30 03:39:58.000000000","tz":480},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nChange-Id: I3677978ce3b7d7c709e65b8e2e353d199c6c7518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccccc303fec3ff71d5273223499c38582ca355c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/ccccc303fec3ff71d5273223499c38582ca355c3"}]},"branch":"refs/heads/master"},"dc7a81b3bf1f0bf1f8a133be3b4190e344ac63f7":{"kind":"REWORK","_number":7,"created":"2022-08-30 04:12:28.000000000","uploader":{"_account_id":35058,"name":"Ke Niu","email":"niuke@inspur.com","username":"niuke"},"ref":"refs/changes/32/851632/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/32/851632/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/32/851632/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/32/851632/7"}}},"commit":{"parents":[{"commit":"38f5ef60740e474d1bbe24f0ffffb46f4052ebac","subject":"Use TOX_CONSTRAINTS_FILE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/38f5ef60740e474d1bbe24f0ffffb46f4052ebac"}]}],"author":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-07-30 07:53:28.000000000","tz":480},"committer":{"name":"niuke","email":"niuke19970315@163.com","date":"2022-08-30 04:11:10.000000000","tz":480},"subject":"remove unicode literal from code","message":"remove unicode literal from code\n\nAll strings are considered as unicode string from Python 3.\n\nThis patch drops the explicit unicode literal (u\u0027...\u0027)\nappearances from the unicode strings.\n\nChange-Id: I3677978ce3b7d7c709e65b8e2e353d199c6c7518\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7a81b3bf1f0bf1f8a133be3b4190e344ac63f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/dc7a81b3bf1f0bf1f8a133be3b4190e344ac63f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
