)]}'
{"id":"openstack%2Frequirements~851785","triplet_id":"openstack%2Frequirements~master~I6927438d571ad4c9dd5a55545bba5c0becdcb20c","project":"openstack/requirements","branch":"master","topic":"openstack/requirements/constraints/noclob","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-08-06 16:46:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6927438d571ad4c9dd5a55545bba5c0becdcb20c","subject":"Updated from generate-constraints","status":"MERGED","created":"2022-08-02 02:35:04.000000000","updated":"2022-08-06 16:47:19.000000000","submitted":"2022-08-06 16:46:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851785-openstack/requirements/constraints/noclob","meta_rev_id":"37abe5f7fe07382456af68c97c815240a758e292","_number":851785,"virtual_id_number":851785,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-06 16:46:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":2,"date":"2022-08-06 04:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"all":[{"value":1,"date":"2022-08-06 04:56:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-02 02:46:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2022-08-02 04:31:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-04 16:25:32.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-08-06 04:56:19.000000000","updated_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"reviewer":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"state":"REVIEWER"}],"messages":[{"id":"cbff8c2f705a3f858475ff42f7dcd2ec23b73273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-02 02:35:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b0d9826851cb73e279f275d1f3768b78ff5f763","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 02:46:21.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fb0bf4fc6834cf3a691c402fcaa4f71\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/682d0e89adff453db36fcf3aaf86b167 : FAILURE in 7m 28s","accounts_in_message":[],"_revision_number":1},{"id":"ae58842f71068ff608e6e60542599cc5c9bf52d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 04:31:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66ce29a2a2b344baaad309c4f520128b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51a057994f0e450b8619e07cd2679383 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0bdca9f80194a609f386a62bb947f4a : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02f10a74e8094ce3a65d6abbd1d34677 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32fd348262da4888ac6234a6dad8b08f : SUCCESS in 7m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e087bde0762f49549d20372ad2cef79a : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/60fd1aab302d42f59c2acdadfa8af747 : FAILURE in 4m 01s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/6125e382c901432fb47baa347892234b : SUCCESS in 2m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/71354b5f1541475ea92915e2f2086b57 : FAILURE in 3m 15s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ec991f9cf022434aba175b28c4dd1fdf : FAILURE in 3m 52s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/7abb5d1e43be49d182554a8ed5767593 : SUCCESS in 7m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/7ee8ceca1a4042cca2db280b068aff07 : FAILURE in 5m 48s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/99da915d08b04d37a1d7a4e62a99dd5f : SUCCESS in 10m 32s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/26cf22b2606d448e85e019a46d8ffac4 : SUCCESS in 11m 11s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/3d412d4a8efe4fbe983a0f26f33b87e9 : SUCCESS in 5m 29s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/3f25a8ef25f04c3ea770e14ba05a3c58 : SUCCESS in 8m 45s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/d653b3a5ad1246aaaeb51efb5145687e : SUCCESS in 8m 40s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/13e4fd3f22d64ca59f8d3fd59934279d : SUCCESS in 13m 47s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/47959db5499945da941a907ebfd6e6aa : SUCCESS in 7m 46s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/6a3ceda0c34c484a95dcadef831d2625 : FAILURE in 7m 25s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/e5dd6bdba3b44621bf041e2b35a75b7d : SUCCESS in 12m 21s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/32492f962eb0470899fb4145fd02e1b1 : SUCCESS in 4m 11s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/106bbf10c06e444995cd9ec1f7bee69b : SUCCESS in 12m 27s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/24bc722c84a84ed89e14e8baff8cf81e : SUCCESS in 7m 48s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/54a4dfebbd91407f936acb5eb4a6599e : SUCCESS in 31m 02s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/acc2110d56db431dbfe0ea198f666d6c : FAILURE in 11m 40s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/d30ce122b651465791a07228391354c3 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/f15d98f0a7c443b580a61eede1d058df : SUCCESS in 3m 58s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/f2c442fdb1d44bc599c2d08e4c9724f9 : SUCCESS in 20m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/fe69f2ce118c432bbd0dbb59a64fb314 : FAILURE in 7m 43s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/f779a2a9a70040c9821c3b2697dfd44f : SUCCESS in 6m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/edad531a90194dacaffd64db7221ef19 : FAILURE in 5m 10s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9e9bb3995f164dd1bf6ef4bcaf4b3a4a : SUCCESS in 12m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39554e7151764c8d8adf21daa9f55eaf : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8f9ac0fa13c64ab58616c5fb741e0488 : SUCCESS in 1h 51m 30s","accounts_in_message":[],"_revision_number":1},{"id":"9cc6c57e9fa788bf51ec05b6ccca403c5b4bc34c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-03 02:31:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"05c7996a0effa8067ebd122cb191d0588133afe5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 02:43:29.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56fc896a163b473fb7cf076e350172f7\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ef50e8e81335479b9ab691698ecac16c : FAILURE in 8m 10s","accounts_in_message":[],"_revision_number":2},{"id":"dcb41693bee09bc8010047cdb897be9f4c4f94ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 04:10:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26100057379142489aa179de3255d853\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9969de6ee7b54c86ac2b5452a41c9dc3 : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0b0c6908044a9ca069911114d228da : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/36c8919b01e5498db3865db755a59de9 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/953921bed38e47d99a989bdacc058752 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f615248f75df40f58a3b0e36e4696f29 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/14da5bb93e144dd2884cbd8dd2dabf84 : FAILURE in 4m 10s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/63044172f02b4e0da22954d4e9237c22 : SUCCESS in 4m 02s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/60d7d909114542d4bad40f48819f7315 : FAILURE in 4m 50s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/e63e912f8d8544f99b61f53fb8946174 : FAILURE in 4m 52s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a8ec618e58b54f22a079649f1809278d : SUCCESS in 9m 16s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/79990ecf5d5c458483dbaaa7ebdf2592 : FAILURE in 6m 44s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d40ac816279b4dc1809783c35d850e31 : SUCCESS in 9m 01s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/166c2e7a31d4476ba254b2125bf8c996 : SUCCESS in 13m 29s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f1dd91a15a144de09ff010b69949c955 : SUCCESS in 5m 39s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/38697e40325a42fc890c49cc166bb11e : SUCCESS in 6m 39s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/eded82cc670644b7878c8fa54b5c2c29 : SUCCESS in 8m 39s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/986c174695064dd589062bd4ac0d0aef : SUCCESS in 10m 26s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/da606b9bc3704d0b950cdee633742f12 : SUCCESS in 8m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/4d8ad7f8f3a54aa29607d224bbb34c81 : FAILURE in 8m 13s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/45e875dd6be04269a485d51a695579be : SUCCESS in 10m 49s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/2aeb366f1dda4c76a00f013d2204d98e : SUCCESS in 3m 45s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/41d8cfe5fb5140ceb0353d130e70694f : SUCCESS in 8m 48s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/393cd7b3e9ed410fb8324515b662f8d9 : SUCCESS in 4m 30s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/229d145d0510448484fc6b7b902966ff : SUCCESS in 31m 40s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/9b87287d1dea4c95ad098be43e6a6eeb : FAILURE in 15m 39s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/b5285568c22c4de58c360c0b8d595399 : SUCCESS in 5m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/bd029180f8d7461c85968f3cda6321d0 : SUCCESS in 4m 26s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/5cb4db0a2ada44b4a97050650b2076d5 : SUCCESS in 26m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/62d9bd1cad0b4d878ca9333a3e7dcbf0 : SUCCESS in 6m 35s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/24867dccbcea4a158c0f83bc879ab93c : SUCCESS in 5m 46s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/179f7ea05a4d47ebb9c816e80f213261 : FAILURE in 5m 11s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/5f0ecbcf3ec3406792252b448504945b : SUCCESS in 11m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b567cee461e4ff29b8930a4c4e6eef4 : SUCCESS in 54m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7b7be15cf3448f7b792d7ac14b01670 : SUCCESS in 1h 29m 48s","accounts_in_message":[],"_revision_number":2},{"id":"6d1ca5726628ba68466f4ef8d29f114aed0b8ba7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2022-08-04 02:20:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5801a1115abe343d6194493166bf63292b4e0056","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 02:33:47.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae32c666fddb47b2ae563f9e5d8464de\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5e76f63ec9a64a8eb247f90622a89e8c : FAILURE in 8m 23s","accounts_in_message":[],"_revision_number":3},{"id":"2ad030594ce023ad9f9058f04441647a2977e081","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 04:14:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c69ce150dd646cf99a3dda4bc3285ea\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61563ded05a14310878fb27e448f2877 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a31854e12e164ed3b39ce25a5961a876 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46b4458224f141fd84608a75c8af1e62 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a28d15ef292846b7bd9fc39067a918ec : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb36544cec4245ae8f8ef7db2085327a : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8e53497dfe14c418310d8f4c73aabff : FAILURE in 3m 08s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/71816a32f112410a9084adf6d496d8ab : SUCCESS in 3m 55s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/da7829e4574f4cea9e975ee3db419bac : FAILURE in 4m 09s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/b9c819786ec640fbb15c8c8e44d1291a : FAILURE in 5m 02s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/b3e67803896e49688e593f6c5135804b : SUCCESS in 7m 39s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e47a59d74e3b41c09e4789803c589afa : FAILURE in 7m 19s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/d53be570adbd4499b01e4c1af7e256d1 : SUCCESS in 6m 46s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0614bebe476145ce8cd37f16e221c1e1 : SUCCESS in 12m 16s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/47472c049f4048d5b2c50721898a3242 : SUCCESS in 5m 23s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/6b69adbfc8ff42e3856627aa6b83b113 : SUCCESS in 6m 29s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/cbf65c5ce0794b00bffe8f09ea92df82 : SUCCESS in 8m 48s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/61e8b19f4d4a4fb39de74e7b1c537972 : SUCCESS in 10m 40s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/8ba21ce3ca26439099e931e1fff78196 : SUCCESS in 7m 43s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/a3b764e0acd242d08234d112cdde057f : FAILURE in 7m 33s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/ed2961623da2434bb7797f8dca349225 : SUCCESS in 10m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/a1375f0b70c74a86ba33429a75ce6da0 : SUCCESS in 5m 07s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/65b5ed66b8484a45ae559dba2307e883 : SUCCESS in 13m 12s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9577578ac6f142eca657d884d32f7a8a : SUCCESS in 6m 19s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/ed5d5a188a08495799fe297117f96180 : SUCCESS in 26m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/68264f97614d4066a540df3a426cd257 : SUCCESS in 13m 22s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/61b74c41f6fb4104877a24ca09661f48 : SUCCESS in 6m 04s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/fdb4d257d9f3434e9dc76d901284ebc5 : SUCCESS in 4m 18s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/ffb63903ab9f4c8c9398bd5168062211 : SUCCESS in 21m 14s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/377a0c1195d743ceb9558acdcef2caee : SUCCESS in 5m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/37fb58c38faf427c848525a5d0ae0313 : SUCCESS in 5m 22s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ccac8348c61246df9c7cd41cc0f418b1 : FAILURE in 5m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9649701e58394b3c8cecf434495735ab : SUCCESS in 11m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ddfc6e7ea32f46b29ba596e67ae9977f : SUCCESS in 49m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8965b2c14c76456d895a78b5ba1cbb30 : SUCCESS in 1h 37m 03s","accounts_in_message":[],"_revision_number":3},{"id":"78b079a11522de9252bcdc061b2d173348cd333b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-04 06:35:57.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"25721145ef144d13522b4ce8d02a949c0d5a4f82","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 06:46:58.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6cbc6147e05434899cb3c9f2ec2e652\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/5079f909d2634db3ae04b94c3a227136 : FAILURE in 6m 35s","accounts_in_message":[],"_revision_number":4},{"id":"bc110f1cdff55c967f7b0b827eb43193b1180c2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 08:03:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/66d7f8f281b54b80acec36e4b20e0e44\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b312bbc84524411b9d0522f71d1553f : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ca1545247c54ce49d230317f46b7ce5 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9267563ec92942f6b7a4578045b70311 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f60dbd1dc5564521abd77d5eb1f85d8d : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a33ec6c5046544818e5f71b9f7f44ba6 : SUCCESS in 4m 37s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bc47e6717688400ea34776e1cb7005f7 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/c633f7d2d0084b81abc4ff0fb3e95510 : SUCCESS in 3m 35s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6e8cf9f88c904676af5e125ef448ba7a : SUCCESS in 8m 16s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/17861eff76c84008899303abe836d755 : SUCCESS in 11m 20s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/cb804a4d90e14db490f173cdd88c3d2a : SUCCESS in 7m 44s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/5152d051fa364cf085e05b2124238474 : FAILURE in 5m 02s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/56b4ba39999b436e8be809e371f438d5 : SUCCESS in 5m 45s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/933228d580a54a7fa920ccb40909abe2 : SUCCESS in 11m 48s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/068c3335684f48e9a6a4223c96579637 : SUCCESS in 5m 04s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/e34313c996814f719203e49b7bd397d5 : SUCCESS in 5m 28s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c09b8e95ce67488ab56711d111ae9b42 : SUCCESS in 8m 10s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/dbe9654591554c9d98768a701ae08b79 : SUCCESS in 9m 13s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/949a92f49e6a4bcc8e343530cd44ef25 : SUCCESS in 8m 33s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/26e1c94c7949474098f3c74d812cb142 : SUCCESS in 8m 05s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/d9a194794b314128b0b821b92e0c0b7a : SUCCESS in 10m 26s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/8e68e08cb5e7461abcf925f80b2b030b : SUCCESS in 7m 21s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/8eafd276d0e04b3b9f1c6a8c21bacbbb : SUCCESS in 9m 04s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/9f5fdb093585421bbc540ce6dc745ca6 : SUCCESS in 8m 20s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/4313ca10502c45cfb98c023ad676e166 : SUCCESS in 22m 03s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/c306ae8e515e47f697fa47ac07d7eaeb : SUCCESS in 16m 49s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/564ee336d1d64c67bd24225b198b22ff : SUCCESS in 5m 44s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/9955d486ae584bb9ab03f2d176a645d4 : SUCCESS in 6m 52s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/45eb8ff80b184a748289fb7ec668440a : SUCCESS in 18m 55s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/e527d8c075f14eb0a5de46f212297c05 : SUCCESS in 9m 41s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/0d36e520b3684d1fbc9d1c49218b54ac : SUCCESS in 5m 23s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ae3e446f393341b5adb7cc6c8cd25936 : SUCCESS in 10m 28s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/0bfce0ef97264cd596030ba890a4eab4 : SUCCESS in 10m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f24be9a8a2df4e97b73aa5c079c8dad2 : SUCCESS in 1h 03m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4e1740bac6f415f8c4efc35526804f8 : SUCCESS in 1h 22m 59s","accounts_in_message":[],"_revision_number":4},{"id":"f4b4c712ef5d07ff988b4090db7f32a1a8098b20","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-04 16:24:54.000000000","message":"Topic openstack/requirements/constraints removed","accounts_in_message":[],"_revision_number":4},{"id":"c92669488aabb749cad9011ecc2805c946d549b8","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-04 16:25:01.000000000","message":"Topic set to openstack/requirements/constraints/noclob","accounts_in_message":[],"_revision_number":4},{"id":"32dd421a01cb568bee2eaeffd23f8ad4b0d2fd65","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-04 16:25:32.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1f15643c56cdaebe99da65a2dfaa2a0897d234ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-04 16:25:54.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"9f9050cf8b05565ec4725425c013ef9a2dc5bfe2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 16:47:03.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a50b4e62386d4515aec0ae08621238fe\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/ed483e57fbad4ca7bf145b316a0d8fdc : FAILURE in 13m 12s","accounts_in_message":[],"_revision_number":5},{"id":"4adaa85fe3c3439dc4b9d16adfebef9ff754700f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-04 19:43:41.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12fa1f53c4c34082b2922cb679f21f97\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/140bf42f30ea4c7689d0e0e27d73f297 : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad8dc007894b41ff9bd129c53c06e084 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d5e30aa62c542e4b0c687556f88c044 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01dac9d413504de48821422124052233 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23cceb34b7f841b1928da521a45946f8 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5ba2a7cbb40c44dea7b8d730ecf75dc1 : SUCCESS in 3m 59s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/ceeabe38031c4b028c9bd73b10f24473 : SUCCESS in 4m 43s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6c59f05c096a4e0783dc23843c76e572 : SUCCESS in 11m 13s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/8fa0844a4bf04601bd0d9d224a82b23f : SUCCESS in 13m 03s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/a48e3d5e8a42469e90c8210c3a6819c3 : SUCCESS in 8m 00s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/d358c9e767a9419aa5cc1990d932124f : SUCCESS in 6m 22s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2cfcda0bdbb94d3b95295fbf84c5fda9 : SUCCESS in 6m 35s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/82837fc5a22944cdaad6dcca3fbc5e3a : SUCCESS in 12m 00s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/204fd8c6a79a4adc918deec2f68434e3 : SUCCESS in 5m 06s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/7a84c79d374449ad80f3176fcf268203 : SUCCESS in 6m 10s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/f3488c8ff7844860b8c7502f9dc8e490 : SUCCESS in 8m 20s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/3655385aa9c841d7b4c0d1133442666a : SUCCESS in 10m 21s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/3f910ba08730457a84a8463bea9cbfbf : SUCCESS in 9m 40s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/22eb3fe6b3ba4eceaa5cbd4f95b2bdae : SUCCESS in 9m 15s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/0353091cfa5749428506bd2f364f7a42 : SUCCESS in 13m 02s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/019a6be417c24020bc06865cc172843e : SUCCESS in 4m 29s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/d333e20a7cb046d0813606a1bee397ab : SUCCESS in 11m 58s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6c3bcd791132441db3f83b0e24599394 : SUCCESS in 6m 06s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/eaed8184076848f1ab94186b4e96d937 : SUCCESS in 29m 30s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/dc12780421654cdb833d7d0e3f0db88c : SUCCESS in 14m 19s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/3c00a44580f84979afd32ebbde0ef133 : SUCCESS in 6m 50s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/66c2e352f4db4b75bb0bf1b2d539ce96 : SUCCESS in 6m 57s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/2673389f6de04144b4e169749b00fc99 : SUCCESS in 19m 32s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/76dbebcb85684b558bd1af8b17c990a3 : SUCCESS in 10m 31s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/028413a7faa843cd8bd6bd80dcb09cfb : SUCCESS in 5m 21s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/57920e9e6b9b4fd7bf181de68247ad6e : SUCCESS in 6m 20s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/4a64b4ef96a249e088a80d90084f4a2f : SUCCESS in 12m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2c4ce96717254a7897ca5c37e9de4ca3 : SUCCESS in 54m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/71574bf33cb840cea6d82fec5f0aff30 : SUCCESS in 1h 43m 17s","accounts_in_message":[],"_revision_number":5},{"id":"a43ebbb1f9bb937460a2003bbb482d03a80a987e","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-06 04:56:19.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"2f3291ab01e4dfaec6a1b1fe7d4dfb43ac68f673","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 04:57:03.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7308e70eac95fab4b545b67a6cce26e6b7554f75","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 06:36:16.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b34c068604de4e1cba0f33dcc08c0f46\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87f949dc27b34c32bb6c564dc28cb18e : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/41412eeb1abf4d448a95be5b1fff9d41 : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d8d2bf22ba3d4f94926f6bd0a99fcefe : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e993ff8ab02e4502ae8a96c6c750b736 : SUCCESS in 4m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/66733d09fb78425f88449f8bf4be70d9 : SUCCESS in 3m 42s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/d1054ac4c37b4441b8b5be36e9a45efe : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/f4ed0b72674c424ab42c8ab8e73484a3 : SUCCESS in 9m 23s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/7fd98c71b8504e8a9ecd075469442f50 : SUCCESS in 12m 39s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/18cf36703b5a47148623a76edb326b48 : FAILURE in 11m 58s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/88fa3adf5b854260af23eeffc234f99e : SUCCESS in 4m 13s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/ae04425308e7443f8503b713bee26622 : SUCCESS in 5m 14s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/c52e9e11976b49d08891f61f6f0a359b : SUCCESS in 8m 05s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/4989d121f4d54a75814aebc82b64ef5b : SUCCESS in 10m 25s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/9e1c5a99c501496ab6fe641b497e8c69 : SUCCESS in 8m 27s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/7873b4667b4f4648af129d79ab28958a : SUCCESS in 7m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/a95b2fc6e41d4775985ceaed6ab29e4c : SUCCESS in 11m 52s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/46d9a34b14e4446ead49be73358553aa : SUCCESS in 3m 55s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/97f001a559c14f09aa4a53f37a0e892c : SUCCESS in 11m 30s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/c153e0b4a8144bafb5335650b8fcc6ce : SUCCESS in 5m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/559f2948897e48d88dd529eee896c4fd : SUCCESS in 23m 22s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3f02a28741ee490f89e270dc47ef6ee1 : SUCCESS in 12m 59s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/fb3cd0b8b503443199baca04510c5843 : SUCCESS in 6m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/0de0b62619fa45c1968eeeaf79562f25 : SUCCESS in 4m 20s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/1ea12a8a9a4747a69cbaa4852a5bd320 : SUCCESS in 20m 17s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/118bee6bab9c4d0daa3057096e637345 : SUCCESS in 7m 28s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/c0cee8862c19449aa4c0d7348a590403 : SUCCESS in 6m 13s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/5b680d9edd6d48d1b98a53549134a4c3 : SUCCESS in 7m 12s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/fb1ae3b2b3f64e6e9ea2db9529aca683 : SUCCESS in 12m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d059a4aa14b4fd88d15d091e5fa962d : SUCCESS in 51m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d4cb35fd81a4d6eb0a31044deac45ba : SUCCESS in 1h 30m 20s","accounts_in_message":[],"_revision_number":5},{"id":"ff71b90203c079f04e4a98938f8566c13196918a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-08-06 09:16:48.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"85af5aa9f3f1bfbb05321d1f36d896d7f558b0d6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 09:33:27.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88d71a4db5944a01b8bb3ce0e51aac77\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/67f120eceb5f4900b94a7a35a700dce7 : FAILURE in 14m 37s","accounts_in_message":[],"_revision_number":5},{"id":"17574091505b9719f71316b1b63a4de68a11d592","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 10:56:21.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca2c9d2fc374d03be1723b594ecb730\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ac2b3aff753435ea4adfb2c63d4a967 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c7d518e44824122b742a2b51893173d : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c99d416761fa48948a2555a28bf18402 : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d7e67cb98cb4b69b81c9fa1d9209aa5 : SUCCESS in 7m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe450ee6e6be47559896d46758de1bd6 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bd851fb4488a4f7ea732618a29b92af9 : SUCCESS in 5m 52s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/773cd9cbf7134300a0ad21ba69a4875f : SUCCESS in 6m 54s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/6308fcef1d77496b9248b2cc2c0175b0 : SUCCESS in 12m 49s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/ec4f0a5582d94f3fb71f434c8f66e01a : SUCCESS in 14m 32s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/ac75cc9d217849b2a5a1d91128cae845 : SUCCESS in 9m 25s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/e4e74df9c0d3418eab7643210dbc02e0 : SUCCESS in 7m 38s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/2a7d3d8b169e47db9f660c3642649186 : SUCCESS in 9m 04s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/c9fda9e256254009a9ff0ea7fa562454 : SUCCESS in 12m 44s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/f9a7928d3995411c9cfb48bcdb856f31 : SUCCESS in 7m 14s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/cfd1bb407a2f45f8946f421de637f4bd : SUCCESS in 8m 13s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/4e7205af57154d8c8d453931c81a8af1 : SUCCESS in 10m 23s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1f892e78e09c495a916163ceaa0735b7 : SUCCESS in 11m 56s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/1e44406fa21f4a6890aa391b9363e61c : SUCCESS in 10m 57s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/cb16512cc51c45f79114498e6a13b07c : SUCCESS in 9m 59s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/c7cbf96f0e3f4f3db06e24ec907a7087 : SUCCESS in 12m 16s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/d5cb516ffb0548cd8f773ee7ec448bdd : SUCCESS in 7m 44s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/b83aa5f69cee4e319c4a8e5491acc8be : SUCCESS in 11m 54s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/e5456df656ae4b2f893f098bcb57e857 : SUCCESS in 5m 16s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/2d33a82490374921adf2a55d6c9a8cb3 : SUCCESS in 24m 46s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/7aede217d3fa4155aa4a6608fe95b16d : SUCCESS in 12m 18s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/9cecb6cc109f44c6bf5a63b8213ddc65 : SUCCESS in 8m 48s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/27bb189ebf2b48e3afefdb1bbaf8bbac : SUCCESS in 3m 47s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/8163dbd6c9794790b6e6695dba384b75 : SUCCESS in 23m 31s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/98da5bf850ad4c1fbc054f894da168f0 : SUCCESS in 6m 07s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/119ae73b5e5043ceaf1ce10c5ea43996 : SUCCESS in 4m 18s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/ba8afa954d4c4d50b00d8b86a9ec9b50 : SUCCESS in 7m 18s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/e80b1615eb484d759c2c67ef7598c9b6 : SUCCESS in 10m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4964a070a0dc4977b500b22aacb9809c : SUCCESS in 1h 01m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51b144388f5f4371a8212526a6ce8063 : SUCCESS in 1h 34m 06s","accounts_in_message":[],"_revision_number":5},{"id":"a1fbc9661d3fdb65ca58ab16cde7dc836b5fe76b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 10:57:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0a2e3ce8db26ee98f4b7480a6aac7b1cf07628df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 12:17:05.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/96bde9fe9770422590b104d1980386b9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d612a7d16a1249849bcae80011e1d88a : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953c9275c37e4524b0b14fc2102c2c6b : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fc9d2f3905a4d939271b6b75aa90142 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c7bc0aec049445a8ba28dfa757fbd5d : SUCCESS in 4m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1006096586b945e29273013b0a17591c : SUCCESS in 3m 56s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/50621c2f2dd44be284dc0c5d1af44b98 : SUCCESS in 3m 38s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/d8c8bd8a6470470db5df4822bf3de638 : SUCCESS in 7m 58s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/146f85531b74469aab3980c4500ec17a : SUCCESS in 12m 42s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/baf3aad15cc34a9e9a02019b9fc43444 : SUCCESS in 10m 06s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/a50cc6befe324a39b8d3654d432a736e : SUCCESS in 5m 09s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/a5907cf3402e4df6abdd24e937a90902 : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/a17b515a8bd4449f925365f17f6b3af0 : SUCCESS in 7m 35s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/1b54b9beeeab45718938c361ed1e37e4 : SUCCESS in 7m 41s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/ce69178296a447e793338ae00789f1ab : SUCCESS in 8m 00s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/9b1f7e280b9743ffae53de0e3c77285d : SUCCESS in 7m 20s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/2c15b3ae03394aa4ad03e4c05a29c08d : SUCCESS in 11m 29s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/979e06340c494820a1fd1d3f68e41faf : SUCCESS in 3m 49s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/9a39cf6b034844a6bd9d823bcf1f2c80 : SUCCESS in 10m 17s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/884c936655c34afca9b17dca4428cd83 : SUCCESS in 4m 52s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/7cae1f6e0ce048a19303283c8f7694d4 : SUCCESS in 20m 20s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/3dbcf787c4b842358c1e9e9d0888b3ca : SUCCESS in 11m 01s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/46148e02fe0642aa9191c6a6c989c2ff : SUCCESS in 5m 30s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/974b67e8ca2b4cd2b47084d4a032a5d3 : SUCCESS in 3m 39s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/fcc0bf1fb17e416c874b440008b53fa3 : SUCCESS in 20m 01s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/d66154f866774d298cacbceed3e73b20 : SUCCESS in 5m 46s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/992d8bbad77b4408a02f07b1d51bbce0 : FAILURE in 6m 04s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/39be418f14b84c74a4031f0a8d19f458 : SUCCESS in 6m 22s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/9af6ecd6f0f3428683dba460d7316895 : SUCCESS in 13m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02260069f01f423bb259f9b23d8df814 : SUCCESS in 52m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/eca717ac7c1b446a83dac2f7ebea417f : SUCCESS in 1h 19m 01s","accounts_in_message":[],"_revision_number":5},{"id":"961f4d201a8fa88cfb73fab3f6222a15419ebab2","author":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"},"date":"2022-08-06 13:31:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"09b467d5375584298396cf52a462dd0031c82a58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 13:46:50.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07fa1129f4fc45bb926fe94318b0f30e\n\n- requirements-tox-py38-check-uc-aarch64 https://zuul.opendev.org/t/openstack/build/d41b22cadd4043d0ad69232d2724a856 : FAILURE in 12m 58s","accounts_in_message":[],"_revision_number":5},{"id":"8106bbedbe5baf2a5da830e4fe5a4f600a07d438","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 15:25:59.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6431445c03c345e481fd321bb4a4358d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74bb75ea27334f2cb450ba5af24b9548 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c0c45d4e6cf47199eaa42c55f0ce5fe : SUCCESS in 5m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/805e744d249341e5aa5689ba46fb9940 : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4495ee7d568743f287509397d7f892d4 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7477b4d7872845d98a22ce103ea10efe : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/418b9539935b46ed98ab0dfff25df159 : SUCCESS in 6m 39s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/33eb78dd691e4f79a391507f29b75a82 : SUCCESS in 6m 20s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/7d1a9a3496e44eb9ab8041097ec70617 : SUCCESS in 11m 40s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/9dea80f6387d484ca94d673ef0a19da7 : SUCCESS in 13m 31s\n- cross-aodh-py38 https://zuul.opendev.org/t/openstack/build/264969c2e6eb40b6b9137efd0064cd1f : SUCCESS in 9m 54s\n- cross-barbican-py38 https://zuul.opendev.org/t/openstack/build/b5d47fc4097242298469864253190b28 : SUCCESS in 7m 56s\n- cross-ceilometer-py38 https://zuul.opendev.org/t/openstack/build/7e6a3aa3ba854a378fa55ea73880b423 : SUCCESS in 6m 09s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/0424da77a8ed4ff88d2e027217ec8057 : SUCCESS in 11m 54s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/cd6822485c0d48fd9bfb5bfb8e86e759 : SUCCESS in 8m 17s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/bd3d8d6931204b38b80040a5cf522610 : SUCCESS in 8m 16s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/9bd2822698ea4eb5b2ec119f021a363c : SUCCESS in 11m 00s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/bda7ce1a09c745d0861d175afc13af44 : SUCCESS in 11m 59s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/61e2fa5907084586a453e83a36dcf719 : SUCCESS in 11m 23s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/74c0d507446a4212826c71418648e4d9 : SUCCESS in 9m 34s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/858fd1abc7d445d991a7eff205a3b365 : SUCCESS in 12m 25s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/1f95b86bc355438e8cab03a8a62b3cbc : SUCCESS in 7m 05s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/c2ff74dc4f6f4cb5b705a68a964da155 : SUCCESS in 11m 46s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/6ea4bf67cfb14fd499d4bb1d96bb8c24 : SUCCESS in 7m 59s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/78fb27a78c8f486086687ff8ae8a7e12 : SUCCESS in 26m 45s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/8126303d90914f82aea60beded71794d : SUCCESS in 14m 00s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/489cb48cc4874eb38cd0b25d3657b1d8 : SUCCESS in 8m 15s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/57f13a3f5bf34c699594b3379950d655 : SUCCESS in 6m 15s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/394b1034fd164d8eb99fc7a30eaebbc1 : SUCCESS in 21m 30s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/f60ddb6b986c40dd966f38820623e0ab : SUCCESS in 9m 22s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/8d50fcbdcc644206989e47d153278023 : SUCCESS in 8m 01s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/b633ba8b8edb40d6a3f1d48fcdd132e5 : SUCCESS in 9m 07s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/c12616f959d642778aa5641466be71fb : SUCCESS in 11m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c29a150384b74be7bafd9ecfeb4cb39e : SUCCESS in 57m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8abb1585a674f18a3ca14ef939c7a24 : SUCCESS in 1h 39m 40s","accounts_in_message":[],"_revision_number":5},{"id":"fb78cc070fd4be2797faabbd8c51cf7606eae812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 15:26:42.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2aea71fa4663242be15aaf566629e32ed76af66f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 16:46:23.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4823bdab6fd949089ef0076e75626add\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df268c9350be4629b2d8130e162621c0 : SUCCESS in 4m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede757b98fa84bdc86d0c48e32ada834 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38f07c48a7bb4a3f9154078f46ebc5f4 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06397f655f224897a1f56243c30afb91 : SUCCESS in 4m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6e30923255fc41d3aafd5ea0e2d7e49a : SUCCESS in 4m 14s\n- requirements-tox-babel https://zuul.opendev.org/t/openstack/build/342ea1fdef3946ea928baab38ff48f69 : SUCCESS in 3m 57s\n- requirements-tox-py38-check-uc https://zuul.opendev.org/t/openstack/build/8178cebd4fc2418283bdcbe67dd2bf80 : SUCCESS in 8m 34s\n- requirements-tox-py39-check-uc https://zuul.opendev.org/t/openstack/build/3ef596708f7247be92c921a7df6f5344 : SUCCESS in 12m 31s\n- cross-cinder-py38 https://zuul.opendev.org/t/openstack/build/588d546820984234bf3af0a696600f99 : SUCCESS in 11m 07s\n- cross-designate-py38 https://zuul.opendev.org/t/openstack/build/6316dcc2a7b548f2907df458171576c3 : SUCCESS in 5m 10s\n- cross-glance-py38 https://zuul.opendev.org/t/openstack/build/d1735a9f116a4bdaa7a7598213d0181f : SUCCESS in 5m 19s\n- cross-heat-py38 https://zuul.opendev.org/t/openstack/build/82d97d3377134da3a39907ea321ccbea : SUCCESS in 8m 17s\n- cross-horizon-py38 https://zuul.opendev.org/t/openstack/build/6287e5707a54469bbc78c0a8c1e483d1 : SUCCESS in 8m 28s\n- cross-horizon-npm https://zuul.opendev.org/t/openstack/build/591b803af5314d8493e9e5140ff3337e : SUCCESS in 9m 09s\n- cross-ironic-py38 https://zuul.opendev.org/t/openstack/build/5c58097fc83e4faf8f15f280c157b743 : SUCCESS in 9m 55s\n- cross-keystone-py38 https://zuul.opendev.org/t/openstack/build/62cfdde85cd445b7946355294bbd0b16 : SUCCESS in 14m 24s\n- cross-kuryr-py38 https://zuul.opendev.org/t/openstack/build/07b3f1022bdc4965bbb681d97a9cd67c : SUCCESS in 4m 09s\n- cross-manila-py38 https://zuul.opendev.org/t/openstack/build/e114d2edb9aa4673874fb26afb7e196a : SUCCESS in 12m 15s\n- cross-masakari-py38 https://zuul.opendev.org/t/openstack/build/91f18c803daa4af2af2b73c6a3915cd8 : SUCCESS in 5m 09s\n- cross-neutron-py38 https://zuul.opendev.org/t/openstack/build/77eed57dde304d95b2759bf8ea2e5277 : SUCCESS in 22m 07s\n- cross-nova-py38 https://zuul.opendev.org/t/openstack/build/70a0602038ad4646a5a479e571a4b466 : SUCCESS in 13m 38s\n- cross-placement-py38 https://zuul.opendev.org/t/openstack/build/689d8adfd83b4ca690e9137b3138da83 : SUCCESS in 6m 00s\n- cross-osvif-py3 https://zuul.opendev.org/t/openstack/build/44a45d2bf31147d2ae9058e6e6343d35 : SUCCESS in 4m 13s\n- cross-nova-functional https://zuul.opendev.org/t/openstack/build/89ae3c2059c7488db59bc31fbbc8926d : SUCCESS in 18m 59s\n- cross-placement-functional-py38 https://zuul.opendev.org/t/openstack/build/3338b340f8134e84a46db0f4c4053399 : SUCCESS in 6m 21s\n- cross-octavia-py38 https://zuul.opendev.org/t/openstack/build/36da50071b1e4086a3dd64354dc060a4 : SUCCESS in 7m 40s\n- cross-osc-tox-docs https://zuul.opendev.org/t/openstack/build/2e5e03d2bca74988bcd1c1f1f7f5296f : SUCCESS in 6m 56s\n- cross-swift-py38 https://zuul.opendev.org/t/openstack/build/06a736ad545143b4a2bde77d208efd04 : SUCCESS in 12m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ec9c9dcfd08e4de887899e9b241752f2 : SUCCESS in 53m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/09443ac21b284a7a85d3e8cdfde2ffd2 : SUCCESS in 1h 18m 43s","accounts_in_message":[],"_revision_number":5},{"id":"6cde21b923e37a3ddaad90f2ff65e5414afc818b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 16:46:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"37abe5f7fe07382456af68c97c815240a758e292","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-06 16:47:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d64624b1d3a94ac8b87d928ba08dbde1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea4bd86beee54339aeb920adde5171b7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"271edc43404fa8cc096cf31a7ff60b8a7cff4b5f","revisions":{"374b0dbb7aa7a61147a92b4e1a214809747d1b55":{"kind":"REWORK","_number":1,"created":"2022-08-02 02:35:04.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/851785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/851785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/851785/1"}}},"commit":{"parents":[{"commit":"bed46f4b7ab58badcddd45e0666481eb89aedcaf","subject":"Exclude Werkzeug 2.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/bed46f4b7ab58badcddd45e0666481eb89aedcaf"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-02 02:35:03.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-02 02:35:03.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6927438d571ad4c9dd5a55545bba5c0becdcb20c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/374b0dbb7aa7a61147a92b4e1a214809747d1b55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/374b0dbb7aa7a61147a92b4e1a214809747d1b55"}]},"branch":"refs/heads/master"},"5145f708580701de567cbd918b51b5bb7cd35986":{"kind":"REWORK","_number":2,"created":"2022-08-03 02:31:38.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/851785/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/851785/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/851785/2"}}},"commit":{"parents":[{"commit":"9dcc4636b1db4da27a74c631333e4fae23c9f877","subject":"Merge \"Allow backports to have arbitrary python_version markers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/9dcc4636b1db4da27a74c631333e4fae23c9f877"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-03 02:31:37.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-03 02:31:37.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6927438d571ad4c9dd5a55545bba5c0becdcb20c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5145f708580701de567cbd918b51b5bb7cd35986"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/5145f708580701de567cbd918b51b5bb7cd35986"}]},"branch":"refs/heads/master"},"94fa4739dc50438ed45a8c7de0c9f5db789598ce":{"kind":"REWORK","_number":3,"created":"2022-08-04 02:20:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/85/851785/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/851785/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/851785/3"}}},"commit":{"parents":[{"commit":"8efe11a74a87f4a1522099ab5734c6af628c540c","subject":"Merge \"update constraint for openstacksdk to new release 0.100.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8efe11a74a87f4a1522099ab5734c6af628c540c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 02:20:36.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 02:20:36.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6927438d571ad4c9dd5a55545bba5c0becdcb20c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94fa4739dc50438ed45a8c7de0c9f5db789598ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/94fa4739dc50438ed45a8c7de0c9f5db789598ce"}]},"branch":"refs/heads/master"},"c903e58dd11c69c3e4bbd43c1b72f29df2a46e85":{"kind":"REWORK","_number":4,"created":"2022-08-04 06:35:57.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/851785/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/851785/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/851785/4"}}},"commit":{"parents":[{"commit":"8efe11a74a87f4a1522099ab5734c6af628c540c","subject":"Merge \"update constraint for openstacksdk to new release 0.100.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8efe11a74a87f4a1522099ab5734c6af628c540c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 02:20:36.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2022-08-04 06:35:34.000000000","tz":120},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6927438d571ad4c9dd5a55545bba5c0becdcb20c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c903e58dd11c69c3e4bbd43c1b72f29df2a46e85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/c903e58dd11c69c3e4bbd43c1b72f29df2a46e85"}]},"branch":"refs/heads/master"},"271edc43404fa8cc096cf31a7ff60b8a7cff4b5f":{"kind":"REWORK","_number":5,"created":"2022-08-04 16:25:54.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/85/851785/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/requirements","ref":"refs/changes/85/851785/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/requirements refs/changes/85/851785/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/requirements refs/changes/85/851785/5"}}},"commit":{"parents":[{"commit":"8efe11a74a87f4a1522099ab5734c6af628c540c","subject":"Merge \"update constraint for openstacksdk to new release 0.100.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/8efe11a74a87f4a1522099ab5734c6af628c540c"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2022-08-04 02:20:36.000000000","tz":0},"committer":{"name":"Dr. Jens Harbott","email":"frickler@offenerstapel.de","date":"2022-08-04 16:25:54.000000000","tz":0},"subject":"Updated from generate-constraints","message":"Updated from generate-constraints\n\nChange-Id: I6927438d571ad4c9dd5a55545bba5c0becdcb20c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/271edc43404fa8cc096cf31a7ff60b8a7cff4b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/requirements/commit/271edc43404fa8cc096cf31a7ff60b8a7cff4b5f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14288,"name":"Matthew Thode","display_name":"prometheanfire","email":"mthode@mthode.org","username":"prometheanfire"}}]}],"submit_requirements":[]}
